)]}'
{"id":"openstack%2Fswift~937391","triplet_id":"openstack%2Fswift~master~I6d8ad06c645f25497dc15173460430fd93747afa","project":"openstack/swift","branch":"master","topic":"p-cleanup-expiry-queue-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-10 18:56:34.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-10 16:01:32.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-10 18:56:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d8ad06c645f25497dc15173460430fd93747afa","subject":"Clarify ContainerBroker tests re expirer queue items","status":"MERGED","created":"2024-12-09 17:48:45.000000000","updated":"2024-12-10 18:58:26.000000000","submitted":"2024-12-10 18:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"937391-p-cleanup-expiry-queue-test","meta_rev_id":"af310dff99dd21c36264251fef068320bf191f63","_number":937391,"virtual_id_number":937391,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 18:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-12-10 16:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-12-10 16:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 18:53:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 19:40:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-12-09 20:04:39.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"ef0b6f6831db8af1f7da4a751801b65d967c58c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-09 17:48:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b1d416a87828b7581ae76b0ae66f1eef78927c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 18:53:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb0addb0f3645108a45761c83177795\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae7f4c9ba754c01a3260e0d15490a90 : SUCCESS in 8m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/afa64efb3475416eafe3a8eabcd8799c : SUCCESS in 58m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1443af27566449b3919798f9f634c6bc : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e329fbe4e754e9e8d049ae56979408e : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee94fe9f1b5f420faf0e8680450adcd8 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc1f59874f254961aa008a328fd035d9 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a711be047f264e8ebdca78a613c779ee : SUCCESS in 19m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c2b76c0cc5347ff96bddcc03f2f9f02 : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/091d22027e3b4dbea649302e2537c76a : SUCCESS in 18m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd3d8f7cf0c144fab021d205a9850689 : SUCCESS in 30m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ef7744fad6104957bee32382c171e0e1 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f860dcae87cf4debacc4bfe012fbfd98 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4c95640ca1741888c1bec4b73533aa6 : SUCCESS in 18m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6bed399bfbe64eb9aeeba52f96f9887b : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4fb91ebf2964a2db26f8f374a1078c3 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/128508577f97495c990dff2e692da078 : SUCCESS in 17m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf1dd2137bc44bbfb8048ce022978cd7 : SUCCESS in 52m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7982c6839a34b6eada37abd83b82acb : SUCCESS in 45m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d75cb16350204d489f0d73477266f4ee : SUCCESS in 23m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ecffe4bd3b45dab6c40984d0734501 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/886872b18f004b4a8ca525bffa075b14 : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f43952621b25d9e9deb54fc33936e6b69553e204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 19:05:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b863600b68d44ff85726ac819f74adf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71d02b42fcff477dbf860742090ed21b : NODE_FAILURE Node(set) request 300-0025887997 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6dc57cfbc72c755075d9fc74cdd9d46ada961609","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-09 19:40:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ad883132cf8aa46d4eba61cfce6232659914234","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-09 20:04:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0ea86877730784a6ce60d547c72750aa2d7a2a72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-10 10:51:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"10a382a833211d8e28ff1215c356e609d009620f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-10 11:02:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34bd25b4534a73c1e039c732cccd28829e5afc65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:56:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5468ad940cea45dbb10f29a342097754\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c39b8fc4f34e5aa968d2840d1ba8da : SUCCESS in 4m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0b9d0a757af4cc1956e21454bce822e : SUCCESS in 1h 04m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42f52b6124f745bbbe5be2b89fee18e1 : SUCCESS in 21m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77342ed1e65240f3b1ffcdab89e918a1 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1eaacc1e99fc4175acb4c368d75239a8 : SUCCESS in 20m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03cea7bd67dd422aa91efa150d1c82f8 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/099e49f583854c7893366d0221f367d0 : SUCCESS in 18m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3db73363ce614d629d659e06a322b648 : SUCCESS in 18m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a2ccc90fd14403988f52663428ee92d : SUCCESS in 20m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/02c70015d0564437be7f066fea767e35 : SUCCESS in 34m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ef3a3385e47477e825e62951606d8d1 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/238e53913c9e49dda3c3a124e879107a : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/789580a56e16483ab2eb0afbe80afa9b : SUCCESS in 22m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b02ec936bf0d4fb7a54f36891ce99f8a : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/852768a7b200470b88f59d944ee4417e : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac8b4265263447f4998700f6c96b50cb : SUCCESS in 20m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/383a1545c7204c0f84c58f8d64148390 : SUCCESS in 44m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10445dcb3fd34fba87995f1058498194 : SUCCESS in 46m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7789a9500c44cedbb5fdc5abe16bd98 : SUCCESS in 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a15e8b5f8543b091fbe436bcc43973 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63e83a22e1194797be10564df91e0bd4 : SUCCESS in 26m 47s","accounts_in_message":[],"_revision_number":2},{"id":"eb0038f30d8e287290bef62ce1ab7f6725ea33ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 12:18:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0772f32221164cd2ac9423691c57075d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a74b574c6bda420fab9a763fe515daed : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"97e55700a3b4ca56584aeac4818c3c3ab50f129c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-10 16:01:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"077f87bbc71d66420a61c9f5c1051236770a338e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 16:02:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4525b13517a9846bd32b01081a6a082b29f62b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:56:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7117a08c713d4dd580855c02d8a9fe69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593fcfae90be418e8d067454ff39341b : SUCCESS in 13m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/306e2e93971f4dc79b704290a0c96924 : SUCCESS in 49m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/61ce5e3e6141405e89fddc01db06f63f : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83fb2a781f6340caaa71de7a3bffd04b : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ee97e3bf1ea4c5589b2d558a64ce62c : SUCCESS in 15m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9274f3aa50ac458ab93615ebe8b86c2d : SUCCESS in 19m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/55d2d86c9bc546d194765e5328097155 : SUCCESS in 33m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5294c59ad69b43e0b8d89f63ec01747e : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4048661c4094e1dab80492a0d483c6e : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa20df7a617b406fae1f26a464efdb86 : SUCCESS in 18m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdb9a6424ebd4e94937ac16a66778812 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f2f134ff4ab9417f8135b1bf3d196320 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0ae5c537cf6c46adb6a77df4283915bc : SUCCESS in 12m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf44c33a262c403d9fa6ce5766268938 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c693f164fa1a4f28a3360c35fe29269d : SUCCESS in 6m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a00a18f2a0b419e957509f09fec7ddc : SUCCESS in 33m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a36fb29d926417b87aec3abf7a4aa2d : SUCCESS in 43m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49bd2cd53f1d472d9bfcce83790c38d5 : SUCCESS in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25cdabefb64a405d8a4a30ca2c3ce220 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4acaeae3dbed49a2b51c3c8352277c8c : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d03dfdfb4301e1a0945250bf27aaad6492754e4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:56:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"af310dff99dd21c36264251fef068320bf191f63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 18:58:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22776c918694422d9226e84fd4d2c3c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5baa47d141944c5823558b68cc8680f : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b7ebe1ad6f124bdfbb013e2a6a5cba60 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a5bbcd7a6501962ee4fd545e99de2e16d2c2853","revisions":{"53139e1b24f3d72666a8ad04d141bd72b4eebc1c":{"kind":"REWORK","_number":1,"created":"2024-12-09 17:48:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/91/937391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/937391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/937391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/937391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/937391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/937391/1"}}},"commit":{"parents":[{"commit":"b4579b414e51db88bfbcae5d3d678ed8fe1a7941","subject":"Merge \"CI: point antelope job at 2023.1-eom tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4579b414e51db88bfbcae5d3d678ed8fe1a7941"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-09 17:40:31.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-09 17:40:31.000000000","tz":0},"subject":"Clarify ContainerBroker tests re expirer queue items","message":"Clarify ContainerBroker tests re expirer queue items\n\nAdd some commentary as a reminder that whilst normal object updates to\nthe ContainerBroker cannot have content-type timestamp newer than data\ntimestamp, expirer queue updates can. Therefore the tests do make\nsense in that context.\n\nRename the vars to exactly what they are; the author of this patch\nfound the references to \u0027put\u0027 and \u0027post\u0027 confusing in the context of a\nContainerBroker test.\n\nChange-Id: I6d8ad06c645f25497dc15173460430fd93747afa\nRelated-Change: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53139e1b24f3d72666a8ad04d141bd72b4eebc1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53139e1b24f3d72666a8ad04d141bd72b4eebc1c"}]},"branch":"refs/heads/master"},"3a5bbcd7a6501962ee4fd545e99de2e16d2c2853":{"kind":"REWORK","_number":2,"created":"2024-12-10 10:51:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/91/937391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/937391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/937391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/937391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/937391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/937391/2"}}},"commit":{"parents":[{"commit":"b4579b414e51db88bfbcae5d3d678ed8fe1a7941","subject":"Merge \"CI: point antelope job at 2023.1-eom tag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4579b414e51db88bfbcae5d3d678ed8fe1a7941"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-10 10:50:38.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-10 10:50:38.000000000","tz":0},"subject":"Clarify ContainerBroker tests re expirer queue items","message":"Clarify ContainerBroker tests re expirer queue items\n\nAdd some commentary as a reminder that whilst normal object updates to\nthe ContainerBroker cannot have content-type timestamp older than data\ntimestamp, expirer queue updates can.\n\nChange-Id: I6d8ad06c645f25497dc15173460430fd93747afa\nRelated-Change: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a5bbcd7a6501962ee4fd545e99de2e16d2c2853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a5bbcd7a6501962ee4fd545e99de2e16d2c2853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
