)]}'
{"id":"openstack%2Fswift~937494","triplet_id":"openstack%2Fswift~master~I1e9d4d5df403040d69db93a08647cd0abe1b8037","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-11 05:42:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e9d4d5df403040d69db93a08647cd0abe1b8037","subject":"docs: Call out that xprofile is not intended for production","status":"MERGED","created":"2024-12-10 23:22:25.000000000","updated":"2024-12-11 05:44:42.000000000","submitted":"2024-12-11 05:42:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"937494","meta_rev_id":"55a01d0cf1086298a7204aa629d8d62c3e3c7652","_number":937494,"virtual_id_number":937494,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 05:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-12-10 23:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-12-11 03:20:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 23:40:27.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-12-11 00:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-11 00:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e619ca54b2549d242ab2527f5bfa81b412f32624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 23:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bedad588c2ce6dac14c156c7f395e287d65b09c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-10 23:40:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"270e774cacdc10008c87f4fbaa75d3d086cf859e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 00:39:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf1acf494884a58a8c4183ffe756349\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a601d6d5a1c8417e9956e239026e63f3 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"35174d2a9688e7c3033ca0060809d853657197c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 00:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d307366f0dc4a73a6f90d2831ba6de1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189a98b992f240de91933529a7d3405b : SUCCESS in 5m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/cd07bf781d5a44818345ffccedba15b3 : SUCCESS in 58m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d00f0bbff454f57b8b62cef2f037f4b : SUCCESS in 46m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b2e03a47d404442b3b0c4c66a69c32f : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c818bb4396430c8bfe2919997843b7 : SUCCESS in 53m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ad3403bf229401fb7dd5616685a3be7 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6de24d8dc3864bc6b8ed0e037cdd6897 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf493b5dd5294e82b7cc3c72839330b3 : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0dda7d90213b4c37b0cba4ad4cdded1b : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fdf372a593a47bc89101636640ba4da : SUCCESS in 10m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e57a935af8a340a393702912e62b8b7f : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/63154c315aee4bb3b1b2dbe61697ac6b : SUCCESS in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/de86a4c04436494b8b1f43188b3a0f0c : SUCCESS in 32m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2435f6b18e7040d9894418ea31419d26 : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3bb7b7798c65429e886b0a31fb2c9202 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a1dd2335bd84f569998b1fe0c23b8fd : SUCCESS in 17m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2a5d13d73d5f48949392d7ad78986724 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cbfa4e4c0f3248cfa37b28dfa8d4b684 : SUCCESS in 9m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7eb6a5b9e7064e8a8e43ed8031144349 : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87767fa8374d41a9953c446ee21a0331 : SUCCESS in 4m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/595eb9ce167e4bf4a2ea3870b6be2211 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/776f243f3b084fbc980407ea39d6d95b : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fba4909540a46fd810a11b32f18fd01 : SUCCESS in 55m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f5be17148f14659bca4ff813d47ef3e : SUCCESS in 57m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbc352629d83414ab082c64029042560 : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/559e48bef4a643928f5639ddbb803128 : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9b9785e766450dbaf5b4b5dc426b0a : SUCCESS in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63014504f221449fb9e4a0dbc0c8a5c6 : SUCCESS in 17m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1b6462eabea47aa94a9a47e976a175e : SUCCESS in 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04f91ea3ec7c4bd7b0b1db6a92060cc7 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ed8619937338314d92e295065e9c50f2f00fb408","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 01:18:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e2805ea8b2d39bdf3b5b9e1b675a00bed024afba","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-11 03:20:40.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"430bd2ddd865057079d4657c334c14627c2c0200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 03:21:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d69cf2e66d93041486de1821947640fd69acba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 05:42:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1262008a7f8e4d4ea90c818444f61b09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8c7a9cb985488d9471cfadea9d9978 : SUCCESS in 7m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/5aae1267f5074e2d906aeca3cb3cd708 : SUCCESS in 57m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a1b190a48eb4dd3a3a3b0ee0f8fe0fe : SUCCESS in 55m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8441646bada742cda5c0a23e4854462d : SUCCESS in 51m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73e25bd7935149b5a692b58a5a6a21f8 : SUCCESS in 1h 04m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/16400b979e2e4a0fb545862b4537eed1 : POST_FAILURE in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/372a33ba09ae49408fdc40f8048cba4a : SUCCESS in 22m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bff4aba15a54b8a9c4da1326e913193 : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6cbc15d0b1a4380bd514f8ce672c527 : SUCCESS in 21m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7dee46955f454149b3659e8c3cb72397 : SUCCESS in 23m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b16980e5ce984543bf25244bc331d870 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd4eeb0a809748f59b68265720cc6f6a : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c5f32dc968744db1bb12e1499c7e8c03 : SUCCESS in 9m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/569c0cb8ba5c464285d965f02163014e : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9389568f65c3496b9cb8f381f1548667 : SUCCESS in 13m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b67baf1284c6478d9942536fe4d79ed3 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2cf12b31afbc482e82d1db97df80aa9d : SUCCESS in 4m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96822bb7023c4111b66cc289bdcab0a8 : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ec11a1d01194877a9cdca4457decfe2 : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3f0cecd8eb54368a1462bb06669a3b4 : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f536b476988e471daa84f2fec22c015f : SUCCESS in 32m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2b74a264157447fb62b806129d14e6f : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32784e224dc491b8312f1879cb9c4db : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/747a01459d6540a09ec6af8147297871 : SUCCESS in 17m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd4a0ba66734438bb51a52af1bf92172 : SUCCESS in 36m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9296142fd174ae7b9ec2eb443278ad4 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2e437b9a067d81df71310f400378d98d5a940b99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 05:42:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55a01d0cf1086298a7204aa629d8d62c3e3c7652","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 05:44:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2da663749784b88ac40a59ae18ef234\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa43074d5b35460aba53cb112a85a907 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5f22cea7c3a84e56a673bf1105d46bbd : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a55a48ffc8bdd04a4876a01b92b204302bd62881","revisions":{"a55a48ffc8bdd04a4876a01b92b204302bd62881":{"kind":"REWORK","_number":1,"created":"2024-12-10 23:22:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/937494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/937494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/937494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/937494/1"}}},"commit":{"parents":[{"commit":"199aa78fbe647f336fecf6d3b76d07964b841128","subject":"xprofile: Stop using eval()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/199aa78fbe647f336fecf6d3b76d07964b841128"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-06 22:24:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-10 23:17:11.000000000","tz":-480},"subject":"docs: Call out that xprofile is not intended for production","message":"docs: Call out that xprofile is not intended for production\n\nChange-Id: I1e9d4d5df403040d69db93a08647cd0abe1b8037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a55a48ffc8bdd04a4876a01b92b204302bd62881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a55a48ffc8bdd04a4876a01b92b204302bd62881"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
