)]}'
{"id":"openstack%2Fswift~937495","triplet_id":"openstack%2Fswift~master~Ia054e4f7c9e5bf29064a66933e27830adbc107d3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-12 10:43:45.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-17 10:55:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia054e4f7c9e5bf29064a66933e27830adbc107d3","subject":"CI: Configure bandit better","status":"MERGED","created":"2024-12-10 23:22:25.000000000","updated":"2024-12-17 10:57:10.000000000","submitted":"2024-12-17 10:55:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"937495","meta_rev_id":"c6c0c4366e739c69e49a3ed2095979e930eb26e5","_number":937495,"virtual_id_number":937495,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 10:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-12 10:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-17 09:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 23:54:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-12-11 00:39:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-11 01:29:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-12 10:43:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-12-16 22:35:40.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"6a18c7c5b635dedd922144054a13800c34ffb915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 23:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48c71d4795fe82d50c2e852dbd3c75346397f615","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 23:35:45.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9a2453f04976962248af11054b1f18dfa2ebcde9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-10 23:54:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7d6678f488c0b15081f9f58c8e8cb2a0cc77467f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 00:39:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0647ccb266947948b9d566eb845cb27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/277eb216eb304ae3bab2de5886dc9f53 : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"248d70946f4c7951d64e36493b8a8abf76fbe043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 01:29:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d782f36901bb48869b6eb8a20ad8de08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edf7c1334ba476f9c4fdebfb22e83ae : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/cf82bc8fe1284067b174b873266b60b6 : SUCCESS in 1h 03m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fba963a5de9646068a8242d0fe8199ce : SUCCESS in 1h 03m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e2dc95dae9e4583b6f91cc60ea03b6c : SUCCESS in 1h 54m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f80f2d987e9146c79fc5f3f4c9a3c36c : SUCCESS in 1h 09m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2fc424031ff414e84f8671b60034df7 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e2ed0e057a9494491c2193fb3c1a382 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27e2e47ad24c47ce8ebee977f6751efd : SUCCESS in 17m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87eef7333ac84706b04b007f3fcbf69e : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cca10bdc759244fbbefe4bf70f7187d8 : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/12a3bcde78eb4e9b8f26e1e126fd8ad2 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ddbad5b5f33f486bbbd1bddd2b39b29e : SUCCESS in 18m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8bfddcf69b514f86a6f093aac61ae51d : SUCCESS in 21m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70b534c427754ced8bc6c37c83e272f9 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd6d622d0eda402897d3fc7603788027 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd12c1e5b0044581a9dbdc848b8c4a3a : SUCCESS in 17m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3380fce133cc4aaab93ad45ea6fd7454 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/94dc2a8f22c5483ab327518a5c509c1a : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e26eee2f0d284d39a5a6ffe939d87d76 : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/180f7d15dfb343cbbacd3b42ee457aeb : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/796fc28e6c944e45bd3d0997412a8019 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73189a03f7b547269555d959cf6a16e5 : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c8ddd2084184ac38b83f62ba0857317 : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec3ff2d5c57049e3b03aa75f6f51e823 : SUCCESS in 37m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d451978ac70e4042932f1ccb03256795 : SUCCESS in 46m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e36187e867444c1988a621b08778d325 : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba9cab35cd54c549b00f1bcdc22bcbf : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd726b6c2e2b449bbba358165a54f5a9 : SUCCESS in 25m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb7dd17940944e588a132aae28c4bebc : SUCCESS in 53m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b7c63a1e86e431d88c7cc28fe629b69 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7f0affeaeceec59df84e04b77957afb40c9b5bc7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-12 10:43:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"44bdc88fec04a6dafdfe00b23af12ac5e28e0f7f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-16 19:29:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f99bc3cf69ac9a4df151a5fea633e5e5411e8615","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 09:53:51.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76931a2591027b5390c31b1ae9374d79be846548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 09:54:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f42a41f0d283732cf51cf306decc8da074b78ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:55:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80a67057ed2b47e1b639618f6be5b0c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a6b2a54d25478d8166cdb23c5cf52c : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/22d0ef6eb5b0458ea2bad7110e911c0e : SUCCESS in 51m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57ddcd4f52b84ccc8f9cd6a50d723bd0 : SUCCESS in 41m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7fb3421124104bc99b13f8e7025c51a1 : SUCCESS in 40m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/047a72d3e9234ceca727ab5c22280442 : SUCCESS in 48m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6ad9ef21e26e41c3ba7b6effb71c003e : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32eaf5a3e6fd4797b19a9e72156deaab : SUCCESS in 21m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3874170075024086aa9699d0c75e24e9 : SUCCESS in 11m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c7fccd1b30bf41d98c83bd7d7f96f665 : SUCCESS in 11m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ea9ea1cf02c4cb8b0b7de8753a634fe : SUCCESS in 15m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2e57d146b9f4b389e626cadb88ab48c : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/47962a962f604d7abed38ecfef623e0f : SUCCESS in 12m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da17a73cd6f049a192870bc4f96f2e0f : SUCCESS in 17m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d1508c1cc534e9a8e0509034ba7c95f : SUCCESS in 10m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/257f5b71c35b450cb79a07e9c708b58a : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b96d2e2a382a4ec791b095db0de821bf : SUCCESS in 10m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7d8999ccfbd4d1f9f5fca0a99acfa6b : SUCCESS in 6m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6c4660564da4e788ab9a6b64dfa9b46 : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd002af0ad494520bb74bb17253a1ad9 : SUCCESS in 51m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5e0726d7d2347e89e1584dc82d521c9 : SUCCESS in 38m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a402491a858491b98ec4a67157c4029 : SUCCESS in 44m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a0fdfcef48045cf95d44a7831d04330 : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322b951104694ed0afd2fbd378cae23d : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2299a62983a40ac914384f83542ec1a : SUCCESS in 20m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61d7b4a226a745bcb56ac6eed1dced3e : SUCCESS in 51m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae0b633addb1444fb300166cdb010567 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":1},{"id":"13c8d4a681109a8ee86cce1993cdc27d3174fb1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:55:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6c0c4366e739c69e49a3ed2095979e930eb26e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:57:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d58343e0c3348a980b171588b44c746\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5186cde0f15941249de9ced695cb4016 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/146ae3182524494b9dc34fa47fd3921c : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e576c5cee0240fbfa87566be906a51d9ac1613b2","revisions":{"e576c5cee0240fbfa87566be906a51d9ac1613b2":{"kind":"REWORK","_number":1,"created":"2024-12-10 23:22:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/95/937495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/937495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/937495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/937495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/937495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/937495/1"}}},"commit":{"parents":[{"commit":"a55a48ffc8bdd04a4876a01b92b204302bd62881","subject":"docs: Call out that xprofile is not intended for production","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a55a48ffc8bdd04a4876a01b92b204302bd62881"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-06 20:21:01.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-10 23:18:12.000000000","tz":-480},"subject":"CI: Configure bandit better","message":"CI: Configure bandit better\n\nDeclare the tests to skip, rather than the tests to run. This ensures\nthat we pick up new bandit checks automatically.\n\nI recently noticed a use of md5() without the usedforsecurity\u003dFalse\nkwarg. Confused about why this wasn\u0027t caught in the gate, I eventually\ntraced it back to B303 (which we explicitly enabled) being largely\nsuperseded by B324 (which did not exist when we wrote down the tests\nto enable).\n\nFlag a bunch of false-positives with \"# nosec\" comments, resolve two\nother errors, and skip some more-pervasive errors, to be resolved later.\n\nChange-Id: Ia054e4f7c9e5bf29064a66933e27830adbc107d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e576c5cee0240fbfa87566be906a51d9ac1613b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e576c5cee0240fbfa87566be906a51d9ac1613b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
