)]}'
{"id":"openstack%2Fswift~937508","triplet_id":"openstack%2Fswift~stable%2F2024.1~I39bb87f9d8e27f2f88410a087a120a0e9be1a243","project":"openstack/swift","branch":"stable/2024.1","hashtags":[],"change_id":"I39bb87f9d8e27f2f88410a087a120a0e9be1a243","subject":"xprofile: Stop using eval()","status":"MERGED","created":"2024-12-11 06:20:35.000000000","updated":"2024-12-11 09:35:53.000000000","submitted":"2024-12-11 09:34:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937508","cherry_pick_of_change":937493,"cherry_pick_of_patch_set":1,"meta_rev_id":"0bb585ff4e4ebd4b6a41eb89e872febda2172976","_number":937508,"virtual_id_number":937508,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 09:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-11 06:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-11 06:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 08:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-11 08:20:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7a869a326b374e11154af91080807f65f474133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 06:20:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e40835d063069b4e2e041a16b6cd64e70d9b5a8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 06:21:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e912523ef9a1774b0cf4a0837432c45c1bb221df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:06:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a69c13aaca43ecb09ddfd934d46801\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5b1eef383714d2ab3f3d17ee008b247 : SUCCESS in 1h 04m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/825635ccd7cf46f5abac65d997586c03 : SUCCESS in 1h 18m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/228124f0fddf429f811adbb917645dcc : SUCCESS in 1h 12m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a50bb2032cee4955b05ee80e2195ac29 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cfe3592cb0148c58ebd4d71a5d8401fee679093c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:20:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bb9c728914d45fd837cc1196686e45f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e0bad351d64657a8de5efc4fdee6f2 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/08fc435f9f434effb3558daaef8554af : SUCCESS in 1h 09m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/39d4d42d24f2491cb7937b441be036f3 : SUCCESS in 55m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c104db1ac259465e822ea06cc5a6e826 : SUCCESS in 33m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4d38b43f181405ba135cdda66b040f6 : SUCCESS in 1h 16m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e5a20d0808f479abf281af909621f32 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0292dabd6ae48faaf3c69aba01c5602 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4eee6fa421dc4532962f83b8e5d795e5 : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aef4a18324f54a8da6301cd6f4b4171b : SUCCESS in 11m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ece3d2fec6142c384ad0fb52e36dc2a : FAILURE in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/31538b284bda45fc9d9ac577e8201c7e : SUCCESS in 14m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5001e248a93148c5ba9349bb88ec99c6 : SUCCESS in 19m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/27c40442eae9496b99c12a475d9a7a7e : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c0e90876aa014935ae74e9c0dd37f07c : SUCCESS in 17m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fdafc292108450aae68e6683eae2144 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4201aeefc04d402cafbe2dd8586e43e0 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b563057f6f234f0295459c26ff6c8dbd : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ff66f0cf2a904cb59339c79fb97ba22f : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/754001b2d3704095930593aa32fcf15d : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0d343c446a7b4901acbd34ea35129dd7 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86e070c7c6694663beb063bb67cb283e : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23bcc1417772468fb07703eb5b4325bf : FAILURE in 15m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b558dc2515e48f5b77c052ce1c32e43 : SUCCESS in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a76bd139ebaa4ecf92f32fdf8df749ec : SUCCESS in 58m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3dff2d784ff34b5bb3ad80a02b02c112 : SUCCESS in 52m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/892c64744b1a4fe084bab82e08f2d580 : SUCCESS in 47m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e60d3b02f8794899b147c5c65052c491 : SUCCESS in 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91eb4b6a4a3482396d599ce5c41fd9d : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05fc13f5d24d4a05b9cd1f5cf522b19b : TIMED_OUT in 1h 30m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d3dab8695ee4e7d92230695b03f23ee : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":1},{"id":"784ca9a650aa762749f5d34847b6c73d4d1b7d30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:20:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"901acb140cdf404a0d10858f1a0ff5785e7e24df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:34:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db55879d6422464ea34326f8f984d70c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526b61628979422e8f532d90ea6632f6 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4bdaaeb2aaaf49f8911481b83332e5f9 : SUCCESS in 1h 04m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c4b1023e699743f793a79263ca402ce4 : SUCCESS in 1h 08m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/416c3300212341e780ca61f9973c2207 : SUCCESS in 51m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f0b051963764a17b95e366be5e66760 : SUCCESS in 1h 01m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f16561d4660544b49d8b9ed663da1fcb : SUCCESS in 10m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f836a9342919457aa93715ea52ce4230 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c63820b590744fbbe98f0ae5e7660a2 : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/748fd045ce744a1a9cc7a7e3a7dc4ef6 : SUCCESS in 17m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/088feabb03234640947a58aad7ee1046 : SUCCESS in 14m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a84bc7fdb98841b3adc71a401cffe0fb : SUCCESS in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ebd9272380846609aafab422b4a639f : SUCCESS in 25m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c5d41baef4247c0ae450e0615f37f33 : SUCCESS in 26m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0711a497f77a46b29279bd6fb3dcc57f : SUCCESS in 26m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3973a43664af4282987513adfe20c6bd : SUCCESS in 23m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0f5fb14c4f3e4131904a253cd1b3f9c8 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/81f684e3c1f94a22885e099ade933d8c : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a6f33cbe8e4455a9b398757a0c72237 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec10b5aefe72471ba5965078b533c793 : SUCCESS in 6m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fce9f98cd4e4df59151ff53ba2daa90 : SUCCESS in 1h 00m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1843241342a74c29aaf04d5b7b649ac5 : SUCCESS in 48m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/852af8eaa9674c2bb5485bb399404c02 : SUCCESS in 29m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04f1fa3120be4289a9a625d00dd72922 : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599fcae23cc6420d87c47646297ac7e3 : SUCCESS in 4m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b8d2fbcb764239892b652d854f2e05 : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":1},{"id":"445dae498ffc78b1aaf13d6e2694ab4c07bc33f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:34:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bb585ff4e4ebd4b6a41eb89e872febda2172976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:35:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae4cabc608fb4a218421d34bae121089\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9adc58fc67c74c448d8d70e553a8abfc : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b565f241543340e4aa950f55483ce6bc : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4fadeb57c3be4b29916b5e5e2b99e14b : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18b0df30b09edac73375121e478b20cd20708a85","revisions":{"18b0df30b09edac73375121e478b20cd20708a85":{"kind":"REWORK","_number":1,"created":"2024-12-11 06:20:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/08/937508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/937508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/937508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/937508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/937508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/937508/1"}}},"commit":{"parents":[{"commit":"e0818f55b0e2528cef267e2590bbcacf0341de64","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0818f55b0e2528cef267e2590bbcacf0341de64"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-05 21:43:13.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-11 06:20:35.000000000","tz":0},"subject":"xprofile: Stop using eval()","message":"xprofile: Stop using eval()\n\nAll we need is int(). Using eval() on user-provided data (or really at\nall) is a Bad Idea.\n\nCloses-Bug: #2091124\nChange-Id: I39bb87f9d8e27f2f88410a087a120a0e9be1a243\n(cherry picked from commit 199aa78fbe647f336fecf6d3b76d07964b841128)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b0df30b09edac73375121e478b20cd20708a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b0df30b09edac73375121e478b20cd20708a85"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
