)]}'
{"id":"openstack%2Fswift~937509","triplet_id":"openstack%2Fswift~stable%2F2023.2~I39bb87f9d8e27f2f88410a087a120a0e9be1a243","project":"openstack/swift","branch":"stable/2023.2","hashtags":[],"change_id":"I39bb87f9d8e27f2f88410a087a120a0e9be1a243","subject":"xprofile: Stop using eval()","status":"MERGED","created":"2024-12-11 06:20:48.000000000","updated":"2024-12-11 09:45:09.000000000","submitted":"2024-12-11 09:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937509","cherry_pick_of_change":937493,"cherry_pick_of_patch_set":1,"meta_rev_id":"3a2e767a76f57a9fd3669e683ac2e06d78839f0a","_number":937509,"virtual_id_number":937509,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-11 09:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-11 06:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-11 06:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 08:34:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8092f63dbf8736e4f0e19c8265123520b7343722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 06:20:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"70f189119f5c2d64451f841f93c04fb036db3da7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-11 06:22:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18dd4411e13ca54cb410220bc2fd68119953f1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:34:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36f6daf179d440da9fb8d0c79183b820\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ea30895b104a1cb9b072391b281165 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/02bf2896656b4e18b8587337cc1d5dab : SUCCESS in 58m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66be7b237b6d4f4cbc01b6a7e5632c4e : SUCCESS in 45m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8e647bd71a94d56a05f0d81cd29d77d : SUCCESS in 50m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d0078a614524a5cb00db77435b56c6d : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f98b38b50b7e46c29971a90e18f38be8 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/978052ff56e84e4b8759b9a826ad0eae : SUCCESS in 17m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bab7844720f458687d6c704de6610fe : SUCCESS in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad734dd66be1425a8b32e9b81c45d671 : FAILURE in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b730cc685c7428f92d54cb8e1eb4657 : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4437966a985e4ae38c74d83b9c032820 : SUCCESS in 15m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8822d1dbc832460a92929fa038d37a5b : SUCCESS in 14m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/709543686c0b4c4f99f22dcce211e9be : SUCCESS in 16m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b276753f506c437a859fcb29193cb8c3 : SUCCESS in 13m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c89b65bf1c9444b79a37884e991b3e4f : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef4ae68bf5d64c3d82702c06d5e657f9 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9fb9000eb5cd4d639a1ddfa8d98bc68e : SUCCESS in 24m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb32726207364ece826c74e87e37500e : SUCCESS in 19m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d793a7def56e40628a3b1560a0dedee4 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e8f9da551d9b4db4a82b9c5e17d2fe2b : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e494fc6e779c4c4cb006147e3d93e821 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6da2e48e63c34e9eb89bf25362b0090b : FAILURE in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e8e86e24c76f4e53a5ff929d2d498ba7 : SUCCESS in 6m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9c20935ff654e2b8384a18a2ccb606c : SUCCESS in 50m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e67e7c4ef054bbcb6399d9422e7f51b : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b7bcb82836848b99af1cf09fb4a312d : SUCCESS in 55m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce0268e446f3494bbc4ba53fed993106 : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3d5effa7f454dad942308e88112cba7 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05345e46652947fb804fd6f2cc1951e2 : TIMED_OUT in 1h 30m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63b0c41ecbd448f1b1d6b7cd668642e4 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"278640270468332c9d15a71a5d21a97d4a2465b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:35:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c7da1427adf7160ec742f810416300703015354","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:57:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e17471e0c1a40ad83401a93ce4e7f0f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee70430ae4364deea8e3c6d4bea453cd : SUCCESS in 1h 08m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b6d732b5dbb46efb67fb60de7647e76 : SUCCESS in 1h 19m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc4b790c6654457baa823c8405bfea3a : SUCCESS in 1h 19m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2c1f680733e40dbaabfa133694c9890 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1ebea5bc94ae7a5a8c72d9954811aee052ee15a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:44:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26002ef0cda44e5a8155a24d293102d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e507d86d40445dd882c21b37e6148af : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5c2ae78d30db4051af053eaa4e8fe7d8 : SUCCESS in 57m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9eceea3408aa4c71ace3c3c095866ad3 : SUCCESS in 40m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5ab7cad7bff4310bb84f7e81140d612 : SUCCESS in 48m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b27a43ff2453464394a01473d78a70b0 : SUCCESS in 7m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/73255b54b726424abe3ba12b5197b30c : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e8b90824e6134dadb56d6dbdc5b4333e : SUCCESS in 21m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/196666ddfc9f485b9fbb21530b89c150 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37274e80f69b4153a4d8de316d208547 : SUCCESS in 18m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ea9c1035c721401a8e54e87b6964be7c : SUCCESS in 15m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/51b7629636eb4e3bb9af2924a21bafc1 : SUCCESS in 18m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/febd46b5295b46e4a665b9c94fe3fb70 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2cf33de8a7674741b8b7eb6d122ca1fd : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a4c015754fc4f02b37b9a050687fe02 : SUCCESS in 18m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c551c538ad3c45eebec87c220c244fcb : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cdf895d57cfa4adc9251c305cdfcf4a0 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9f7f5af33152405b91ba2f818d3a5c37 : SUCCESS in 15m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a971afbb4a642aeaf4393decf373e61 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a77ca3431d4a492aa948a59dc379b183 : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56b5ac8b23f3470791382dd2d007cfb0 : SUCCESS in 58m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b47cf178b385476daeb3d511a4129370 : SUCCESS in 50m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4dd652003e664879bae325c614624992 : SUCCESS in 55m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3d9e7835d8c444cb527125e3a1ae627 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b9608b8af0485c811196f82ae5870c : SUCCESS in 4m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db399067b7cb446e95283935ac050bf8 : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dd9e461257f9f47eb4d8005c784639526fbf3eb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:44:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a2e767a76f57a9fd3669e683ac2e06d78839f0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 09:45:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be0782de069541ac85bf44db0bfa40f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bba68368b0445d9f7677c139d0fde7 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d06d2cab624542fab79fa8adf4635f30 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/afbbe1fbc29344c0a11f9cfd9f253750 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0236ddaef322ef371a3218aacef2026bbaf98ea9","revisions":{"0236ddaef322ef371a3218aacef2026bbaf98ea9":{"kind":"REWORK","_number":1,"created":"2024-12-11 06:20:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/937509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/937509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/937509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/937509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/937509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/937509/1"}}},"commit":{"parents":[{"commit":"b47d0cf53e6151354168f007f7d8255dd65eb854","subject":"Ensure correct content-type in container HEAD response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b47d0cf53e6151354168f007f7d8255dd65eb854"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-05 21:43:13.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-11 06:20:48.000000000","tz":0},"subject":"xprofile: Stop using eval()","message":"xprofile: Stop using eval()\n\nAll we need is int(). Using eval() on user-provided data (or really at\nall) is a Bad Idea.\n\nCloses-Bug: #2091124\nChange-Id: I39bb87f9d8e27f2f88410a087a120a0e9be1a243\n(cherry picked from commit 199aa78fbe647f336fecf6d3b76d07964b841128)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0236ddaef322ef371a3218aacef2026bbaf98ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0236ddaef322ef371a3218aacef2026bbaf98ea9"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
