)]}'
{"id":"openstack%2Fswift~937561","triplet_id":"openstack%2Fswift~master~I0e62e2865df1937f0296f479e5e6c6337880785a","project":"openstack/swift","branch":"master","topic":"add-alternative-config-to-manager","attention_set":{"37271":{"account":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"last_update":"2024-12-19 02:02:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0e62e2865df1937f0296f479e5e6c6337880785a","subject":"Add alternative config to manager","status":"NEW","created":"2024-12-11 19:13:51.000000000","updated":"2024-12-19 20:34:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"cf5802b97f55deb505e1966333377fdc7fb4b7e5","_number":937561,"virtual_id_number":937561,"owner":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-19 04:05:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-12-19 20:34:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 20:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-19 20:34:19.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"37ad8fc6eb26c1c2a713300aac20f19f08268877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-11 19:13:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f29a7a8a2757a5f944aec49dd1d0514341cf0ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 20:28:20.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a16b8a41f4c6467fb2de84a840e70e08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbc76f0ad604ff5a1a7457fc6ced6f6 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/cedfe58cabb045bd86762590f455fe92 : SUCCESS in 43m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b42d20d8152487fa31a493d5235883d : SUCCESS in 58m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a85ce86afbd4079bf9328f3454f1d82 : SUCCESS in 52m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/418cea1b30194489986b0dad45d3784f : SUCCESS in 1h 03m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e75fccd2a0c74ce5a8bd25c3487dd8a6 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/931711902cb64cf98a93741995760b4b : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20c47ad0e66a41d686c1eb23508df20a : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e331537c8f2d4e5f9f98f262b98cf5ac : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c00a10fb90dd48bd8919390949348f34 : FAILURE in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a288d2d47848aaa488a579b10089c8 : FAILURE in 19m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a155b3656d68476f99c1918b7ddd4ac5 : FAILURE in 13m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8e32e3d02a0d4bfa9a28d01ddb721c34 : FAILURE in 27m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88615af677cc4ae5ad74d683b34fe2bb : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b318f73da9b473eb2831deadbbc5c47 : SUCCESS in 20m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/538b418bb1f0409bab496bb5bf82fde0 : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5cea5772abdd4b46a8643d9034afc111 : SUCCESS in 13m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f366473e2424c7aacd9f1f885c56a37 : SUCCESS in 13m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/45658a533ebd4a8fbff6753768674cee : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5c1566c4275433c90de312d62fab112 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfd19ceb7c524be1be04e12571074bec : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7165ef0b7f6c4348b7d8d5432d65abfa : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf5ee12e50da4c36ba8b3666f9ce8f37 : FAILURE in 55m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d621611c6f849fa86bbd4949d24dadd : SUCCESS in 56m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc36077b55694ddba11b3c163526714e : SUCCESS in 29m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f0c1e0f5382445aa5aa00063066d6a8 : FAILURE in 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf6c7426a2240d786bf9e42c8ad33b2 : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03ae4087497d4a5e8085eb1c81e1e523 : SUCCESS in 17m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07fa3ed2e0684b1ab996be18975c97a3 : SUCCESS in 39m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da3a31bebac14d4eb8ae73b1f8385505 : FAILURE in 40m 43s","accounts_in_message":[],"_revision_number":1},{"id":"027a29b8074fff4e92c456051cbed7767b2b38b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 23:06:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e3d79c4712449c78ff62aed221b30c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a75c2b1d6ce24f83a1d08400a22c5609 : FAILURE in 1h 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9598bdc8ec9419cddb73b72ce43070c8974632e4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 20:30:27.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"dfd90083f3806da85e43ca5c20789c0cccc2e81e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-18 21:52:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"95d5781714675337eaf58777c73bf70a33fa9e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 23:12:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec1eef3df8764bd7896103f75f3f3511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07786922b9d842c1846e0f9693849825 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/455c081154b74792a339d4fe25fbc641 : SUCCESS in 1h 03m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dac8b40170c34ad991a32a9a40665776 : SUCCESS in 1h 03m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02600c7c3d15482fba184e6256c04612 : SUCCESS in 1h 01m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00ecc07c985840baba27a7442301c2d5 : SUCCESS in 1h 11m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67a0cb71605847b08c1f0be44edfce22 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36933d1080934860b96e52b6c4ac084f : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bd7496e32ee445f8151b04522ca5ef6 : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b6316f2b5f7411aa487494c3fe968c7 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af66b3a19ce3448d84758bf12bb3a12c : FAILURE in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9d03cb3565e43b1b437ac40f6db757f : FAILURE in 9m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e90fc0c84ace4eb0b35e17fe9b0d9715 : FAILURE in 17m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f24c5c68a7b4419abd31ce16f6f7f9aa : FAILURE in 16m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ad8ff456a4549c6b570b9bd6e83f300 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/233e403eb77c45b9adecd14e623f5cd1 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/caecd6c6c397430b84ba79c245bf6d66 : SUCCESS in 16m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/56907b38a006475cbbeb031a9ee57297 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c9644da4224d4b30a1a1fc9a6d67ac8d : SUCCESS in 13m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd60b38c8c3d4ce39d604b07fa753ca8 : SUCCESS in 9m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7041c4a78404becb834838f025f4bb3 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd326b3fccbf46e9b2a31e863f6a3157 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0cf07c7280e346a1aeec1c3e52bdaf9c : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e121fc22fa854e07972e4f8646a2525f : FAILURE in 38m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30524efab9114bf19c124e971b3d8203 : SUCCESS in 53m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25f11d862eeb49b897316e326f3433f5 : SUCCESS in 29m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34c6aec6232141b2926f90d3df32a65a : FAILURE in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0952d3a3a1478c82feae74c70003b2 : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80426fd6c42a44a1935866cb82eec27c : SUCCESS in 15m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9bfb603a90c4de3b6ab3962ebab8240 : SUCCESS in 36m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c0747071c5a48c48e017a0d15d8234d : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4de3a35bd2a59d483a70096bd5977e90f5bf9ca7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 23:54:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c475f3332e4540bff13e4d4a439400\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42d76a0cfb074b28842bb6e4925e6d29 : FAILURE in 1h 48m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5e7ac86e39cd2bef9ca55713fa37c80be8c3cfc4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 00:15:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"54da4d607082c8cbca623fa19fd3fa2071df3476","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 00:50:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9845f0bddcc01fbc4d0790e1a7947e291b83f2bf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 01:03:41.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"a50868dd8f096987d6a3e889ecd79004d1e6cc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 02:02:49.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b4ddd2e4225481885f122ee02b93d4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81554c6d3f074e1db516b336d46c7aa2 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/0377c97f220c46f6b752e61569540ca9 : SUCCESS in 1h 02m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ecf2cef3b564aebaa103be5a2319a87 : SUCCESS in 58m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa5899b5731641ed82863e3fdbf774ec : SUCCESS in 42m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c11618dfe2f14a7da1b1a5439a6e19c6 : SUCCESS in 52m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b70b529af35d4338b8639589c4c83d39 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca61e77c41e94770b9791b858ddf9935 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/462051476b9744f69064dab25b49866b : SUCCESS in 23m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a347ab1d713b46c4bd19d9a5e245571a : FAILURE in 3m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a33c9122919a44e3a951f903de1c5c24 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d14431385eec4569830af15010684b32 : SUCCESS in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e1bc8e1c177a4543b9fe2df887b8c22b : SUCCESS in 17m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b042e0ba9eca49a691c7192c33c8414e : SUCCESS in 27m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86e9e5deeb3e464c9e98b5efc24917aa : SUCCESS in 19m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3db28a4c704a4bb7942f77188ff652b9 : SUCCESS in 10m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02383a71b0de4916ae488e3f718dd2ce : SUCCESS in 15m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a13945ec0d954a428818d3760db17f16 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7c44f9792a3f4c898d7cc176b880f984 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d5a4bb90163843e681f40d8e39a893b9 : SUCCESS in 9m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/416e2892bd7e462288fee33c2e549936 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6448aad09c39429b8ba2f11f84adffef : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e61360da2ceb419cba747e1f7f6188f0 : SUCCESS in 4m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/acf4432f93594ae3a7ea5382226a9d2c : SUCCESS in 49m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2da814801354951b63426d45313937d : SUCCESS in 48m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6fda31014dbd40d2a2a840b556f201b1 : SUCCESS in 26m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/083d5e141f14431ea6bd0e72e3e9898c : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebc7f54fdfd45a8a4a7a37a6d9d3bd6 : FAILURE in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a64609c3ae949619cffd544ba955502 : SUCCESS in 20m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b3cb9a441814ba48c09acbce503df9d : SUCCESS in 50m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2f062d0ab3c427e8e2cb9994d55b658 : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":4},{"id":"eb73394a3e832a8b8d38bfe466d042865ce4e6ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 02:35:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ea2df46d4e28b796bf7e361b9d9a9b7e4c505852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:05:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d90e0a44ff4171b7b9f73a5981168a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df11ca36f8e482bbec265b8f5b39eb2 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4353ae4b00a349809a084893c3b17b3d : SUCCESS in 57m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b97ccda1664648529b93e6c426a612ac : SUCCESS in 37m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05ab4c3662c94cb9b8e6ba1818948d07 : SUCCESS in 55m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bd05a23e7fe400382a506f9e11d8713 : SUCCESS in 1h 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ca582b2b3b5410ab07c7cd21cf918bf : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72012a8e793042c89e4db9cb53fe43ff : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adb00730c5ad4d3bbf4cb97df5589c75 : SUCCESS in 27m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ec976ffd9fb473ba09d4c655de4dfda : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c609b94e34a4d97bee2a84863ef6838 : SUCCESS in 20m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e4c7d1ac94f54b3d9f65ac38acd900ff : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9362e6b027ab4207b92aba5bcaf038f7 : SUCCESS in 16m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7291cfbb140e497b8b64b26455a39d74 : SUCCESS in 22m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d664adae84e940a49ff16d1412e8019c : SUCCESS in 27m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5352ca06ef914dc8a9307516de29d696 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19963d4a0d54463f80ff4b2371eb9da2 : SUCCESS in 16m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fae19f28eca74bc8bf567af8f672ebf9 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/af1273e3e18a45f7ba2a9ade54daa069 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4b84b380fe3d4f8aa322ef72a5df72e8 : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8553f1fef6a34087b2c6b280d3ee2b88 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fd31867315a43cd8439137ed5205377 : SUCCESS in 13m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/839b427e64b8482d8246a7dcd8230458 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b11fce51e28a4fdf94c2c52c80816915 : SUCCESS in 51m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e34ee2d814354f3dae0c3cfbedb6f07b : SUCCESS in 48m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a35e5db4ee434ad5be727dbc1ef7d082 : SUCCESS in 35m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f490fca4adbb4aaba04ec1d10f2b3dc9 : SUCCESS in 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b558f3ca9545ccb878334b39146eea : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe5981a9133f4f948937c22d89d426ae : SUCCESS in 17m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc8ff8d7e7c4437288154acfd887cf7f : SUCCESS in 53m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/506ab252af2344cca921eb2e46785ed2 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":5},{"id":"8e22640fb5efab0778410f324a7f354016d7faa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 04:30:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb418d5a8bbc43a9a151fc50065271b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19cdd93be2ad4edfaa53ec57b8aa7d6d : SUCCESS in 1h 48m 05s","accounts_in_message":[],"_revision_number":5},{"id":"cf5802b97f55deb505e1966333377fdc7fb4b7e5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-19 20:34:19.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"69bef3c31d98401497782084e131da2b33a691e0","revisions":{"ffffd4897b0ed94b34f32a4805d161fbb259d606":{"kind":"REWORK","_number":1,"created":"2024-12-11 19:13:51.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/61/937561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/937561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/937561/1"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 19:13:37.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 19:13:37.000000000","tz":-480},"subject":"Add alternative config to manager","message":"Add alternative config to manager\n\nChange-Id: I0e62e2865df1937f0296f479e5e6c6337880785a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffffd4897b0ed94b34f32a4805d161fbb259d606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffffd4897b0ed94b34f32a4805d161fbb259d606"}]},"branch":"refs/heads/master"},"7645633fca048c2efafc94d2fcc1de07894624ab":{"kind":"REWORK","_number":2,"created":"2024-12-18 21:52:06.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/61/937561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/937561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/937561/2"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 19:13:37.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-18 21:51:54.000000000","tz":-480},"subject":"Add alternative config to manager","message":"Add alternative config to manager\n\nChange-Id: I0e62e2865df1937f0296f479e5e6c6337880785a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7645633fca048c2efafc94d2fcc1de07894624ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7645633fca048c2efafc94d2fcc1de07894624ab"}]},"branch":"refs/heads/master"},"c1a4971689e4a0f428c5117820aba59d33994120":{"kind":"REWORK","_number":3,"created":"2024-12-19 00:15:36.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/61/937561/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/937561/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/937561/3"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 19:13:37.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 00:15:28.000000000","tz":-480},"subject":"Add alternative config to manager","message":"Add alternative config to manager\n\nChange-Id: I0e62e2865df1937f0296f479e5e6c6337880785a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1a4971689e4a0f428c5117820aba59d33994120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1a4971689e4a0f428c5117820aba59d33994120"}]},"branch":"refs/heads/master"},"0d17725d6d7532d4f933f0d793b724639a2aaca9":{"kind":"REWORK","_number":4,"created":"2024-12-19 00:50:00.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/61/937561/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/937561/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/937561/4"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 19:13:37.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 00:49:51.000000000","tz":-480},"subject":"Add alternative config to manager","message":"Add alternative config to manager\n\nChange-Id: I0e62e2865df1937f0296f479e5e6c6337880785a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d17725d6d7532d4f933f0d793b724639a2aaca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d17725d6d7532d4f933f0d793b724639a2aaca9"}]},"branch":"refs/heads/master"},"69bef3c31d98401497782084e131da2b33a691e0":{"kind":"REWORK","_number":5,"created":"2024-12-19 02:35:16.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/61/937561/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/937561/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/937561/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/937561/5"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-11 19:13:37.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 02:35:05.000000000","tz":-480},"subject":"Add alternative config to manager","message":"Add alternative config to manager\n\nChange-Id: I0e62e2865df1937f0296f479e5e6c6337880785a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69bef3c31d98401497782084e131da2b33a691e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69bef3c31d98401497782084e131da2b33a691e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
