)]}'
{"id":"openstack%2Fswift~937632","triplet_id":"openstack%2Fswift~master~I22674f2e887bdeeffe325efd2898fb90faa4235f","project":"openstack/swift","branch":"master","topic":"add-object-updater-last-stat","attention_set":{"37271":{"account":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"last_update":"2024-12-20 16:46:13.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-19 23:18:43.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-19 21:45:35.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I22674f2e887bdeeffe325efd2898fb90faa4235f","subject":"Add object_updater_last stat","status":"MERGED","created":"2024-12-12 14:55:17.000000000","updated":"2024-12-20 16:46:13.000000000","submitted":"2024-12-19 23:18:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"937632-add-object-updater-last-stat","meta_rev_id":"89307a59f17f9aede5bcc0fd70ba9d9aa81ac4f9","_number":937632,"virtual_id_number":937632,"owner":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 23:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-19 21:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-19 21:45:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 16:07:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-13 17:06:05.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-12-19 00:51:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-12-19 21:45:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f8762ff124f0e65380c5df73cf05107747ca3b4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-12 14:55:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88b37a2580f7c6b66422e2d5dd3d4f4b97549a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 16:07:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94b0aca106234a0c94aa45a4140740e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fc38f288424c4290e608b708d6434a : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/465d6c5de88749ffa24ab4e10ec17689 : SUCCESS in 35m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84c3da35c22947bbb7b94f413e9f294e : SUCCESS in 34m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d1fc98c43c241c18345a198ba3c3ce8 : SUCCESS in 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5901beb846f48efa6b0fc1b1dfae3bb : SUCCESS in 52m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4d3a6adfde44752b3579fae645fb69f : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8ffa7bde5a44bd79112d89de5aaff6b : SUCCESS in 30m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d719aa150d24458b8fbede8b69d3c692 : SUCCESS in 19m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2cf2d2559aa7423e9503e58c8cd735b6 : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04088f1e3b314bd69104d604365944e2 : FAILURE in 21m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/576ea128e9024d74a5940daabbc68bed : FAILURE in 21m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d52c474b3884652824637d4af2cc6d6 : FAILURE in 30m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1b5de3f2a4904815941fffff05b9466e : FAILURE in 25m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/caf94c9406474cce843f2f054ec5391b : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f3fd5a71f68407d9e1e086dda5feb88 : SUCCESS in 29m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e18a91369e54499a0f3bc4f761ad0e1 : SUCCESS in 18m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8df59776852e4d42970349cfe2885bcd : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/15f9844bf7754a1695b1f9080842dd14 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0bb61163e96045a983ebb4a7fe87bc1b : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75d6b19395cd4fa3a7b94b512a939555 : SUCCESS in 8m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b3e25dd94d6a457ba7b576a3dd441273 : SUCCESS in 8m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6056b9f9920546fcbef76836d0e6320a : SUCCESS in 9m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/86ea032182e84a1a9b2fa03ddd277772 : SUCCESS in 52m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16de6f5b62ba41628883af852b5c35ca : SUCCESS in 46m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a67f72f35a184282ae84857978cb2e0d : SUCCESS in 56m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe81f9980594440b81a06f50718af070 : FAILURE in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4456d04ec96d49829d94ba3686a724e5 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c21376bbf664ec4b250725a600ee7f8 : SUCCESS in 18m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf05ef6bd9004bed95f8455c3cfdb267 : SUCCESS in 37m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb34eafd0aa14a39919db1de6531b62d : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ae5e3959b1dfc7c61c3fb68485528c3e3221efdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 19:44:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/156b76aae64447be832b9513c4daa9b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46f7255b6f5c4d14ae9d7677d83d5fda : FAILURE in 1h 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"241e7656b23c024c0b6ad023ce18b0b1f95acfab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-13 17:06:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fbdaf852b4cfa22a8f68a24fc452ce62f608511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 18:54:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ec7b48257f127d9535a73d63f5b9c8084dc1a66c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 19:20:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"12fe278bfccc48b8f3de88bc79843577d1d53041","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-13 19:22:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c75c4d5e321754e9ee670ba8b7482c100409093b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 20:30:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7aa1fbcde54c2984526f174e557370\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad08d90bd504dec8303ccf4018fa944 : SUCCESS in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/9b6361ac811c4d479b18b357abab25b5 : SUCCESS in 1h 03m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/248ed149cc834101bbf3d2ba70b20b65 : SUCCESS in 53m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7912d89c68b4290bc6fb95400e7fea6 : SUCCESS in 43m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45c2ba84c0de4a5fbb17992d57a4dca2 : SUCCESS in 1h 04m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78c795b2d09f401a86a30f4a26f0747a : SUCCESS in 11m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cf25bff9b2a485c89f372a68667a12c : SUCCESS in 23m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6fb7090c879490286107d5e847d9b97 : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/907a07e6c69241399293125fcf154f91 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/793df0913c3048318a3afdf285e8e8f1 : SUCCESS in 18m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c27bae1ccc6483087c6f1db97e85a9b : SUCCESS in 19m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/723ed3bc30f84c60988c731f0a905f4c : SUCCESS in 24m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c4baff1c32c64743a5217aa98d07b25d : SUCCESS in 26m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32adb23c43ea4908817a13248ee3085a : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29ceac86204a4972a28cfbc6e6f9e2c5 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9467597c237b4458993e4b4d91fea2f1 : SUCCESS in 16m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3eb6e3c61fef47ddba60e99d61736349 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6db2c6ad7fc940d6b52885de70562ecd : SUCCESS in 17m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/991b945b366847a5bb40e8353d82ee6f : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d18128ec218d4d289df3f6ac7c09774b : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/619f1c62e2194ae9ae74ecf7ba729323 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/131c5d6492e14718bd53d0db393658e5 : SUCCESS in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/60f7cec8428842d8bbf85c54b06d5a03 : SUCCESS in 57m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c80b5b307ab846b89289fd723e24c7cc : SUCCESS in 56m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e1168dceba9468386c613ee06595502 : SUCCESS in 37m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8eb5c6110674dbcab5dfbf517cf2e01 : SUCCESS in 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9107e773a5c34a93bf309612cf6eac4b : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9794c45a88b6455b9fcf96f1d2643d4c : SUCCESS in 24m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19e55ff30ff64157b78c939dfaec0c69 : SUCCESS in 35m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8c06cf812494cec8f5c79db53e49893 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":3},{"id":"59381721c1dd48e4eb1f1d36746f143f18e62984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 21:05:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/034273724dea47fe8716c79b47317d44\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0c65726da7d49679d60a5ec65a7a490 : SUCCESS in 1h 42m 03s","accounts_in_message":[],"_revision_number":3},{"id":"573477285c841c206d05c16ad70bfa9c999ea2c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-16 22:56:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"91901b117449dade5cc934b738f19ad2fc1ec31a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 00:12:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ad9afd7ced547f3b17f9e8bd9dd1308\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8056f0ee0fe4310ba78bef87604e42b : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/5abad18cfed64da49e6ba0c60fa4d2d3 : SUCCESS in 1h 05m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/838f42c2294a4243b7df0bc437b6bae7 : SUCCESS in 1h 06m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/abf6edcd988840419352c73f4a2c133c : SUCCESS in 1h 00m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fb5bff6a92a44a1a29c7d459fc7cf8a : SUCCESS in 1h 06m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9c78010f4714e71b59f51b7aadbae2a : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8a4cf54bdf74e2086b2e45860bde7d4 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0431604b23554faba705d4b35eba952f : SUCCESS in 19m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dab78fe088704e6886d607f07dd09153 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d732d162fc814f4bb2a7d456b05728ce : SUCCESS in 17m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db99a4690289455cb27f9d8b68662fb9 : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72652d15058b4dd1b266c195697bee3e : SUCCESS in 17m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/75475a8e9e174751a62ff03e6b408e23 : SUCCESS in 32m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5de66a46bdff422093c6a070f67a5f5f : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3551caaeb3b848c292c137989118fa5f : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/845669b45abb4ef59934290a981223d4 : SUCCESS in 17m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/49f34bb6f7bc46cdbdc37ba159a2b630 : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bb14bd7475184cc0bb2d73952278d8b2 : SUCCESS in 13m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b9c6bb2eb154aea8020560861237af2 : SUCCESS in 13m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54c87bf78e2945a081ea1203778a9165 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/828f25e92cc142c2b3807a3e4668629b : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c12dbefef94c421fa0cd68399dfb0000 : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/56e4352917164035ae86426c874b0b5f : POST_FAILURE in 50m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77a6d8b1f6ec4ee084ad8d453bba8283 : SUCCESS in 32m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba86dea8c780431face98b8d5a282b4b : POST_FAILURE in 51m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40da0d73763e42afab65cd723ff326a6 : SUCCESS in 22m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53cd23e15fae44e3a123ca4b93225e30 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5c108e359b549ed992358ec3769b7e4 : SUCCESS in 17m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c6e5d29962841e8b06263c3a7eaa208 : SUCCESS in 50m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d563258272bd46dd88d43fa25178ad08 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":4},{"id":"eb2ee1fc35c68ca1ceea222bc212c10d428017d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 00:39:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58505272e4a9453dae487a914e224189\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29b82dce3eb64bcf87a4e6beafbf12ec : SUCCESS in 1h 41m 12s","accounts_in_message":[],"_revision_number":4},{"id":"b2faeab8ccee5ec9cf622fcfc7280bbebb3ce44e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 00:51:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3792751abd5a64eee9fa8e739e249f01ea413a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 02:20:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a83ba9ef56474190e3e00d3f9a5973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386ff6f2a56543a4ac0f3a1289bab2a2 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/593fc6c61c7545838c1b707cfa717ae4 : SUCCESS in 1h 01m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b670ad59f4748b3ae7ceb62d555c168 : SUCCESS in 1h 05m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c24da11665a7482c80b874996c66888a : SUCCESS in 55m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfc79098afdb4551a17ee5dfc874c2e4 : SUCCESS in 1h 11m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df084454e5294b08b995e6446de27862 : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e76357396b444da9283f129b6e1a20d : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ddc58ecd3f044d8a933eab9c0d855d3 : SUCCESS in 19m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b699a4a4c80e4f488b645dfefee06e67 : FAILURE in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03fceaa9b18148f9bee36003dec671fa : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7e6d4a587da41bca0080b9b5c7b34f9 : SUCCESS in 15m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59d331ab2d3a4851af8c0c6a157fedfd : SUCCESS in 13m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dccfddd6a35a4be0ac9fe7fec425860b : SUCCESS in 33m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/990784c03ff24190846435b05ddb2c4e : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42cb41dd577e495096ba765581a0d8f3 : SUCCESS in 10m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9eeee65417a84c139ba2367db9ac362b : SUCCESS in 19m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5b063ef7d6464f4c9bcdd8b85d696b7d : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/68b75bec01d34e738c8e9592946f8354 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/652f573c864c4040a55d0c7d22c25fc8 : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0a93450a3fc401496553e793143f4ae : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1fbdb9ea06c54fc391615f02e211b517 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7233c5984e3444b1873d5f8e2f2df54f : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/949d795ecaab40938afc8a1e45b28ad2 : SUCCESS in 54m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b5b55abfe2c4dbb9491f83d806ca2e9 : SUCCESS in 49m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bc6f0f6ca62431080725e8a078eba5e : SUCCESS in 52m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5274623207c842e891d758b593ac51d4 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab622631419464aa1e4d1f5bc5e97f0 : SUCCESS in 2m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e5cef493aa845ffa8d91136835b15ab : SUCCESS in 27m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1975aa3ac954681b9bafda14adb7bc6 : SUCCESS in 38m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcf7268889c045b9ad7e268bc5624ea4 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":4},{"id":"02957c616d70cbdba8e4c1f6240ee39aef8f43c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 03:00:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e80ce0d61804f3a891aad277c5c3a9e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3b9bb2af820f40bbacedba9e9afdc25d : SUCCESS in 2h 07m 16s","accounts_in_message":[],"_revision_number":4},{"id":"0720460b820813be040f8acc676e734b3c26009c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 06:11:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e42e49670270067ce49ae8727c3c45668143fa4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 09:28:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5548f566e13be73c643656592752ffd584a688bb","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 09:28:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"22e2190ed39ce4fd8d6256b421bed1183e0c2f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:55:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d100da76da447dd85cfb13891933910\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c1892007e0400ba342c486d975d025 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/82a2ac1189134b05b2c441f17d81c7a0 : SUCCESS in 1h 02m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7fb5f5a6305c40baac089f2cb4d003d6 : SUCCESS in 44m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3acb9118398146d1b0de1bf82d01bb0c : SUCCESS in 1h 01m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07efdd309e3a4157850d996828cc9c8c : SUCCESS in 55m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfc09f2e3021405299613f7bcab7627a : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba8e2ae283014ff3addd78ca317d21bc : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86926db31ad94523b2a76733914dcac8 : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/447a96ef73d44516a42dea6838d16e86 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5df97e24d5c544ca8a5398861e656f76 : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c483635aba64c13b64730e285e8cf7a : SUCCESS in 13m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a45dbe38268d475e95ff5cc2893aec42 : SUCCESS in 18m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7bb0f92e152b4df3a712711743704f3d : SUCCESS in 29m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19471ca28f434d8b88cac793897de070 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64185348e6a8447594a0082701e69fb2 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d5138e16659d4340b3eb38730784aa9f : SUCCESS in 17m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f837bbea58c84253913f085d85b33ef3 : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/553fa454f9364e11997a38affc532b5e : SUCCESS in 12m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6f77aea4d73f46dcaa0ca3cbb5215079 : SUCCESS in 13m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/056923bb912e471dbd828d824f8e071e : SUCCESS in 7m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d79149daecb430f8ba327f2055941c8 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f98af3f1f6bc44e1a2acfe27d39ddb14 : SUCCESS in 6m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2256222186a04207a8ef00fb98df3f7b : SUCCESS in 51m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/486bb7f0e2b640ef845fd47caf76745c : SUCCESS in 54m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71e1d845314e4f6796d3ca76cc233013 : SUCCESS in 38m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f021134b959f441a895f8cd14e01b7f3 : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1a9cce60b3e4900abe8dcf3e66f408c : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f40e11ae04d44db2bdd0781dd40ba03b : SUCCESS in 18m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb04ed7d011e49b18ff75cce805a4238 : SUCCESS in 1h 03m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d069f4e74014d26a89014b2b83c2da5 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":5},{"id":"efd9942c4fc90165c4f00a3d4430488d48b3ff75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:09:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6fcd1fdaba248acaa1cfa657830d1fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c38c7c9cc914761a73c6a1cf81d156e : SUCCESS in 1h 39m 07s","accounts_in_message":[],"_revision_number":5},{"id":"e9dd49c4e92b08105a873ef507d61c0ec2a25da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"date":"2024-12-19 19:11:04.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4222371c14fe03a2762a4c6b9e4271255ed7a007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 20:30:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b367c52338c4fa69ebc72c00d042a5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937092a00f064dfca8a27fbddf989225 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b8948e9d4dc34d4ba10c3be38e7b8e71 : SUCCESS in 58m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11727e94deda44119acb35c10d1d38c3 : SUCCESS in 57m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0eeb63e63fe8444d848bfd2e40e3cbf6 : SUCCESS in 1h 03m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/134ad226afeb49e1af284f33157e0417 : SUCCESS in 1h 12m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/450ca07dd59c4fed8ab80d1931dda642 : SUCCESS in 23m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d99c6a7f7b94e41a52f990b12a0b80c : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f1c40ffe2ef4b9cad15281b245b1578 : SUCCESS in 24m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33997fa791c54247a4002c8b85b4af17 : FAILURE in 3m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/13d1e1f7b64d40dd98ea2f768c7ecf5b : SUCCESS in 21m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/699af71ec57540b2a2a83b9babeac29f : SUCCESS in 19m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c8cc7213af2844bd857fd8e6e3ce1205 : SUCCESS in 22m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/18569275dc3a47e5a90b9bde766d7de5 : SUCCESS in 22m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/413a91b4ceed40beace03b8ca2c38708 : SUCCESS in 10m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/93cee0f3b08b4cd0b79d13b98c4b4255 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/30c0e5c02e934a75a1b84ad213238131 : SUCCESS in 21m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/09e636d7c28b418c91378bfce03fd14d : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85b0915996a347169ac2340d1901299d : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/714504e2107d4b079b3beae480d2629c : SUCCESS in 10m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0dfd8785c26e434f93a1299ab1bc0b07 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d1003acbc3744eb8cd6a7827befe57d : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0b3a92d37574efcbe70645a4cdb5a2a : SUCCESS in 4m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68b2a0ec2dfe45b3abfdeb69a8aef78a : SUCCESS in 38m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2259788516ab4a86b67932c36f3ea8bb : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/94a5e1231cd94dabbe091e67322e98b3 : SUCCESS in 51m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92eff3197fcf44229022f9d5a2c67b33 : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf1654b66bb4a87b4711ff3ff9e4ed8 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f211687b5dad4c6182aa29a27246d36d : SUCCESS in 21m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/263823b5d6eb4b22b47030e7e27df012 : SUCCESS in 47m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b010c1b1afd840c79ac94c403e24f1b3 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":6},{"id":"ff671cfb837104c1cf507f87539f9e1abbed8907","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 21:23:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce5c6bfbd3847bfb2e12a2fa7af0e7c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a2fdff1704a492e9e4f157011b851f9 : SUCCESS in 2h 09m 44s","accounts_in_message":[],"_revision_number":6},{"id":"eccdc2d83d0222f6f4a75ad58d365647b959b150","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 21:45:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ffd80f6512a90b73f8fb7dbfd2a3702f47a0fd9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 21:46:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dbf906dbbc92e9e3a4c8b14609ec2d4619954026","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 23:18:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87eec0cfe2c94206a97a64254307fb60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd882710c5ff4bd180551355ab1e64df : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/8b7910cffc334571b841c318728ae819 : SUCCESS in 1h 19m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7af4174cb0bc44339383527016c2b0d7 : SUCCESS in 58m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00e7d1e6bda84412b67cb2ba8d148a4b : SUCCESS in 45m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e7a33b341fe4cecba65d6669d43da0f : SUCCESS in 50m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/eaeb42f68d7a4b15bbc3d468d0bb48d3 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d161c12920842e0b474d16820a4a01a : SUCCESS in 21m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/258acd56b6bf45e9a54c2a5f80a75590 : SUCCESS in 21m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c613fa6f3004f4a963900dc97393213 : SUCCESS in 20m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/80801fdd0ba740f49c98f71aef4da2ba : SUCCESS in 22m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/baf97c050fee41a49d073caa86333683 : SUCCESS in 21m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c92ee978e2f4cae92d53a20428ddc25 : SUCCESS in 19m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b387eb77f3b4e25b9d7d051a7f7f7e0 : SUCCESS in 20m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/639986032ab54b8896feeec907273ef6 : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/369a90c8a8344a16905a25e963ae45bf : SUCCESS in 16m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f7b5ab7d455149459a17d659731e6592 : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d22bd01b177f4502b0236beeaa7c2bd2 : SUCCESS in 5m 31s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b2a4d25f520459aa6887c5a8bfd4b70 : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf97c0c4e7674cfd9d6e26b71ddc8700 : SUCCESS in 55m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81aba70692a64fa58f9abb8be12682b6 : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b1d04622d174c41a72bbaf593482b85 : SUCCESS in 1h 00m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/272b6c5fa82147a3bb60e509fb5b8e1b : SUCCESS in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0567be9ca0774d05b8d796e55ef3cc90 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15ca5d71472a408ab9ca2a857e0183ab : SUCCESS in 17m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97d970e1d90c4ef8a44e06a37bf2c831 : SUCCESS in 55m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e84096f9ac94046984cbab149e6bcf6 : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":6},{"id":"ec39051e4350f6fdf145a68bcdc15227764235ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 23:18:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"be6044a11bbf0efd4f2672598ef984e350d811a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 23:20:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b5fc0edbc064760baa45d0f478d4c35\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243ddce560dd4eef854b290dd6b9f424 : SUCCESS in 1m 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b95373738dac4fa88f99062ec9b97846 : SUCCESS in 1m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"89307a59f17f9aede5bcc0fd70ba9d9aa81ac4f9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-20 16:46:13.000000000","message":"Patch Set 6:\n\n(12 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5281af5cf25f1a88b23441d1c45f0d86735395db","revisions":{"05e70117581829ef22d5a9a60b24e4e4b20e8ec2":{"kind":"REWORK","_number":1,"created":"2024-12-12 14:55:17.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/32/937632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/937632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/937632/1"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"subject":"Add object_updater_last stat","message":"Add object_updater_last stat\n\nChange-Id: I22674f2e887bdeeffe325efd2898fb90faa4235f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05e70117581829ef22d5a9a60b24e4e4b20e8ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05e70117581829ef22d5a9a60b24e4e4b20e8ec2"}]},"branch":"refs/heads/master"},"6a7d22febf56e8f46a6076e3c5fe577a78f02761":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-13 18:54:46.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/32/937632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/937632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/937632/2"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"committer":{"name":"Chinemerem Chigbo","email":"cchigbo@nvidia.com","date":"2024-12-13 18:54:46.000000000","tz":0},"subject":"Add object_updater_last stat","message":"Add object_updater_last stat\n\nChange-Id: I22674f2e887bdeeffe325efd2898fb90faa4235f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a7d22febf56e8f46a6076e3c5fe577a78f02761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a7d22febf56e8f46a6076e3c5fe577a78f02761"}]},"branch":"refs/heads/master","description":"Rebase"},"5da44949e16a613d8a8660bf6aee7e3a23c7654e":{"kind":"REWORK","_number":3,"created":"2024-12-13 19:20:59.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/32/937632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/937632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/937632/3"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-13 19:20:47.000000000","tz":-480},"subject":"Add object_updater_last stat","message":"Add object_updater_last stat\n\nChange-Id: I22674f2e887bdeeffe325efd2898fb90faa4235f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5da44949e16a613d8a8660bf6aee7e3a23c7654e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5da44949e16a613d8a8660bf6aee7e3a23c7654e"}]},"branch":"refs/heads/master"},"753f17d76b708aa252503fc3146303031141b83e":{"kind":"REWORK","_number":4,"created":"2024-12-16 22:56:20.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/32/937632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/937632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/937632/4"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-16 22:56:07.000000000","tz":-480},"subject":"Add object_updater_last stat","message":"Add object_updater_last stat\n\nChange-Id: I22674f2e887bdeeffe325efd2898fb90faa4235f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/753f17d76b708aa252503fc3146303031141b83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/753f17d76b708aa252503fc3146303031141b83e"}]},"branch":"refs/heads/master"},"dc10a47f8f405c67161040ffdf61c9b0e4195de3":{"kind":"REWORK","_number":5,"created":"2024-12-19 09:28:19.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/32/937632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/937632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/937632/5"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 09:28:06.000000000","tz":-480},"subject":"Add object_updater_last stat","message":"Add object_updater_last stat\n\nChange-Id: I22674f2e887bdeeffe325efd2898fb90faa4235f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc10a47f8f405c67161040ffdf61c9b0e4195de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc10a47f8f405c67161040ffdf61c9b0e4195de3"}]},"branch":"refs/heads/master"},"5281af5cf25f1a88b23441d1c45f0d86735395db":{"kind":"REWORK","_number":6,"created":"2024-12-19 19:11:04.000000000","uploader":{"_account_id":37271,"name":"Chinemerem Chigbo","display_name":"Chinemerem","email":"cchigbo@nvidia.com","username":"chinemerem"},"ref":"refs/changes/32/937632/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/937632/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/937632/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/937632/6"}}},"commit":{"parents":[{"commit":"ea06ed44940afb5352bc715d62dba1c742e2cd5e","subject":"Merge \"Aggregate per-disk recon stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea06ed44940afb5352bc715d62dba1c742e2cd5e"}]}],"author":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-12 14:54:58.000000000","tz":-480},"committer":{"name":"Chinemerem","email":"cchigbo@nvidia.com","date":"2024-12-19 19:10:52.000000000","tz":-480},"subject":"Add object_updater_last stat","message":"Add object_updater_last stat\n\nChange-Id: I22674f2e887bdeeffe325efd2898fb90faa4235f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5281af5cf25f1a88b23441d1c45f0d86735395db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5281af5cf25f1a88b23441d1c45f0d86735395db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
