)]}'
{"id":"openstack%2Fswift~937653","triplet_id":"openstack%2Fswift~master~I4f015d711e4d602cd04f04548747a0323a657b49","project":"openstack/swift","branch":"master","topic":"fix-oldest-async-manager-run-forever-recon","hashtags":[],"change_id":"I4f015d711e4d602cd04f04548747a0323a657b49","subject":"sq? maybe better assertions/comments","status":"ABANDONED","created":"2024-12-12 21:27:39.000000000","updated":"2024-12-19 17:07:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1024019fb046365acb727efaee5557d437df361","_number":937653,"virtual_id_number":937653,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-12 22:28:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-12 22:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0baf00853adda7a688bea5fe2bbe5a18315997d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-12 21:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a376d2e6d040f371d09459874712ecdbf99be1f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 22:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e52aa3e1b89b4c48b28d8a19075a2a03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a99c97c7f74f0580945423454bdf7f : SUCCESS in 9m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af27e30250fa45b0a8377efc9e7c5df2 : SUCCESS in 31m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffd84f5a72d34e0d9683548495279b5e : SUCCESS in 51m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d17ac4ad0b6644b9ae68b6e99432704d : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a6f9ef1a56fd6b038be653951621bebb8fbd0905","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 22:53:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d55f3ad811545439c6dab9ac0c11008\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ffc7b754f2e841af8bbabec47a4a1acb : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c1024019fb046365acb727efaee5557d437df361","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-19 17:07:46.000000000","message":"Abandoned\n\nlooks like this got squashed 937267: Aggregate per-disk recon stats | https://review.opendev.org/c/openstack/swift/+/937267","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64e47bb0c29cd56ea1ac6d6d3344edae72d9e4bf","revisions":{"64e47bb0c29cd56ea1ac6d6d3344edae72d9e4bf":{"kind":"REWORK","_number":1,"created":"2024-12-12 21:27:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/53/937653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/937653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/937653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/937653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/937653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/937653/1"}}},"commit":{"parents":[{"commit":"9f423d1fbd5503caed504715a09800963f9fc611","subject":"Aggregate per-disk recon stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f423d1fbd5503caed504715a09800963f9fc611"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-12 21:27:16.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-12 21:27:16.000000000","tz":-360},"subject":"sq? maybe better assertions/comments","message":"sq? maybe better assertions/comments\n\nChange-Id: I4f015d711e4d602cd04f04548747a0323a657b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e47bb0c29cd56ea1ac6d6d3344edae72d9e4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e47bb0c29cd56ea1ac6d6d3344edae72d9e4bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
