)]}'
{"id":"openstack%2Fswift~937659","triplet_id":"openstack%2Fswift~master~I38889dca04f846999d8669228a2794e7895d5c01","project":"openstack/swift","branch":"master","topic":"fix-oldest-async-manager-run-forever-recon","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-19 17:07:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I38889dca04f846999d8669228a2794e7895d5c01","subject":"sq? maybe don\u0027t use self.dev_path as alias for self.sda1","status":"ABANDONED","created":"2024-12-12 23:30:49.000000000","updated":"2024-12-19 17:07:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76d8b6b99ef7884bb76097dc57a7b7bdafec8abe","_number":937659,"virtual_id_number":937659,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-13 00:29:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 00:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4deb4f3d9d188eb3b02d603e06bb82d66b9bd572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-12 23:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79e138848f2e4e41f52c5464845c001244b2603f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 00:29:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47dd7b004814a4b9bac952559927c65\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14c70200d6f402cbca901d21b27e644 : SUCCESS in 9m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35a213e81da74480978513c2f97cf316 : SUCCESS in 55m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c67098d777b248e78bc706bc1c48bd4f : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb077a5d45214472a50076f259383a82 : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de8bc5ca46a34e9a93bce17b65db157f : SUCCESS in 18m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2b5d1ba825147e8852e24ee42fd70fa : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/054f0047f0f8471bbada7cb152f292c1 : FAILURE in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb68f5441ca479fa1e9859473fb4b8e : FAILURE in 13m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29aafc01d8a24ff4a3b6e851a6735e89 : FAILURE in 15m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1a4bc43b009c4d9cb50ca6aa8363c511 : FAILURE in 19m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c48a1b85164a4a12b31fc1b8889f4ccf : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a06734b5aefb4c5e8e854631581e2f55 : SUCCESS in 13m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f84195507eb948d9842979234b03d7c1 : SUCCESS in 16m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/01f2853100044ebc95ebac8dd50c55f9 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f91930eec6954d628f1c1be690393406 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/66d2aefc1b804430bbb6c7222d9a89e2 : SUCCESS in 14m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbf54e09201d485bbcf9a18841355a88 : SUCCESS in 32m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f02d11e65b8404196be48d26450d534 : SUCCESS in 32m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80640d8e047644828ef89c8ae1bdb954 : FAILURE in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459914344c82400688d1b3b896e64e67 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab4f12f4d63d49618e65d89be4e6ebe9 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e02fe0cf53261c3f9eeaecde7e5cdf9c0f8a2fc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 01:39:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97390f1f745f4d54a0948a9a65d670a4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee18ca6cd4d0450d994bbc77a3759f40 : FAILURE in 2h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"76d8b6b99ef7884bb76097dc57a7b7bdafec8abe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-19 17:07:29.000000000","message":"Abandoned\n\nlooks like this got squashed 937267: Aggregate per-disk recon stats | https://review.opendev.org/c/openstack/swift/+/937267","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a7b249ac3ae079bd039710e91fd720c629bbe1e","revisions":{"3a7b249ac3ae079bd039710e91fd720c629bbe1e":{"kind":"REWORK","_number":1,"created":"2024-12-12 23:30:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/59/937659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/937659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/937659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/937659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/937659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/937659/1"}}},"commit":{"parents":[{"commit":"64e47bb0c29cd56ea1ac6d6d3344edae72d9e4bf","subject":"sq? maybe better assertions/comments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64e47bb0c29cd56ea1ac6d6d3344edae72d9e4bf"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-12 23:19:11.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-12 23:19:11.000000000","tz":-360},"subject":"sq? maybe don\u0027t use self.dev_path as alias for self.sda1","message":"sq? maybe don\u0027t use self.dev_path as alias for self.sda1\n\nChange-Id: I38889dca04f846999d8669228a2794e7895d5c01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7b249ac3ae079bd039710e91fd720c629bbe1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7b249ac3ae079bd039710e91fd720c629bbe1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
