)]}'
{"id":"openstack%2Fswift~937660","triplet_id":"openstack%2Fswift~master~I73c71efd52d7262cd4fac379a846ad485d1d0445","project":"openstack/swift","branch":"master","topic":"fix-oldest-async-manager-run-forever-recon","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-19 17:06:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I73c71efd52d7262cd4fac379a846ad485d1d0445","subject":"sq? some more unittests","status":"ABANDONED","created":"2024-12-12 23:30:49.000000000","updated":"2024-12-19 17:06:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"598bfb4e04e10aa411776dc2b65e65916d4cb460","_number":937660,"virtual_id_number":937660,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-13 00:48:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 00:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f636f48758382b7970afba417ee437073f7266c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-12 23:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57ecafa8308a72518c5a374f5dc04aacd99616a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 00:48:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/256317bd59d048b19148a8aea432a7e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13bec9b984b942b58623098955d2b0e6 : SUCCESS in 8m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9d4ad24139c4ea989fb90851507df23 : SUCCESS in 1h 11m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/494ab754755f414088a6dcf97dffc39c : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d286afa375644038840a98beffdd8ef : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ca4a6092afa4aeaa4e84580bc67889b : SUCCESS in 17m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3257c5fb180c496cbc8e9ccf23eb9137 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9bae41487d84a9086f7f9963c639278 : FAILURE in 19m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19a0782c1e6944e4ab72a6fea195bbb0 : SUCCESS in 13m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3b305924b27045fe9f36627d18c64af6 : SUCCESS in 16m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bebbab0713144b0b958f30a699227ea0 : SUCCESS in 27m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a82c60e206594cc3a9b410350e6a7ae7 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db722bce5d4e4dddb94472c855b82029 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd4e1dd33cd240f682b246ef21d849c5 : SUCCESS in 17m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5dc9336c1c7443dfb2713f93533b155c : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f5abf08a2c614f82b67468dd7418282c : SUCCESS in 13m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bd15f3a62e2d4c0ca50158896d069051 : SUCCESS in 15m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c86000a34fb4087aba5ad073befe16c : SUCCESS in 31m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/55d477abd41642e9b53cb9910eaab3a1 : SUCCESS in 51m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/961542427f0a484292c8e6b7a502d047 : FAILURE in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f775d9a881c649c6b89b86d689758129 : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bdaf22a4e15349fea9da995430106e93 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3a5894cd87d7749649fbdb1785b8ab4a324821a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 01:26:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd601a4654c5492e911b49e0d545e44e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25a94eab3f674af095af19be5a5ad370 : SUCCESS in 1h 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"598bfb4e04e10aa411776dc2b65e65916d4cb460","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-19 17:06:06.000000000","message":"Abandoned\n\nlooks like this got squashed 937267: Aggregate per-disk recon stats | https://review.opendev.org/c/openstack/swift/+/937267","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8d47494ff12d18df86643b679f8c0ee78ce38b2","revisions":{"d8d47494ff12d18df86643b679f8c0ee78ce38b2":{"kind":"REWORK","_number":1,"created":"2024-12-12 23:30:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/60/937660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/937660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/937660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/937660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/937660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/937660/1"}}},"commit":{"parents":[{"commit":"3a7b249ac3ae079bd039710e91fd720c629bbe1e","subject":"sq? maybe don\u0027t use self.dev_path as alias for self.sda1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7b249ac3ae079bd039710e91fd720c629bbe1e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-12 23:19:22.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-12 23:19:22.000000000","tz":-360},"subject":"sq? some more unittests","message":"sq? some more unittests\n\nChange-Id: I73c71efd52d7262cd4fac379a846ad485d1d0445\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d47494ff12d18df86643b679f8c0ee78ce38b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d47494ff12d18df86643b679f8c0ee78ce38b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
