)]}'
{"id":"openstack%2Fswift~937663","triplet_id":"openstack%2Fswift~master~Ib8945aac730c2c2912061896864202324d29e8bd","project":"openstack/swift","branch":"master","topic":"fix-oldest-async-manager-run-forever-recon","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-12-19 17:06:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib8945aac730c2c2912061896864202324d29e8bd","subject":"sq? more real recon dump testing","status":"ABANDONED","created":"2024-12-13 00:39:26.000000000","updated":"2024-12-19 17:06:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c3bad0c697d9cdaac0ee6176afd64888b9248608","_number":937663,"virtual_id_number":937663,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-13 02:04:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 02:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe682815a89453e472bce0cef933be78657cd6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-13 00:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97a7e6704659d6ffd50f65aa4c5be634941e65c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 02:04:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4c9da9a83bc4a57b4683da45e4e15b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916307d827fa489396f6832d82489275 : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b90ff12d442f41c9954a5069dd79a4c8 : SUCCESS in 1h 04m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cdb016b1859542cdb19fe450a1d90bba : SUCCESS in 38m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5096ebb8d9f4525a44a7e9035aff462 : SUCCESS in 38m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86015f476268491f8cf14e09d7ac40e3 : SUCCESS in 1h 18m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bce656b4646427eada201993a568413 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f0c745c780f49fcbe46981f176836e3 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecf0a46acddf4c66a11a83fcd71c66a6 : SUCCESS in 21m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecb760d2c9c5428e816f6e345772d0d4 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0865f7d4cc99426c99727aa29075ad0f : FAILURE in 15m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c12018428d384d7e921055cdc74d8c6a : FAILURE in 14m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/50a1154eeb6d44e0a5b3d6209d2b556b : SUCCESS in 15m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a032ba9e6fc6448d87fdc73d58ca93ff : SUCCESS in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76fe697b810c4f888c993861d8885f8b : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ec61eef8a554eb4af82c2372c659a93 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfabb6d37dc84f83a6ef26da81701387 : SUCCESS in 10m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b66a4d0905164bfd80a4703ebf7adc65 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ed99469eedc241649dcf3c20ab6f2e6d : SUCCESS in 14m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ac348920e704921aa236b48ad26afb1 : SUCCESS in 13m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/40cc2fa2e8be498284868a8aa3bfe57a : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b802667913684a669b744df20909b510 : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0693c8fa7c484d8cbc2ff550aebcd146 : SUCCESS in 4m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0bec4164d744a97a04f7c24d1a1470d : SUCCESS in 42m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a7d630ad9cef4eecbc105cdfd0e81772 : SUCCESS in 29m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7f9fe64a1864a37b8cff310d03f8f70 : SUCCESS in 29m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2711b694104940279ba85c895e66ceae : FAILURE in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3929aca718445c8fe2360b67fd1805 : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18c7532ccf4a434b8ffc37f98d8e5b1e : SUCCESS in 18m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/311c544ebd6e45678126625b16880376 : SUCCESS in 56m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d354838228bc46a0827e0deb604b6a4a : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9e90b160be3fac0683f3fc6f9025d3991d4f0740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 02:11:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c65996598ec4af8bb8cc8776aa951a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c1bcd888b16e47b79c39d707d2697213 : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c3bad0c697d9cdaac0ee6176afd64888b9248608","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-19 17:06:50.000000000","message":"Abandoned\n\nlooks like this got squashed 937267: Aggregate per-disk recon stats | https://review.opendev.org/c/openstack/swift/+/937267","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7509594e6d7c8b0cbb77a7b61ae90be3e384c838","revisions":{"7509594e6d7c8b0cbb77a7b61ae90be3e384c838":{"kind":"REWORK","_number":1,"created":"2024-12-13 00:39:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/63/937663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/937663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/937663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/937663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/937663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/937663/1"}}},"commit":{"parents":[{"commit":"d8d47494ff12d18df86643b679f8c0ee78ce38b2","subject":"sq? some more unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d47494ff12d18df86643b679f8c0ee78ce38b2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-13 00:38:54.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-12-13 00:38:54.000000000","tz":-360},"subject":"sq? more real recon dump testing","message":"sq? more real recon dump testing\n\nChange-Id: Ib8945aac730c2c2912061896864202324d29e8bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7509594e6d7c8b0cbb77a7b61ae90be3e384c838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7509594e6d7c8b0cbb77a7b61ae90be3e384c838"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
