)]}'
{"id":"openstack%2Fswift~937667","triplet_id":"openstack%2Fswift~master~I50fb99af575ba42900de90a87a2584b604723dac","project":"openstack/swift","branch":"master","topic":"per_policy_request_node_count","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-01-30 00:29:33.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-16 00:56:22.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-18 01:13:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I50fb99af575ba42900de90a87a2584b604723dac","subject":"Allow request_node_count to be overridden per policy","status":"NEW","created":"2024-12-13 06:16:43.000000000","updated":"2025-01-30 00:29:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":23,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"3a8bb715b1a3e1ac64b9130bec51475e9c277aa1","_number":937667,"virtual_id_number":937667,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-16 07:42:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-2,"date":"2025-01-30 00:29:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":-1,"date":"2025-01-08 18:22:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-06 22:54:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 07:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-13 07:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-14 01:22:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-01-08 18:22:05.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-01-13 17:22:39.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"444ffb3ae652d0fe633ef361ee69d53573df378d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-13 06:16:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31957e1f08aa0c266cc0f95359bf41e3baa829a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 07:26:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e58d6eef8124fed82cc9e2d8f1a8fe0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5d7298eab7a4f9e879077d6b32cefdd : FAILURE in 1h 08m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0038002dce2093f9a680474a35677120558b4a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 07:30:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/677736068d0d4cfeac94c6ce4e1e7067\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954c7b07848144d4a80e226355d5e868 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b563f770a7f4428886deb7be4fada18f : SUCCESS in 36m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89643fd2b5804ae1ac6524bb5f3505e1 : SUCCESS in 35m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/424de1d3de134bbbba1702a30dd790d2 : SUCCESS in 36m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d6854756b58426aa5569eec2067b96b : SUCCESS in 1h 01m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a7125196bd841b0ab68f72fbad44988 : SUCCESS in 24m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4eab382e35b740309136f296dab48d42 : SUCCESS in 28m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c19a9c08b1a54b7082123b92a0df0261 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d64f8a8d4b2c4381838f1d2adc7cdcdc : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/037d1155d72a4e0a9a91429a5f9d7ccb : FAILURE in 19m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab69a8bf5b714942a3f2d5944d93dc04 : FAILURE in 26m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/824aa7bca2b040a8a346bdad91477d54 : FAILURE in 20m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26bf1e84a2c0438a89081d92d20286e6 : FAILURE in 21m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4613f26da78e44bbb2241be67655b355 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f0e41661d2542c4a65f23837bef2aad : SUCCESS in 24m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6af25ae5e0ab4fa38f22c73b1f553f71 : SUCCESS in 18m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dde1883296bb45abb9933c6b9fe95f0c : SUCCESS in 20m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5294395b85c946fe8cd49b7e3c4dd97e : SUCCESS in 14m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b503fadc60aa49459371b99d2d3d90b0 : SUCCESS in 23m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc6dacbb8afe4f9b8b635f197f792908 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/698ddcd27e3c490db81605b57fd860c5 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d06823a7f3d641f58fc2d976ef5bef68 : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9d77e94e2f54a49bd5c7cbf2c623a40 : SUCCESS in 58m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4299fefff432432ca55f86227f2635b0 : SUCCESS in 46m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fc6635cd22344f7ba1327046c5fc2b0 : SUCCESS in 51m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a2186294ab644f0a6e436c1c05a21dc : FAILURE in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fba91109b214aada7a7c54831332b1f : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9656e6d914e54860826e892c6461f79a : SUCCESS in 10m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2c65204ea3844829cb1b596e5c0514d : SUCCESS in 55m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/909351c4b4c04062b7ecef3c24108e40 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d71caa35616ad67dcc2f653ccc97d03a5ced9526","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-14 01:22:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cef2f057c9a4e726bf0824a13a24f60b352b8e70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-16 06:26:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7c079d03c9419204b4d5e35ddb901a4af7336bf8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-16 06:30:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5a3e91d9f205a48efb59f8f01493dfae14517a33","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-16 06:35:28.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"434e1e5c37ea7f53ff0f1edd14b7fc82f99ad06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 07:42:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93248a136c41445faaa07b944c1fe998\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe30ca74cf44aae9797cd2e59c989d2 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/8bf2d850b7744b6ab5b39086696577f1 : SUCCESS in 1h 09m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00133d682ca44a51947f60ae4a39f8d6 : SUCCESS in 1h 07m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0c6a14f89f043b7bcf2584ef8dbe695 : SUCCESS in 41m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8966c346bd214786ba5b7a7b983948ef : SUCCESS in 1h 15m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f23c928ddf44f89a945ab29ca43b1b9 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2025097d6c74aebada521d5e97cb42f : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9997e1633d284331b704df450ad8ff2a : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52cb59e932b44d45be719f3fc37855b7 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0780382bd01455185917dcbdfec73bf : SUCCESS in 9m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb3efa85fd25472cb007c6091be55c62 : SUCCESS in 10m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/458a7f5f962e4109b1c4ff75a58add49 : SUCCESS in 9m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b0af00ad2bd44dc84a1183407f8ca68 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4915ace3c55a46968ee8904c45947a97 : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69712ef4ebed43389cb20c82e2b29392 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c955c38f271d4a34a4d7f68c2720c4db : SUCCESS in 18m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fae435d109af4c629dbbd28bb0928b5e : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/96be1e3055e740b1b35cede52f0fef46 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6102fcaea0d430bba55f8602a8c3a7e : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17b0d928c967473dbe92883c22bea878 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36c4761442234129885c876a31e809c7 : SUCCESS in 14m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9be1d160706c45c5aac0cad775d72b81 : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/00475162cbe7416db90d6828f22397e8 : SUCCESS in 53m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1f3e52a61b647d28ff54adcc402096b : SUCCESS in 44m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22c095d35b704bb4bf110609e894238e : SUCCESS in 43m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/844801568b6d4e79a786515a0300d3f5 : SUCCESS in 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29dea032dada435fb2a2c959dd6cf34d : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8dd0bf2d7bd4480282f053e6416e94d3 : SUCCESS in 19m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e50a5e85ec44458ba277c0d4963c0893 : SUCCESS in 56m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/033e9ca6174d423a89f7ebfb6862b6a2 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":2},{"id":"52b01da65df8a62c6ede816dc2a87fabfd561b0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 07:54:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ddbbfc39eca43769feea7e3923afa1b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/816243b1474f4dcc95db7d7d805ad483 : SUCCESS in 1h 26m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7d434ffa4717155e1e29cc13ffcf499caa9e823f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 01:13:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3a40ad281f360e89e311321c56e5d0ac8756223","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-06 22:54:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c16484f4e13bf0feabbf9d4866261a0441c03a7c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-08 18:22:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c430a559b09305c041c7d5d5a420977fb1e22a5f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-14 06:07:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3432abdb52435b9633b48bbf798e3c581b9c1679","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 00:56:22.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cc4afe271647b80fe1c224df4d8cfe7baabfa85c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 19:53:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a8bb715b1a3e1ac64b9130bec51475e9c277aa1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-30 00:29:33.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d94c098f9f50812f00c686ee9820c96c67c58c8","revisions":{"758200067f8e6e0f2a6aa81bf24c872154c38b3f":{"kind":"REWORK","_number":1,"created":"2024-12-13 06:16:43.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/937667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937667/1"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-13 06:12:17.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-13 06:12:17.000000000","tz":660},"subject":"Allow request_node_count to be overridden per policy","message":"Allow request_node_count to be overridden per policy\n\nEC heads that 404 can be expensive because by default it\u0027ll hit\n2xreplicas, which with EC is expensive.\n\nThis patch makes the request_node_count optionally configurable from the\nper-policy configuration. So it\u0027s now possible to change this value.\n\n    [proxy-server:policy:1]\n    # Maybe making it the same as parity means if it still can\u0027t find it\n    # it probably is non-durable anyway.\n    request_node_count \u003d 4\n\nChange-Id: I50fb99af575ba42900de90a87a2584b604723dac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/758200067f8e6e0f2a6aa81bf24c872154c38b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/758200067f8e6e0f2a6aa81bf24c872154c38b3f"}]},"branch":"refs/heads/master"},"6d94c098f9f50812f00c686ee9820c96c67c58c8":{"kind":"REWORK","_number":2,"created":"2024-12-16 06:26:21.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/937667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/937667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/937667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/937667/2"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-13 06:12:17.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-16 06:26:04.000000000","tz":660},"subject":"Allow request_node_count to be overridden per policy","message":"Allow request_node_count to be overridden per policy\n\nEC heads that 404 can be expensive because by default it\u0027ll hit\n2xreplicas, which with EC is expensive.\n\nThis patch makes the request_node_count optionally configurable from the\nper-policy configuration. So it\u0027s now possible to change this value.\n\n    [proxy-server:policy:1]\n    # Maybe making it the same as parity means if it still can\u0027t find it\n    # it probably is non-durable anyway.\n    request_node_count \u003d 4\n\nChange-Id: I50fb99af575ba42900de90a87a2584b604723dac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d94c098f9f50812f00c686ee9820c96c67c58c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d94c098f9f50812f00c686ee9820c96c67c58c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
