)]}'
{"id":"openstack%2Fswift~937681","triplet_id":"openstack%2Fswift~feature%2Fmpu~Id5e11872a07544b61aacadb7bd61f3e0f1d81ef0","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-streamline-func-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-12-16 14:29:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Id5e11872a07544b61aacadb7bd61f3e0f1d81ef0","subject":"mpu: separate object name and upload_id in session listing","status":"MERGED","created":"2024-12-13 12:58:02.000000000","updated":"2024-12-16 16:49:06.000000000","submitted":"2024-12-16 16:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937681-p-mpu-streamline-func-test","meta_rev_id":"168cabb5aba0175ee0d4c6cdfcf8614ef7a6364f","_number":937681,"virtual_id_number":937681,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-16 16:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-16 14:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-16 14:29:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 14:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d851c06b403c8a0b3427d3e62d5c4abcba775bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-13 12:58:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01a6513e54acc4c6f8a448e0b6045bfa3f732c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 14:01:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c85ade46ed124c51b8c9b9989558f8d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9944b01d42794cfa87aa6c17ff5f1b55 : SUCCESS in 7m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4083b92a9f3e424b97ad28f16cf92420 : FAILURE in 26m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78618ad66107423cb554bec685b74ec9 : FAILURE in 24m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf0fad3af72b4e4bb32624fc935a45f6 : FAILURE in 29m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47f8a8cc58874a26bb9fc8279c1df346 : SUCCESS in 4m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/635cbd3cbd0a4fc49995d5044185d45f : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/224731a508fa4c55b252160ef39bf852 : SUCCESS in 22m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df1b59056868479c87cf7b69dcc0cbaa : SUCCESS in 23m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9cff534b9f64886b964ab213dab4646 : SUCCESS in 28m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ba74fd042b349b4a1ed74dd57c26480 : FAILURE in 29m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/381e6414a67143a0a089e0d250b32890 : FAILURE in 32m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d368b3f8cba44280814b0d127b11c34e : FAILURE in 32m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb6bc425ef514cfcbcd4395ca21e867d : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fef07b0e986f48bbb3d4ca68221a19ab : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/898cd9a49b3749e8bfa68694f5b193b5 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b609d25ee77c483b93bbf166b40eb6a7 : SUCCESS in 8m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/779275e99e56446a9137cc17380ddb4b : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38f02698bf404175bd446bd7ad2d954d : FAILURE in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfa91e03bed6481da15ad32f12fd50cd : SUCCESS in 55m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/270f6ce1f79e4b698087bc872e266466 : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2844fdd3ae20432c8e8ffa02e57c09d4 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d25fa750a8d64bfa95aebc7d1b122880 : SUCCESS in 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6939ac36329d4a41e0f7a97e025fe9d8356712df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 16:12:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6301ca864e149d7b0b1ca272c3213ad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05bbddd507e14decb88e30692521ac1a : SUCCESS in 2h 19m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fda228b83a67b588bd94f728ff22d5b2c6bd9677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-13 18:50:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8664a452799a3c85a35b3652a8abbb4ef85a339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 20:01:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a023ebcf93bb447fae0a0a7c787fca94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e37581102d4459f82e0b43cbdbb7bd1 : SUCCESS in 8m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/744087ce022648af9368a3cc7a7899cb : SUCCESS in 24m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/277b98a495224566b186686bced174c9 : SUCCESS in 21m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9e3920560f6448abc34311969233457 : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/566fff76b3ce46c39d435fb967a5ad9d : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e6965e4ced5426e8a9f8d050e353155 : SUCCESS in 20m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26df519277b8403da822280584090f83 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a83aa3431696499db41542e721e87238 : SUCCESS in 14m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/852515a1965f4811b27d16c8178e5a37 : SUCCESS in 34m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fac9b8ab0bf642ddbcaa71bccd9c15db : SUCCESS in 10m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5a9efac8c64410993ca233b1383470a : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bad78055ff94849b5d980cb8afc78a5 : SUCCESS in 21m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/acbb50c640024b478f75f643a9d29869 : SUCCESS in 8m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6be32e1fa43425d93f347e121a52f26 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/705990b1165443eaa194c7f186b8be3b : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07e73ef8a649469d9f3d2636d7819bf0 : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f725f7b94ad4b638e8f427a521c4f0e : SUCCESS in 8m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f01c66c3ab845318559b6b7aa48c94c : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/51f4ebea79e749a7b2325f6ea7ed1eb6 : SUCCESS in 59m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/839ff8ff0409422baa1a24fa41c8f68b : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4991281dfb4e2ca447f760f34a1468 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/109018e3b8f24295856e891b8b8546e4 : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":2},{"id":"9c6a08b9a186d9b4313c99be3a32e8d2a44c7170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 20:54:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/051818d85526485487cbc468de7a4adb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6deb35b069c4c97b4156bd8b9f20df3 : SUCCESS in 2h 02m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7dbe141e0cb830c99993b4bd23595c88dc511e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 14:26:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c524093779d8e38f9d6b1e709c49fc184e1ae216","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 14:29:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8970d107ca4ee407364a8ff469bbf38b7a6704fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 15:47:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19c330afbc14b1082a47f2993836b9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c960884df2fd4ddcbf5191c8f8ffe944 : SUCCESS in 8m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8d2cc79918048baad874e61b84cdacd : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8da9cf33f097480bb3b09a6d6d5cdc3f : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f974c38ab1407683e3fc1b511bcdab : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98c5045c528f4734856a65c64ab98183 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17783c1c9d904ff3a9952d076aae5072 : SUCCESS in 14m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bed4e9ca1ae3448bb9f579347fd442bd : SUCCESS in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8630907f53444d1ebae8c52181235d45 : SUCCESS in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ad88dbfd5644ff9b05f36ad164da746 : SUCCESS in 20m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4addbc7f37bd4f86a17d562861993b88 : SUCCESS in 22m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92253c56b2ac46dcaa7bc674bbbc6eb2 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/64f21de46a904414b7a0b6b67956ecf5 : SUCCESS in 19m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/39426e9cd2244cf79ac94cfa65d8feb2 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa6d56e39f1242fa868e67eef8653ba6 : SUCCESS in 13m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc9cf221673649c6b6d2788e9e18b13c : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46e91a592bc449a4a234482d2644e4cd : SUCCESS in 5m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64d72c7a00a84ce0a75e6f54ce6ad773 : SUCCESS in 13m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b322017323a42bd87e3eb881e3840ae : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b277357d3b74087a858434cb624fc91 : SUCCESS in 53m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6da2218af98b403bb2c3ee4358d96ec4 : SUCCESS in 22m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc7b96f65300457393caa267291bc8d9 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff1bf5dbb9c04fdba54f2442b8dfaa98 : SUCCESS in 24m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ce9ea64e2cb51793a0eb389d81ba31065d0996cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 15:47:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e4b6b5ec8eed67cc5dd19df540dc3513e80d9073","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:46:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e941331a3a3240fd8de59034f45f9f19\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73620d885dc041fe8dc909a3189bd1dc : SUCCESS in 1h 41m 47s","accounts_in_message":[],"_revision_number":3},{"id":"51a1d02ddabc1fd4bb662167b3e947d4a563fbad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:48:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61100b2f2a374038b9a6f0ef7222a04b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af9ce2fa1434c2eb7ca8ad6b2ee5089 : SUCCESS in 10m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b51937af340c475886631bf822405996 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38fa525ceddb445c9613084a409b16ee : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db32569f19324c34b4348645a02a5699 : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b5e941f81af0405494a189d1b2dc7863 : SUCCESS in 22m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b07d89da157e4f698dce4aeb7f10d2dc : SUCCESS in 23m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fb00cc3022f4c369ee32fd048143e20 : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/051517c8b4f8482e8ff8b805296d9593 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9991f3693d643f09656a8c08bc57e35 : SUCCESS in 19m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31c3c2795c924051b60a7898d76a4816 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f15e29bde9b448638ee622cbedb46fdc : SUCCESS in 13m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3742abddbee743f499fde76f1ad5b922 : SUCCESS in 18m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69db4295a5ca4208bfa9c620e43702d9 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bcd16da9f97a4a7f913390dbfedec600 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f421a7443864706bc919e8abfb32eaf : SUCCESS in 53m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b4e8fbbe7bc44568616878ea0410468 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc6e2b752094bf38654ab86ee7d4ad8 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88477514fd88464483ee2d37bd23fc4e : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":3},{"id":"6db0e791feb52373954c62da6f57e560702a15bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:48:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"168cabb5aba0175ee0d4c6cdfcf8614ef7a6364f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 16:49:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9138188c905442a3bbc51ba9187799e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbef3af58b4d4b909e2bbc8ecc20b693 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c06b30b4f1f54ea590705f6c927bf575 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6db95c00d4ca7aaec416d7623d85812313147181","revisions":{"4497949aa72cd7c5fb5a7becfc1f00e2cea466d0":{"kind":"REWORK","_number":1,"created":"2024-12-13 12:58:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/81/937681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/937681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/937681/1"}}},"commit":{"parents":[{"commit":"acbab92f518fc1763358d1ca2ffabd117e2b1300","subject":"mpu: add min_part_size to registered swift info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acbab92f518fc1763358d1ca2ffabd117e2b1300"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-13 11:48:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-13 12:08:12.000000000","tz":0},"subject":"mpu: separate object name and upload_id in session listing","message":"mpu: separate object name and upload_id in session listing\n\nChange-Id: Id5e11872a07544b61aacadb7bd61f3e0f1d81ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4497949aa72cd7c5fb5a7becfc1f00e2cea466d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4497949aa72cd7c5fb5a7becfc1f00e2cea466d0"}]},"branch":"refs/heads/feature/mpu"},"d404cc7d8b52b3d64edfbcc4a78071b9fa0c35f6":{"kind":"REWORK","_number":2,"created":"2024-12-13 18:50:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/81/937681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/937681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/937681/2"}}},"commit":{"parents":[{"commit":"acbab92f518fc1763358d1ca2ffabd117e2b1300","subject":"mpu: add min_part_size to registered swift info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acbab92f518fc1763358d1ca2ffabd117e2b1300"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-13 11:48:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-13 18:36:58.000000000","tz":0},"subject":"mpu: separate object name and upload_id in session listing","message":"mpu: separate object name and upload_id in session listing\n\nChange-Id: Id5e11872a07544b61aacadb7bd61f3e0f1d81ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d404cc7d8b52b3d64edfbcc4a78071b9fa0c35f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d404cc7d8b52b3d64edfbcc4a78071b9fa0c35f6"}]},"branch":"refs/heads/feature/mpu"},"6db95c00d4ca7aaec416d7623d85812313147181":{"kind":"REWORK","_number":3,"created":"2024-12-16 14:26:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/81/937681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/937681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/937681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/937681/3"}}},"commit":{"parents":[{"commit":"acbab92f518fc1763358d1ca2ffabd117e2b1300","subject":"mpu: add min_part_size to registered swift info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acbab92f518fc1763358d1ca2ffabd117e2b1300"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-13 11:48:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-16 14:17:14.000000000","tz":0},"subject":"mpu: separate object name and upload_id in session listing","message":"mpu: separate object name and upload_id in session listing\n\nChange-Id: Id5e11872a07544b61aacadb7bd61f3e0f1d81ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6db95c00d4ca7aaec416d7623d85812313147181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6db95c00d4ca7aaec416d7623d85812313147181"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
