)]}'
{"id":"openstack%2Fswift~937819","triplet_id":"openstack%2Fswift~feature%2Fmpu~I7bb1b1d5b6eba92b84b3ae84301515de120cb278","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-streamline-func-test","hashtags":[],"change_id":"I7bb1b1d5b6eba92b84b3ae84301515de120cb278","subject":"mpu: support upload-id-marker param for session listings","status":"MERGED","created":"2024-12-16 21:43:40.000000000","updated":"2024-12-17 16:11:01.000000000","submitted":"2024-12-17 16:09:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937819-p-mpu-streamline-func-test","meta_rev_id":"0b86bcca8cd469037abedf5d0da3e38601103813","_number":937819,"virtual_id_number":937819,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 16:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-17 14:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-17 14:59:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-16 22:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bb152495bc6d03b0d34894940f94c0ad2875412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-16 21:43:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95ba4c03e11a8a74d7a2184c0003769ede9f82f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 22:39:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd8a1efb1af408eb1c5612b69350a8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1a40e1c2d54c4c817363f6e50dac64 : SUCCESS in 7m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/662436ee289f493da8faf53046929f15 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fab6de1a543f4814ac2001ce4c0d2fca : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/463368f93db64259a3f757c4c1560038 : SUCCESS in 21m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfe2da58c5cd4059801b936c38419596 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d02fc9e039df42a09fd0832cb6b215e5 : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3daf9ee3630448f998ea5015b363c2a : SUCCESS in 8m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/651992d3d3ce49aaad75a2ad4f2c3619 : SUCCESS in 9m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed99068011544e838045f3f071526c65 : SUCCESS in 13m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5a000bd08354481f8ea902bc910c1704 : SUCCESS in 9m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5515913b67df4a1e9303727f813dbf42 : SUCCESS in 10m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e6c251a5c9d4860b2d90093120929b8 : SUCCESS in 9m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/41a54bf5a437425f9d30e110f4837b40 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/809c070bcc384d8d80660447efbd0179 : SUCCESS in 9m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d2d175bab46a47c48a46de26c05d7ede : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/218c660a1bbd460c9815dbc597da4c40 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c0429ba749c4c7e97f1c890c677148c : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/35fe6b9347c6458c9cc788c081299b4d : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a5ef25b4748464eac17e129136ee194 : SUCCESS in 55m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a572d13ebe449ce8b728fc59d71551f : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a313e694744b447890b587523eed5e12 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7a6b0699e854e079ffdf59f077d0b78 : SUCCESS in 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e9ce30288424f6054ea7b79c96b716f9b6a26e50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 23:41:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2bff37b69a4d2b8f851720bc1fa648\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96dc0f7634b14b42b1777186f9976fae : SUCCESS in 1h 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5ccb7d4e00a35f9403a96f6d70a8458ecbb3e025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 10:46:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2c9981bf984fe0e7b71875095df1d4b4e970a5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 11:29:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc69781797bb433b8d8a5a05fd56dddc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781dbb1df6bc4cff81a1d6d5629bc5b9 : SUCCESS in 6m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/767c58aa2a794e5d81038bcad25a2493 : SUCCESS in 23m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a008427ca5be46f7b3a11b7f5b154483 : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1deef7af32f8471da37b7dd4368536b8 : SUCCESS in 23m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa6cddbcdfda4048932d226d70cdf226 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c282ab283d9a46f3bcf1c3bb4ae321a2 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/057e6305cba84279b610b5875177a352 : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac4da306a0c849f6989e1522d85861c8 : SUCCESS in 20m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d1d9787c235045fd8d90eefba1fb37f9 : SUCCESS in 19m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3de6a297d56c4d89be9b1f5f746bb17c : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42100150f22e48468579bba24d6be41a : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1caf6c15d6f54d0fb0203a1bf97d506f : SUCCESS in 18m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/909aa4e1152947b3a529075b8fef61af : SUCCESS in 10m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6c935f57f1254e9fa937ad5208f1c6ef : SUCCESS in 15m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc3f71ab930b491596781ee5e5b97a68 : SUCCESS in 9m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66a5a6aa2db74726b0324cf9530cd016 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dd192b7ad8b4913ab2567dd5ce2f3b5 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec8d37dade894bf0a43fff3a24787388 : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e4aa0a0d45a4dff87c2e0d1b1dcd76f : SUCCESS in 41m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b87143d0ad694feaa3d6ff76a59fca6e : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a232faeb930142279a45bde3977cc991 : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b2cbddf66bb4cd88bed7c8c2fc5aa7a : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":2},{"id":"70716a03a5f55546598b1d054dbfce7942a37ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 12:20:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cd82fd615b15422a8aee2e6f03b085ad0c5d29f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 13:25:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad57699c57d941ba8834a151ee79cf2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904d817af9e84d8e967a535633b89437 : SUCCESS in 8m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1398bfd6a04642398748b1465aed5f3f : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b29e4798d5d439ea25b9a21f995e8d7 : SUCCESS in 22m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a593bae47d44088a96b3ad324b0106b : SUCCESS in 25m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c50b5ddb51c44ae9b69bb8de97a72df6 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/01a152dfb5964c05b0230b79a7791a52 : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8c9faf7d04485c9b6770b2336ac881 : SUCCESS in 17m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9975a10267bb4d1881a4ef0d7d392466 : SUCCESS in 23m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3a5f8e33fd06440795ccd32df02e7aee : SUCCESS in 22m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4394016670894b4e8d1f94f66bfb067e : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f61b831f53eb43c3b81040f40bb8ad16 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2329d9c2ec064f569fcac57cc72920f9 : SUCCESS in 18m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28f0da4a1a624f8bba8241f0b1e27ead : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/670e7d03dd4041f18c797f237388d84b : SUCCESS in 17m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/874c2fedd8be42878b5b37bb68d07ac8 : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3620ec9c7f9460382832a2ad1a26602 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d587b682dbf47d684ba65b1e8f3839a : SUCCESS in 12m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/be20dde545814a1788a389d2c140cc58 : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/76b0431002974a93b17e0cdab432eec9 : SUCCESS in 1h 04m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fb1c50777c941a288118f080b34c4fc : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4c1eb6f4db4dd3a380f16ad250ea90 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a728d16900d34c3cba472f99b4f81e3e : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":3},{"id":"7e5bfe0c0833533deee867f0a84839f70a5729ed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 14:59:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"564a4ec5808478b83c0d8e47434d60a6c264f935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 14:59:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fc1e66bebfeb644741393f090cac9b4c6791a3c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 15:43:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28052799063b42f9a18d7b48268f8050\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8154b99b9d9447aca2b0ef939323c52f : SUCCESS in 2h 38m 48s","accounts_in_message":[],"_revision_number":3},{"id":"16e9f9c969d0fe05e957398ec7e92e7e7fa930af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:09:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caff6b078be64ebf801a1708b64e834f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b54707e8cc2e43a2a54f1f4e4d4703e5 : SUCCESS in 7m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b2b811ae1638408b8cea804e3dc8750e : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/844deaa88c44466db7999f8083f66300 : SUCCESS in 23m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f44e45e5d214458bba95e4610da6d0c : SUCCESS in 22m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/34d465ddf15743909ff4ff42bf99a9d8 : SUCCESS in 19m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a2e50064ee014893b0ab23f0dcea4bce : SUCCESS in 31m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7ca8bb04f4b4c7ea9818098879e095d : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7accb513faca4730bec6fc4c7a12aff4 : SUCCESS in 23m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aae408fdc94f4eaebc2603798789d659 : SUCCESS in 18m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d2e59ee39f940ea9cc9e7f32efd9cb7 : SUCCESS in 19m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a50e938063d2414a8db7d8b49d453460 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1d62b309f4134a1bac006f89fe0ed873 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e018c102ece4390b909c2f7d7935195 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f5272a582d774f53985838c1537cdf0f : SUCCESS in 9m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/95639c6ae3794d9c9caebf80591ae5f9 : SUCCESS in 53m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33d2bef6dd3943abaaae3fd87b04f20d : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e15e4f31c344c7b477f74fd0f9437b : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf27bee281b74fb9b1b0f71f6c090224 : SUCCESS in 18m 40s","accounts_in_message":[],"_revision_number":3},{"id":"a3a40bc6cf825200f808c087439f445834382731","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:09:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0b86bcca8cd469037abedf5d0da3e38601103813","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:11:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4a97c975254462b25244ce6a7d8431\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1764a225eda34ad2af37a792de53175a : SUCCESS in 1m 21s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a1d3e0c0da1a4f8d97a282d61cc916a6 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f131432d917c10224e8f5da71a1f37fd1ec78ace","revisions":{"5d02881bc59bdda6391efa52359e18bc498815db":{"kind":"REWORK","_number":1,"created":"2024-12-16 21:43:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/937819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/937819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/937819/1"}}},"commit":{"parents":[{"commit":"fa9a04099540b9b9289e13cef23f712901dd0f1e","subject":"mpu: streamline functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa9a04099540b9b9289e13cef23f712901dd0f1e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-16 21:42:26.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-16 21:42:26.000000000","tz":0},"subject":"mpu: support upload-id-marker param for session listings","message":"mpu: support upload-id-marker param for session listings\n\nAlso, drop support for end_marker for session listings since the s3api\ndoes not have an equivalent.\n\nChange-Id: I7bb1b1d5b6eba92b84b3ae84301515de120cb278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d02881bc59bdda6391efa52359e18bc498815db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d02881bc59bdda6391efa52359e18bc498815db"}]},"branch":"refs/heads/feature/mpu"},"c8d009910cfcea82cc9545c7fa7a6eed74b51727":{"kind":"REWORK","_number":2,"created":"2024-12-17 10:46:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/937819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/937819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/937819/2"}}},"commit":{"parents":[{"commit":"fa9a04099540b9b9289e13cef23f712901dd0f1e","subject":"mpu: streamline functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa9a04099540b9b9289e13cef23f712901dd0f1e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-16 21:42:26.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 10:46:24.000000000","tz":0},"subject":"mpu: support upload-id-marker param for session listings","message":"mpu: support upload-id-marker param for session listings\n\nAlso, drop support for end_marker for session listings since the s3api\ndoes not have an equivalent.\n\nChange-Id: I7bb1b1d5b6eba92b84b3ae84301515de120cb278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d009910cfcea82cc9545c7fa7a6eed74b51727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d009910cfcea82cc9545c7fa7a6eed74b51727"}]},"branch":"refs/heads/feature/mpu"},"f131432d917c10224e8f5da71a1f37fd1ec78ace":{"kind":"REWORK","_number":3,"created":"2024-12-17 12:20:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/937819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/937819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/937819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/937819/3"}}},"commit":{"parents":[{"commit":"fa9a04099540b9b9289e13cef23f712901dd0f1e","subject":"mpu: streamline functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa9a04099540b9b9289e13cef23f712901dd0f1e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-16 21:42:26.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 12:17:24.000000000","tz":0},"subject":"mpu: support upload-id-marker param for session listings","message":"mpu: support upload-id-marker param for session listings\n\nAlso, drop support for end_marker for session listings since the s3api\ndoes not have an equivalent.\n\nChange-Id: I7bb1b1d5b6eba92b84b3ae84301515de120cb278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f131432d917c10224e8f5da71a1f37fd1ec78ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f131432d917c10224e8f5da71a1f37fd1ec78ace"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
