)]}'
{"id":"openstack%2Fswift~937865","triplet_id":"openstack%2Fswift~feature%2Fmpu~I4ed51491286be948853b0fc13371ece3d668b6ec","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-streamline-func-test","hashtags":[],"change_id":"I4ed51491286be948853b0fc13371ece3d668b6ec","subject":"mpu: enforce max name length","status":"MERGED","created":"2024-12-17 12:20:07.000000000","updated":"2024-12-17 20:44:31.000000000","submitted":"2024-12-17 18:16:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937865-p-mpu-streamline-func-test","meta_rev_id":"50741a33d55e7772c1fe4ca8d5934d3554f710c9","_number":937865,"virtual_id_number":937865,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 18:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-17 17:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-17 17:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 13:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d962ee642beee1f9e5a595af72c0bf4473313a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 12:20:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49881b8d1f3b87beb3c53611a2f8e9e1750cb7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 13:28:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c930412f810b495ea9309ae841cf1014\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fee69f9bdb44e8da70ce1c4b1221df4 : SUCCESS in 9m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c22ab7fabb4d421ebc94e9248a8ed988 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd9790e1026a42ddb1b64da3a71c3172 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c075eab98b0c4b3696343004e69044be : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0cdd7812df249f2b2f835682746dd6f : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e549d061e55246dab89936c12b7185fe : SUCCESS in 19m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/448a7e7728fa401d9984367693418f13 : SUCCESS in 11m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/66f2dd87644047de91c478c91158214c : SUCCESS in 19m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00a3bfcb045c4473b9c8c555b83173c4 : SUCCESS in 30m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e999a6b5b8d4016b937c223a39bcff4 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a996c8e8b49495a8d1bffa9c9758ac0 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee29abd7808b4bc68705e4a4b3bb3d0b : SUCCESS in 17m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/33341c7ec12e40bab4277b16139703ad : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/321abdccd5014a97a61e4eb54ad19e27 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c6c2f25785e43758e72dab0ac107e56 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69ce121e56cc4f42a2baf2e8db00e59b : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4e6f5aa0bd54ff7b492065c0dc4ba1c : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/80b284181a9e491593557447635257d5 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/00088b85a09d41f18dc3ee0e4c39919a : SUCCESS in 59m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3440beb8f51741a69ed99f79e9493ae8 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25810df6082e4aa691855124c222b73c : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92cf0726180344b1b14ee33dfa2bcbef : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4bc2f5891832baf1bfb9f6648c3ddf00330738da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 15:03:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a01919f689e75b2ac2e9df93f99d10a691752447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 15:58:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0952870459554fe78e16d9602ce1668d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1377ec06d1742f2991df984e4d0e076 : SUCCESS in 8m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09d8f59fa8164f37b00f2ce2d8db75c6 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b72fabc5dd2f457095587b12f77eb7ff : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e49e78d13b8c4fa3bbbd20f80123f658 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e06001fe1d6470b9909a6dc3d7c96dd : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7780cd55ad354395a17dd5a947ed05b9 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2103c0d4675492b987079104d8e8745 : SUCCESS in 21m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39575f7d798a42058d4b7b092b9d8941 : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7c2ff78c25ab4537aa73d68e3e05d2d0 : SUCCESS in 23m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8e262b7edb443f0b350d7f553af7896 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce341f1dc1d74ab1985915722b2eb83a : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4457791d4bc4cde92e1e7821d77bbf3 : SUCCESS in 17m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff393dff4de1447eb4684044519e4fba : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ec17650a527c4c609216d39b1b648489 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed400aea86ce45e89b1bf6a249062e17 : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e09f9948a05d411e91d9f70a15a9c8d6 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/50d3b421ceb040288bbadfe9c32f21fa : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3bfe4a6bc9fd42a7be086c5212951959 : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/719570e3cd944ea5b5f3d4852e57fb1d : SUCCESS in 51m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d1f50ed2ecb4f8c88b9c62edaecd275 : SUCCESS in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c2a5c92807487bbe162f5295adfc1a : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88d04ae707dd4f9882757b37e35c9504 : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a52d7d0d8a416275a6d5b60fd89be9ba42955e5a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 17:09:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2bf760ffe490b6e0e82e62b547e0eb0d4d39a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 17:10:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1664c15ff63f6fff6e1583a8b36435ec1e806563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 18:16:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d123b2c76c2d422881f94970f833afc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65110e700abd46a59c12d11b2316dbc4 : SUCCESS in 9m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/86abc464021e41dd80c5851645b7986b : POST_FAILURE in 7m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/639cd4b196804804a2fee69081b3d3bc : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f19553f404c414aa82f666a50c0d809 : SUCCESS in 19m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/458262fc53914f509b8cb05736f0bcd3 : SUCCESS in 12m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cca8bfba9e364f499939349105c162c2 : SUCCESS in 18m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bfeeef188e324577b97be48d09e8d92c : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9abe42cee07c4181be86704b0dd04970 : SUCCESS in 21m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90702d2b345a4f96bda47f1e0d0adee0 : SUCCESS in 12m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cfe4e8e5313740eab6ac7715ab8b40c7 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f3401e0773c947218207e5610dbd1a47 : SUCCESS in 11m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0f0073e98fc7466cb5cf8e1296f034aa : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea72964b728d40caad46e9e033b2a668 : SUCCESS in 4m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a0027820b9a41a8b4725b9b5899d527 : SUCCESS in 7m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad98b974098743e48644922310ac8a3e : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/864646d5de41425387aa16d08202a62f : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a74b1f75894a79af9488ba9ec5e66d : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b07759bcc9d449d5a0d202aedea51d0b : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":2},{"id":"40f731f6469283c5fa080deefc63a01b422583c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 18:16:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5e4f7b8c5fd59d704f9b1253b132909a2781be7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 18:18:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2722e5d6fc4ddca5267ac43d16b8ed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c905e93b247f4df7a98480e6fad522b7 : SUCCESS in 1m 25s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f1b296137b174a8eb43eb955c627e6da : FAILURE in 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50741a33d55e7772c1fe4ca8d5934d3554f710c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 20:44:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0d6ebbcbc114d879301f8af60c59ebd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d295d7a098f54d69b34bd4ac665bb453 : SUCCESS in 2h 54m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a9b16537d0b0e190983c5985dc6836a749bec298","revisions":{"f68b82e90b3c0e1dca4b4cd63e5466afb8809f9b":{"kind":"REWORK","_number":1,"created":"2024-12-17 12:20:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/937865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/937865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/937865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/937865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/937865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/937865/1"}}},"commit":{"parents":[{"commit":"f131432d917c10224e8f5da71a1f37fd1ec78ace","subject":"mpu: support upload-id-marker param for session listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f131432d917c10224e8f5da71a1f37fd1ec78ace"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 12:08:04.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 12:17:30.000000000","tz":0},"subject":"mpu: enforce max name length","message":"mpu: enforce max name length\n\nmpu names must be shorter than the max length of a regular object\nbecause the nae is expanded internally for the manifest and parts\nresources.\n\nChange-Id: I4ed51491286be948853b0fc13371ece3d668b6ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f68b82e90b3c0e1dca4b4cd63e5466afb8809f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f68b82e90b3c0e1dca4b4cd63e5466afb8809f9b"}]},"branch":"refs/heads/feature/mpu"},"a9b16537d0b0e190983c5985dc6836a749bec298":{"kind":"REWORK","_number":2,"created":"2024-12-17 15:03:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/937865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/937865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/937865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/937865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/937865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/937865/2"}}},"commit":{"parents":[{"commit":"f131432d917c10224e8f5da71a1f37fd1ec78ace","subject":"mpu: support upload-id-marker param for session listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f131432d917c10224e8f5da71a1f37fd1ec78ace"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 12:08:04.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 15:01:21.000000000","tz":0},"subject":"mpu: enforce max name length","message":"mpu: enforce max name length\n\nmpu names must be shorter than the max length of a regular object\nbecause the nae is expanded internally for the manifest and parts\nresources.\n\nChange-Id: I4ed51491286be948853b0fc13371ece3d668b6ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9b16537d0b0e190983c5985dc6836a749bec298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9b16537d0b0e190983c5985dc6836a749bec298"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
