)]}'
{"id":"openstack%2Fswift~937882","triplet_id":"openstack%2Fswift~feature%2Fmpu~I36428e03ebd874892095e13a81e722a3d5550c2c","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-streamline-func-test","hashtags":[],"change_id":"I36428e03ebd874892095e13a81e722a3d5550c2c","subject":"mpu: store max part number in manifest in sysmeta","status":"MERGED","created":"2024-12-17 14:55:08.000000000","updated":"2024-12-19 11:36:16.000000000","submitted":"2024-12-19 11:35:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937882-p-mpu-streamline-func-test","meta_rev_id":"9a217a06d71d4e07cf5ea3ea4409d437f2fbaddf","_number":937882,"virtual_id_number":937882,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 11:35:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-19 10:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-19 10:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 16:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"feee66a27c73b70ea8ab2d8e0295a8420e90fb63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 14:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a9ad2c903a683054d5251328340f5f1820c73d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-17 15:03:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2fd83151365c888d421d3ff3273195e36de5c2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:19:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49542c822b8e48948e9e72b0e25f9c54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f1b5d8ab064aab84e4b8f6ffceb9c9 : SUCCESS in 10m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4612afa260c64c73ace196b9cea18fdb : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1cd06fcb8ad4cb6bc6285a5ffddc77d : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca2b46da43e941a1845548f4524eff0d : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7320120ca87f440c8fa1dd457b3bd572 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/db6e4c86296f417b8265e3836b5a776a : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4fe1d52b5a28467fb7be064821740555 : SUCCESS in 19m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f28edac5e8874816bef78e6ab24fdb98 : SUCCESS in 19m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1839ed977d2d45f6a0ba4457b7007955 : SUCCESS in 30m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a85d66de3cc442bdb9e7ea4f060f7601 : SUCCESS in 12m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54a565d77db240baa40dee6e5de410be : SUCCESS in 19m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/adbd7272f53d43ed8ad35e325ecee912 : SUCCESS in 19m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3c0de8ccc6b14a599844f468f7ad17ae : SUCCESS in 10m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61fe2babfd5c4a38beb5a275debaea77 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9374641f6d3344129e9f0f14154be262 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/910bdf2cee64477cb1c8232027a35330 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46cf76fdf6da4b3b9cf2307422f9d274 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7eea69ff275d4dd6b9dcf82a9b612f01 : SUCCESS in 9m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c7474bb44bc4e658c6f72f806f681c3 : SUCCESS in 1h 00m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95116e714539479e90a3ac2175ae7f31 : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb1059658b4146de88900df85fcb4dfe : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bbf7b59d94a4b34ab8d18d79457920c : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f3d055cddbc12f87e2bf053bf814ddca17a6b389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 21:29:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b700b7cf0074e699ee4ba85aa9dcb29\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d40ba395712434689d14741c598ca0c : SUCCESS in 2h 42m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5eed0b49ae9a15bc880a1cd881d4735f20fc0944","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 12:13:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5138c04d8ed95f160dfc3cc3c4936ecef4cee343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 12:20:03.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5daefbd31ff9c8df93c77685b53fba4949bb21c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 13:16:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ef8292d12d49ab81e71f36ef6d1aa6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7963c5561a84a6eb6b6cc918bb5f335 : SUCCESS in 8m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ea294f32ba84d19812693afe3c3e4fd : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d59bfd3219f0483e8684ef2d681d9457 : SUCCESS in 24m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/669e205a61cc443b89b2f4fe0179967b : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c1e6a525b63410d87ec908f7052ffa4 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fde2029db3fc4ae5b08c1a348d6284ae : SUCCESS in 9m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/55c518edc4fd42259fada2efa989388f : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6e1f116454d4ded8de761f24eed85c9 : SUCCESS in 9m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8768562e9224465383f541a325724154 : SUCCESS in 30m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b13d95346e54de28aebc36a8b902f5c : SUCCESS in 9m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7bd3b51ef1db4f97bf637c3e86c71500 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ced07563be74f9c9fc30f9e771e83e0 : SUCCESS in 18m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5627b493d6d84fa6a6469ba7323574c8 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eed8e6a08c544bd495e2d884e7365333 : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f5d769983773470797a80a379061d625 : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9781c7f3bc8245178128ecf57e306ecd : SUCCESS in 8m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f83f466f13b4f358248407f8070cce1 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/87c0e174ad3a48edbb20242849f0d35e : SUCCESS in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/223cd61fc6f940c894669a1c4b628d90 : SUCCESS in 55m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ad73a20e1c74a6281d4de2300c99807 : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/627e9f9466344c3f8e888693fde38c99 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c495986076b4d0b931782006a6c9dce : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":4},{"id":"aac6986532528fe7914a3d8b904ff03e850a149c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:06:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef84c2a3da541ae96e4930a3c32a8e7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f85ebd5b3344c4eacb999bdccefab0d : SUCCESS in 2h 39m 57s","accounts_in_message":[],"_revision_number":4},{"id":"2a043a750cd6ec3786e65fef3b3278a632ba8b99","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-19 10:51:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ef3aef1c14d0a7d7a1c8727de3d1679b83dfe416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:51:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bb19c68a016844556d9ad13a5e986583decd5b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:35:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a56ce8fc59498799e25de35b407cfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4037663cabf4d01803907c6460531b9 : SUCCESS in 9m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/80feaf77a14d4383967faa78e493888f : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed8071425fe4466ea1aafd63981c19cd : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/711c794f99a74b3ead138cb51689a59b : SUCCESS in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b9f0cb7d7bb43f9bbf4934b622e090a : SUCCESS in 12m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6f2411f6b7744be9a45a268b719daf79 : SUCCESS in 29m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e74f41ce09dd409fadd5fefebd59244e : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/161b4575656a447b8e28f52c49025798 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/193c9f83349a4a55aa5754fdd62bef15 : SUCCESS in 16m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9410e396feb44d49b7685626e32ecab3 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/be814a03b84646a7b04d3fe209f60a51 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c4137f13c3d5433996e49284fde6d5de : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7964d4f2f684b6a971d9c7b83518a70 : SUCCESS in 4m 43s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fbea145164794c70911ed5a87799faf8 : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ccf389b31164514999781c0e18af48b : SUCCESS in 43m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3825af4ba05a4d9c8497ffe4f811b21b : SUCCESS in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684ccbd991aa4cde9639ed68d75e8c07 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d09b65f5761f4628b8c14e265f15fb3a : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":4},{"id":"6f11389ab9d4fd70f648a05c62fb671e9aeb7e42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:35:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9a217a06d71d4e07cf5ea3ea4409d437f2fbaddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:36:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6db4ce6cde746e585cc03dd9af41a7e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9efe6efab68c4dd9ac5e0cb41602815d : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/365bdf07625c4ae78c6b3d0cf8a23db5 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cde06df3daa39d501acc0b6391cfffda7c111f52","revisions":{"b1542b548e8c8b655509075052a4678bcb50c274":{"kind":"REWORK","_number":1,"created":"2024-12-17 14:55:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/937882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/937882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/937882/1"}}},"commit":{"parents":[{"commit":"462f3a2b35a620a93dfdb14e05fe7cf0e3e84ddc","subject":"mpu: check session request timestamps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462f3a2b35a620a93dfdb14e05fe7cf0e3e84ddc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 14:45:37.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 14:54:55.000000000","tz":0},"subject":"mpu: store contiguous nature of manifest in sysmeta","message":"mpu: store contiguous nature of manifest in sysmeta\n\nIf the uploaded part numbers referenced in a manifest start from 1 and\nincrement by 1 then the manifest is considered to be contiguous (free\nfrom gaps). This property is saved as a flag in manifest sysmeta in\ncase it turns out be useful in the future. For example, the path to a\npart can be inferred from the part number if the manifest is\ncontiguous, without having to read the manifest.\n\nChange-Id: I36428e03ebd874892095e13a81e722a3d5550c2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1542b548e8c8b655509075052a4678bcb50c274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1542b548e8c8b655509075052a4678bcb50c274"}]},"branch":"refs/heads/feature/mpu"},"0cb7cad6843d2681bb7d3909911fe70eb4a6d227":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-17 15:03:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/937882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/937882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/937882/2"}}},"commit":{"parents":[{"commit":"e9dcb909bcb6c6127ff785e9844c259c15f9aec7","subject":"mpu: check session request timestamps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9dcb909bcb6c6127ff785e9844c259c15f9aec7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 14:45:37.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 15:01:33.000000000","tz":0},"subject":"mpu: store contiguous nature of manifest in sysmeta","message":"mpu: store contiguous nature of manifest in sysmeta\n\nIf the uploaded part numbers referenced in a manifest start from 1 and\nincrement by 1 then the manifest is considered to be contiguous (free\nfrom gaps). This property is saved as a flag in manifest sysmeta in\ncase it turns out be useful in the future. For example, the path to a\npart can be inferred from the part number if the manifest is\ncontiguous, without having to read the manifest.\n\nChange-Id: I36428e03ebd874892095e13a81e722a3d5550c2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cb7cad6843d2681bb7d3909911fe70eb4a6d227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cb7cad6843d2681bb7d3909911fe70eb4a6d227"}]},"branch":"refs/heads/feature/mpu"},"6663ea8ed8d4c072ae7934e77c60bec3af986d8b":{"kind":"REWORK","_number":3,"created":"2024-12-18 12:13:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/937882/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/937882/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/937882/3"}}},"commit":{"parents":[{"commit":"e9dcb909bcb6c6127ff785e9844c259c15f9aec7","subject":"mpu: check session request timestamps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9dcb909bcb6c6127ff785e9844c259c15f9aec7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 14:45:37.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 12:13:21.000000000","tz":0},"subject":"mpu: store contiguous nature of manifest in sysmeta","message":"mpu: store contiguous nature of manifest in sysmeta\n\nIf the uploaded part numbers referenced in a manifest start from 1 and\nincrement by 1 then the manifest is considered to be contiguous (free\nfrom gaps). This property is saved as a flag in manifest sysmeta in\ncase it turns out be useful in the future. For example, the path to a\npart can be inferred from the part number if the manifest is\ncontiguous, without having to read the manifest.\n\nChange-Id: I36428e03ebd874892095e13a81e722a3d5550c2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6663ea8ed8d4c072ae7934e77c60bec3af986d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6663ea8ed8d4c072ae7934e77c60bec3af986d8b"}]},"branch":"refs/heads/feature/mpu"},"cde06df3daa39d501acc0b6391cfffda7c111f52":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-12-18 12:20:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/937882/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/937882/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/937882/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/937882/4"}}},"commit":{"parents":[{"commit":"e9dcb909bcb6c6127ff785e9844c259c15f9aec7","subject":"mpu: check session request timestamps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9dcb909bcb6c6127ff785e9844c259c15f9aec7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-17 14:45:37.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 12:15:29.000000000","tz":0},"subject":"mpu: store max part number in manifest in sysmeta","message":"mpu: store max part number in manifest in sysmeta\n\nThe maximum part number referenced in the manifest is stored as\nsysmeta with the manifest.\n\nThe part numbers referenced in the manifest do not have to increment\nby 1 i.e. there can be gaps. It may be useful in the future to know\nthe maximum part number that was referenced in the manifest. For\nexample, if this is the same as the number of parts, then the path to\na part can be inferred from the part number without having to read the\nmanifest.\n\nChange-Id: I36428e03ebd874892095e13a81e722a3d5550c2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cde06df3daa39d501acc0b6391cfffda7c111f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cde06df3daa39d501acc0b6391cfffda7c111f52"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
