)]}'
{"id":"openstack%2Fswift~937884","triplet_id":"openstack%2Fswift~master~I26f75fb61ba0d13ca16810f13cc6cc193898211a","project":"openstack/swift","branch":"master","topic":"s3api","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-01-19 20:59:46.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-19 18:47:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-01-19 20:59:46.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2026-01-19 20:59:46.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2026-01-19 20:59:46.000000000","reason":"Change was submitted"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2026-01-19 20:59:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26f75fb61ba0d13ca16810f13cc6cc193898211a","subject":"proxy-logging: Add \u0027api\u0027 labels","status":"MERGED","created":"2024-12-17 14:59:36.000000000","updated":"2026-01-20 20:11:01.000000000","submitted":"2026-01-19 20:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":72,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"937884-s3api","meta_rev_id":"19da05547b94b69679c0de253126c1bb96df8532","_number":937884,"virtual_id_number":937884,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-19 20:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-01-19 18:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-20 20:11:01.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2026-01-19 18:47:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 16:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 17:18:44.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-11-06 19:14:01.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2025-11-19 17:29:14.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-01-12 06:09:38.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-01-17 01:59:33.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"2ebe5e61930169b1cf77a74126ef8ccf641789e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-17 14:59:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"998b836cb3385d471865a2b1fd8ad32d18b43dcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 16:13:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd7fa4cb32524c80b3187dc8508863d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ae37883c0142fbbf060357fc8140b0 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/3cb5add238e248e491db7571c3afe617 : SUCCESS in 50m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb11dd64950542eaba29078cdc9cd9ea : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6455354390914cb3987038127a7cd489 : SUCCESS in 45m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad394a2f2cf344af854ad11c07a4bc06 : SUCCESS in 1h 07m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb28bb59f549448a8b3c44d66109bf66 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/520b6111e78c4aa09a0913877eeb6ae2 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ff54614cdb44f848318c75bd9cadbb9 : SUCCESS in 23m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/802e89cc05e142cb91317a83c484b0cc : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a5276a83750465eb3e2e0788662b3e2 : FAILURE in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a90a85fccd58417f8a5bd5b394487227 : FAILURE in 21m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd480a9a71a94b29b1db266461c25350 : FAILURE in 20m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c501478643d41d3aa7400d0282f1b13 : FAILURE in 33m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25d2c86c0f2749cf89e23554c49bbd81 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d6cafec3eae42839adbd46e705ff244 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9a37c23576d46eb83c6e2ecd028b63d : SUCCESS in 19m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/db6a5c09b8d74fb09eb4a66241856fac : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b78cd19d0f514f22924447f5b34613db : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac1a865a362246f4a6d81c77a88d6e77 : SUCCESS in 9m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8613b6ab4b6e4ad58b7062ae0ea61908 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04f834d3abde490dbff59f3943a5716a : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ffe1ac19e2ea40b2919f985b68270167 : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeccf7cd7b944071b3da21086cebbac5 : SUCCESS in 53m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6083a7fb4f1442f281bf2f0e2ce2cd12 : SUCCESS in 52m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f912abc34884b77b501c84e8234685a : SUCCESS in 46m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68bfc7fa59d4402dbf5ebd216f4e1445 : FAILURE in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d793308b51cc47f09ed21af4231e4179 : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c359d83b23f144e9939c4ff00f2c41f8 : SUCCESS in 21m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5996e501e9e143db8fcd0b72f376d51d : SUCCESS in 51m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54fd955a573c42f4bb63db83c86fb5d9 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":1},{"id":"93a20b14c4ef6216d1ef5a13287413e3ab92e6c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 20:12:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cd3fa4c99240f2bf46e523571d6a99\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6891a48353ad421dbc4012138cba7fc0 : TIMED_OUT in 3h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"68a6bd009438be3c352046aa2e920d05043827c0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-15 15:22:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e2f6f5889039922a9452bc8d0eab5da3bc09cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-10-31 13:59:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"53a90b108fadf2e3fb530f7055ae0c04898036dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 14:59:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55c9dd9e548f4f77a3ceb60f3137fd19\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4e36a35077e4b6f967addf821b886dd : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":2},{"id":"8078ef324eefc892ca36a7687a59a560c47fafdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 15:10:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9562e97ad04007809f7ccaf84e25cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0cb72598464aa59ecc10da29bb1e96 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/38e8c6e068214f798df0bb62ddd0d5d0 : SUCCESS in 1h 03m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/702ac50d8c424a6995f1822166b82b6a : SUCCESS in 55m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/177264b4358543839a764a43a81504fc : SUCCESS in 1h 09m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/098ff87dd1994ff0bb014842b713185c : SUCCESS in 1h 04m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c2de56e58da49d684f40429137adfcd : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/169af37be9b548789800c84371a8519a : SUCCESS in 11m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/169df1177b30420caf06adfc438fb278 : SUCCESS in 12m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7cf91e3100714eefb93770dd56466126 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c1c75a2efac478f8f5a4667b7edec92 : SUCCESS in 10m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1a24ee1ac3064b0f929e7a12db7a573f : SUCCESS in 8m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/56a802cebdc24e668b872c312b05aea4 : SUCCESS in 11m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/25bfbebd39e4441faf1d0d863c716318 : SUCCESS in 27m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1cfe399861bc491ba0613a2336ba8a9f : SUCCESS in 8m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/67faf36a3062488f8a1978b8672c4e90 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/946b8e1967ad40258e457fc3a7d5d721 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b106ac80451d42fb9ba601e87d3d68e7 : SUCCESS in 5m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/419b27353b244117b47a2ff3a5c4e070 : SUCCESS in 9m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e2104342adc4dba86e8151dcbcf01eb : SUCCESS in 8m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4876783956b49eb94400f788b28eb05 : SUCCESS in 38m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7802119c9084f539f42a4867a075ec4 : SUCCESS in 22m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8571a59c7e14440f844e19317ab9a424 : SUCCESS in 55m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad69ccf90e40454d98eb4d2e55138155 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf4b9c27350411e8d2bdf97351c7c4d : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/665af53b4b8c45e782187f180b0b5630 : SUCCESS in 18m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8a02b87621640deb24c18e21e4d6f03 : SUCCESS in 51m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ce8bc4806e044e3b18dfe5c9a59747b : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cdb2093ef5f99eda5b9a93018928b2ea188674d0","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-11-03 15:23:59.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ab6ef665aa92b5ca5bebf69e49d8b7af9691c0e3","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-06 19:14:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e97bfed03c7e5d99d65729ea04dd3df911e93ad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 15:37:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e4760a62cb7e6ab3b38d820bb1c58923dfdc3425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-19 17:21:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c500d6a32422689b8785743bc7f0261573e8e7dd","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-19 17:23:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a88b01b3f0d18fdd4b92828733182bcc85918238","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-19 17:29:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b2f2099a102850217dae786e7aaf3eb04ce7a62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 17:56:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb0053e9e4c44e39da321010e1d0949\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2db76da4cbeb4234b03f4434a6682b12 : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":3},{"id":"335ae672ff7b1f21a10fbd07b6df3af64bae2aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 18:30:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b481496a9674e7eb8a3132eb488a8bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8ac3623d7745c7bb8f8f88af88262b : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0159265d48f94b8a92344447e71cd9d7 : SUCCESS in 1h 02m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a64979ec3d77449c8825c0bc03d76e01 : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f0ba6fd8f684f2e971bbc024272c4fc : SUCCESS in 52m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e876da1711a34d74ba297f77b1c2d92d : SUCCESS in 49m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86681910607948bba4d50136df2c059e : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93ef8ee969754bea8fcae75c553fe7bf : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3bf3bd90a0784b4990ae6603b18b07b4 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f29d646cce304045b277e66b1d17bee0 : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba04fae3d7cd4ed39c1d408590900ad1 : SUCCESS in 24m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f6f6801bd9c84fdbb5ac69f982531030 : SUCCESS in 17m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/64e1e386a859462a863d7b540e543599 : SUCCESS in 23m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/273e1ae04e384acb932be1dafb765c93 : SUCCESS in 24m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a07e88a5b7f247498feb4eac2110cb6a : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/681ec4356586420aadcc1543540a2647 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d3b81c8e820a442bbb2adc2e9275fd65 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2082d4ea302f4da1837a9d850ec80aaf : SUCCESS in 7m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69bdd63fe9354f42be508f11854fb921 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b514b3d0fe3040dcbd093d94a1bd91f5 : SUCCESS in 9m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3062f07e65b349a9bfd699e3db38ffd5 : SUCCESS in 1h 00m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b88190c73ce422ab82322cbac0018dd : SUCCESS in 26m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8498ee4fa684122b9f8de76699a728d : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e99189c694344919454331a431e41b0 : SUCCESS in 33m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036105edbcad43b4a25ffb7be8735b47 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b5c01c11e364ac2908511b583983058 : SUCCESS in 11m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0c1edf532e543c9bed2ea9eda64f2db : SUCCESS in 1h 00m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56979b0ce52f4c08b66c9d0465dead21 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":3},{"id":"64afcf9d3a3f9912ddbc7eef0f136c0ed9c5ded3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-20 20:50:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e2847c6f8ca4024291fffa78352c4804f894a6f2","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-20 20:50:42.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80d29ab178e816b1de91c7e0da58fb823ab1ebb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 21:54:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea8ed53dcd640588352c335b5ae45ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41be2377a4104cc2901297d55201be1c : SUCCESS in 6m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/6bdd135803e64f099353df3845942918 : SUCCESS in 56m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02b28b79aafd42fbb65c1fa9329bae6b : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59c773ad74c8410fa3bcc01b4a7f6df9 : SUCCESS in 50m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c8218f431424b9399dcb1d395ba5b20 : SUCCESS in 37m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d304863980a74e27b75aa3211d3af49a : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94586671a1694d87ad09d3e022fb1e16 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7a83160db524337a6816c3a118044bd : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30339ff88bd848ceb54db4e51017e2f2 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89c16ab0053145ae94873054e6a9ec56 : SUCCESS in 25m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16496bd3c6d8403b9d09a0df70a2d2f8 : SUCCESS in 17m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e67172107d65428b91d1cb50ed4e0c64 : SUCCESS in 11m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f3e6b4f034aa4818bcdaed9f6ce60818 : SUCCESS in 14m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/20c8dc6fc64d43b09f9016c2e87c71d1 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0b472915921c48ff891c5c59722e3e27 : SUCCESS in 12m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d75bb873311a4bba88dc3dfaa589cddb : SUCCESS in 13m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e23747fcf8514e049ab497fe7a41d12e : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3be57da1b647491f89cbe2784dff8582 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1267f874d6d4f4c86a64ee714c9d1cd : SUCCESS in 5m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8462e899dfc040fa8adc6607adefcc0c : SUCCESS in 38m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e74fd10fe42942abb0d41ca3ef74ef14 : SUCCESS in 47m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ead4219dff04708a18fc9b3a41af636 : SUCCESS in 37m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02c3a8bf14f9489a955abb65cd0be936 : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01bcb2d35974aeba5a088a89552672e : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01705197094b4690994b2c1ce41f2aaa : SUCCESS in 17m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b5e9cf5e070496b941bff9d20eaca4d : SUCCESS in 47m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bf0cab984834b2d965da8186e6360de : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":4},{"id":"24d877e7941ffb6f485747b8e247d55705e1b360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 22:31:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2b350a497a472989697291d4cc0147\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/912e707b6ef54a64a86a4d244355b7e4 : SUCCESS in 1h 37m 51s","accounts_in_message":[],"_revision_number":4},{"id":"96bd26a4d198a1b8992a1664d5d34d48e5ed6e23","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-21 13:16:09.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bff0148f5f174579413ef9a4dfaf20173ba8609f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-21 13:16:21.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"04b77dc4bbf1dad005336a82ed3bc5489ccd0273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-24 17:04:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f5165eab063e1e0b3dd928899eb40e90a04ae348","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:17:14.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78bc470fd0fb4cc9990a9ddf4c069336\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f6669d9d14945748af28c9bd2c1cbd1 : NODE_FAILURE Node(set) request c982e9e312fe4875b57a51f335f43e8c failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"2d749309212995855970187ed1cd21da9d5fad53","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-11-24 17:25:57.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e1581afbbba74a4f03e2152b092215f3e0fe7ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 18:20:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b7eec38ddff43d5b354967d7d78f4f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ee8607664541c4a1f6a6a5fa3689c0 : SUCCESS in 5m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2ebe1d9d0d07450aaac2e03347cb86aa : SUCCESS in 47m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f2c8f1701024120ab7f0ed0cc065e38 : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04344e5cd8e54e96ba08d1e460bb884e : SUCCESS in 37m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e72b7f794b8d4b09a3f46bcfefe11a65 : SUCCESS in 33m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d98cd6af8e04c4a906a503e14b2fb0d : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1beed423b934660a27a36a13fb08c2d : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7363c08d6ba4839b1a1123ad012755f : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/432aea2425254bdc80748a6ee4379473 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05f4d3d707944fa5ac7cc1c42d34d8c3 : SUCCESS in 13m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97e8137d0f944615a235ed409fd41f9a : SUCCESS in 17m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6fde8df4b4d04825a3c69e438f4e8128 : SUCCESS in 12m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c448ca0176e845998282a9d80ec69675 : SUCCESS in 11m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7552db36b6e04393b40579fd74e05c45 : SUCCESS in 8m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22f498c47c7a48cf8d051c9ed8d4b505 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/78c53faf6115423f9db9a8ac915d696f : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b69df6af81c4b00a09df9e583d04e62 : SUCCESS in 7m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a677e591e7bb4cfeaee93ad8da491c9f : SUCCESS in 8m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad4b6da4ee5240789493d2febf5c80c3 : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0112b5c92b464af09fcac4f7aeea2eff : SUCCESS in 34m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a90a53036b52408e8fd21c90998ad7f1 : SUCCESS in 51m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10089e4cccef47548f27401b97f600a0 : SUCCESS in 1h 07m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac6215ef99b543e38f441a84d6fa4a55 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3815c8fc80d84e1fbfeb4ee3fc7ff7d4 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1afb539a09df474f803b010b4612085b : SUCCESS in 25m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2b72c0b05ed4c45a07532f75b6d59c6 : SUCCESS in 43m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ca347a65f994f91bd41ccd8bea88b94 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":5},{"id":"f2e39ebb9ae2d5058bf2b23c976047a1f4842eb6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-19 19:00:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dcf2c4cfdce8668b75b7dbcfc911fdc7d1d72531","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-22 13:59:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f597e9b974dd5cb90fedbc29509d07edc0fe3c5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-22 14:55:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"52bbad21684dd5adc89619ad8e813a790b16c00e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-06 17:37:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c831186981d1063261b7510e9a35a1cefc1b59a4","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-06 17:39:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"48de3fe27ceed29e98759355d272707150e0e0fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-06 17:44:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8d6abc0deee85fb465d1687a24679268ed5c09ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 18:20:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f3ff6d5ebe4b4cabc98a4722d174d1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/849ac17f3f2e4f6089cbe0c6f8c52ae0 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":7},{"id":"1ef3ab7574d556397289101c927d57dd2d435d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 19:01:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c207b27666e4080abec87826ad5f939\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01b558e794e4b1d8afb5c5a0a5d217e : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/f9bd0c05725c4d51a0fa9ffff90e0976 : SUCCESS in 52m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb2d11b606364da5b788bb6a0e8ae3de : SUCCESS in 1h 05m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d08363cba59646d6b2d6ceb494f48d1e : SUCCESS in 47m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec8c2c03798440b5bdbf51fcc247e84a : SUCCESS in 45m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9741809d3a8a43f89dab379cccf16726 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ab3a43a97cf4a3ab5ee5ab9d8791738 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc614846cd124abb92901e4f7d27336f : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86902b5b471e4d0585a25724d3220c43 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad164d8f6bd44fd9a336a33f1a03624a : SUCCESS in 20m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b9e440e2f884a10b3b185ce4d2ef6d5 : SUCCESS in 10m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e2b3041b73764d70adef8e3724f49133 : SUCCESS in 26m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b8c83170cbbd40a18d312578956e3867 : SUCCESS in 24m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/10166ae2e67e4c3682836d02535ed9f7 : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/923734a8a39540c7b1662a996227ed37 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/395022876bf145878c0cd29d4043c0e7 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e732b4df7cb04822864b6ffb62e88173 : SUCCESS in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02522ffd359649f281dcb9c718a1343f : SUCCESS in 15m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/68ca260434554ea0827a685a9f6430ae : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e000821fde84db8854c14fade0e7cda : SUCCESS in 57m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d446e0b1e45944b5a07e6aa9aa1c00e5 : SUCCESS in 55m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e77967c30d184fc8948c144828da6282 : SUCCESS in 35m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e74c446ed8a4b4fa06c0f27e9f6dc34 : SUCCESS in 26m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cdd5fb774c4d0b937527d2136728b3 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36a8bc079ff44afa97414e67a3138577 : SUCCESS in 25m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2c9a72d81564219a6a228296468da46 : SUCCESS in 49m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64d5654b6f204772b30dc39081dcccd9 : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":7},{"id":"0b6eaee5d47661e7e11a4fe41515c5c0ce76737b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-09 15:22:24.000000000","message":"Patch Set 7: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4d2ed402d1a5c48422e73b015c5f314d4fc9f270","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-12 06:09:38.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)\n\nHad a play and tested the patch and it seems to work well (see comments inline).\n\nWont +2 yet because of Al\u0027s comment.","accounts_in_message":[],"_revision_number":7},{"id":"fff62157ad06e84b6db0fbfca46bffb237845c5c","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-14 23:00:47.000000000","message":"Patch Set 7:\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b0affe2ea1e58809bf79a96722c5dbc881999d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-14 23:14:33.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"175c92354759b25692c4ae2b3957bdc23221cb40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 23:58:30.000000000","message":"Patch Set 8: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d13ec556a1f43f6b4977248e3534c6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b05eb4705f845d1a531d1f53c20d0f2 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/1eb5aa3afa2a4b2f8a115bc7e2910e33 : SUCCESS in 43m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e6619077d2f410e83a30680daa20c0b : SUCCESS in 29m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dbb4bdd3168b496b867b73e54d4cdec1 : SUCCESS in 32m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fa44d6a9b064423b3e3ae66efd8aa54 : SUCCESS in 33m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b83c367014e4963993196d55696bfa2 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adc30a37657548dbb8d579a1778992ea : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/350cd4664e894b4c9e2b9cf9fb9fe642 : SUCCESS in 13m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/545e5165d0d34c5b91d5fbe12f57f6b6 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bcdd15470e3c4b779de16d4eedef1e32 : SUCCESS in 12m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04447cb7e19a4ca193c990219d8200ef : SUCCESS in 13m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/86b4c57c124a4613969ee1d566df8c26 : SUCCESS in 27m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e89487f239c344c7bbdd50c47c2fa90a : SUCCESS in 12m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34028ac584f94966b752cfb5dc4610a5 : SUCCESS in 11m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/56b51240cdb1477392d25c60c93983d5 : SUCCESS in 8m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3b283aea418e4c76934e9eb968017528 : SUCCESS in 7m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/624669fc819a42998aa1a7eb5bddc937 : SUCCESS in 3m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f4455bb18d7475a91de6fe9540064f6 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/936750e136c84ead8242cb8a7ea167b1 : SUCCESS in 9m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c8ae6d63c334446a3e17d5751277102 : SUCCESS in 34m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/948345841730486ba813eee21328ced6 : SUCCESS in 33m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2efd1a15c497458fa01c8e5f9c8755a9 : SUCCESS in 33m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50f0aa16110848e08e26e6a5cebd922b : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdfa8cba9f440d290f550077c5ca08b : FAILURE in 2m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c73e1aa1fda47c7a3a7b0e38f6a5bd1 : SUCCESS in 16m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbf2730d598145a5a1a00bcbc8354192 : SUCCESS in 38m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee3b57296cae4d66a436d7839f4377ce : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":8},{"id":"fa24ad228b7b3454cdf30b0c3931f0f062b6462b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 00:19:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76666a9a3079419c859934383dd6d030\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/288f8855d31f417c84be7334f558d876 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":8},{"id":"a95854d8975991c7b36bb17442287735ecb5fb4f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-15 15:00:16.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"728ab82b24e8f166d717aa8f0beb2143b2e1d28f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-15 16:16:38.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5b18c5cb75a2232f15e3ab4fed07f4640ba11eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-01-15 16:21:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cd678c2146c44078c07b6b0f53ffd79b53a659ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 17:26:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba6468c8c38949e5ac2590356ce7e9ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a128c0387f4e399196f3517e107878 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/efaf93cc121f44de8d4b68b34504f038 : SUCCESS in 1h 02m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c80a31645f74953822597ce02451fd4 : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cef644463a9b4961aa7215f2921f644e : SUCCESS in 55m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1aad62649cd4046ab261862ad51d0f1 : SUCCESS in 1h 03m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6415be8257a64adfa15ced3b7a8a357d : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4790fe7266ee4a56a54ab1479e820ed7 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2ca30be79e347c5b2035675b4bbc1b1 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7086c4b564247aea0e64be62b20d140 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1304e1d017364f519369feb46f4ec60b : SUCCESS in 11m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6981d2cd6a44367a1c5af317214865b : SUCCESS in 24m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e40ebb4ebabd4595bb69169039d94485 : SUCCESS in 44m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/63e188e3a83543ddaedb8510e1804a90 : SUCCESS in 28m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/99124bff74504f4c8415cfce595ef5d7 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/12e0a06b330c4c01981eefe106a143e2 : SUCCESS in 9m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/44660d0db76d4fad93964ebf8ad723d6 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7287533b0fc4ba7a6906a8249ae775c : SUCCESS in 4m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/465993f23e30416a87de4299300b13ea : SUCCESS in 14m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/754f8baa00d943a2a8350d699e0644b6 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/748a813b5b6f44bbb98b2a67b47121e5 : SUCCESS in 51m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5860582935e6445183174b980eae47a2 : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96446c5271bc42abae94c96c958cbb48 : SUCCESS in 56m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e20da5a743d44a9b96ae651962b52bda : SUCCESS in 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a31266351a04a3c8ffbd5c1d8b38ae9 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/703d0062dc0a43498ff31930f9e491ee : SUCCESS in 25m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/453becaaf4644c07a864683362056db4 : SUCCESS in 34m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b99d1465c4a544a4a57655ebbfd17540 : SUCCESS in 31m 28s","accounts_in_message":[],"_revision_number":10},{"id":"e607fb150010bfd053561acc9b709fdf98e0208d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 18:01:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5bcb8a416724a92b938b6a342a854fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d83a94c8f7349b793c459b91b40ff58 : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":10},{"id":"48412623973e9afbf2edddb2df6a288a52a890ae","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-17 01:59:33.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7c7ee54064af8e3303a4941ddff5807ddba60daa","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-17 02:14:04.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ec48342600b6d8f19cf3186c0fe2e5d7a2995c6c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 13:13:23.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"089fd0ba5ab96a81dc723a8a1a138acd96380c9c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 13:13:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"24b6f6ef50e45fc3f091568e936d0689b76c87b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 15:55:08.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"0d346d4543434daa467d4402682a360d7f73330e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:01:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bbe8e51c7bb457db5463ba6b595b05f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7dbe7660884d6f90c34b985f0c2e86 : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/60ec467cf1d14230a67cab4d8ec43c58 : SUCCESS in 29m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25849de2806c432686d1023f6183c8bb : SUCCESS in 59m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e0f864b7bc744ad99582d823d89860b : SUCCESS in 54m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efee9f9f359f49079c679e6fc230f9a5 : SUCCESS in 47m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c4a302fcc7642408b4f919bb5b0a0ea : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a79a6406f5284003ac7c328e4ffbfe0c : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/addfbab58f714fd68dbf4b6dfaa2b696 : SUCCESS in 12m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c676e5a1525409ab4f18ff191e238b2 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb4d49701be84c11bda2bedd9abb2a50 : SUCCESS in 18m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/80e35b10145740259dff26d0633fd8bc : SUCCESS in 9m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/032623ec443c47f8987528caa50d0d52 : SUCCESS in 12m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3202c820455c4654877c3e9c47a65524 : SUCCESS in 24m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/109480f3ddde472cbdb10ee8a9ae48b2 : SUCCESS in 10m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d28975279fdb40b7bf3f8d053668b192 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6dbc07dd5ff44444ac7b6b566ad16bb5 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa8b58d314d84ed18d686282de563010 : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70765cca4ad044b7a72798a0dcacab59 : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c6582a60df064b96891f98b9af04c20d : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e76427d572fe4bc19caaa16ae5d3d323 : SUCCESS in 50m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/416acb91a1bb47649a53141362d6340e : SUCCESS in 39m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7e2e365f5514e8db30b974c00d056da : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/460fc9ffc8cd44cd90dffd0bc60f3b79 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7599fb43801a48068351c61e3b0f0158 : FAILURE in 2m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bfba6ee254341c981d27921f0a1378c : SUCCESS in 18m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce96c31525154e30ac6e6cffb35ccdcf : SUCCESS in 1h 00m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f109afb462e43c5b26ede15d7d7a326 : SUCCESS in 51m 37s","accounts_in_message":[],"_revision_number":11},{"id":"9f574f51d25488f28427ecbfa45fd5e846c3c3a4","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2026-01-19 17:22:29.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a14e3e91aeff1308669224ef24166926dbcff311","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:50:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab3898ddfeb4bd3b0e525aaac9cf8a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4b6b46bf89347c5b89cdc7ca72c0f21 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":11},{"id":"a4451898e3d9364be0af758a921ad812c9cd03b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 18:31:47.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/726263a7faf44745b58417f62f7ce0aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41794dab996d424f9c2d264c0e509faa : SUCCESS in 5m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/013315669b3547e7b20be8dd39bf8755 : SUCCESS in 38m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8fef4b0fa5a4c738436457515fb29ea : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e19b52039ba648e4a29b71da73210266 : SUCCESS in 53m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55537a893aff4f30a4f8dc2a8a75e2a4 : SUCCESS in 39m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5e8f0bb329542089387b775f635b6b4 : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1daa474caa5e41e999ff6cf8b560dd5d : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60b3054fd78a436a963dcb390de3b4bd : SUCCESS in 13m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0d2dbd4a26e47a0b9290a27d5d74ec2 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9ba551af44f44dcb15d5660b9d2283f : SUCCESS in 23m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e393727ab1254129b288e883ac8032a9 : SUCCESS in 8m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/809f8781ae9744bb9f347bb1a28e3495 : SUCCESS in 14m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0f3647f4743a4786bb40968c1a2e2ac7 : SUCCESS in 31m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/065c73c74d06450db1933ff91e57b174 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/32f4b277e6244b83a1542358f4a1dc90 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1a894edcf21b409c86038ab5d6deecd9 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a6263abbda04b11a58d3a6a9831f840 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8866f55e0e4e41e58d8546e4a3cbe522 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f22fec8056f4bfabeb67969d51c18b9 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/641af41bc0d94bc5a80083e33ff9ab2f : SUCCESS in 39m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17d6128378774edea5b9fe2530eae137 : SUCCESS in 50m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2bb81bc0ffc54bcc989329904feb3017 : SUCCESS in 52m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5ed62a3f1984f3abb02f2bb2bc96172 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cdce2675d84092ab0c8811b1368900 : FAILURE in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a50f15551204fb993b8e91258298f85 : SUCCESS in 19m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4aa7ac6ded024f78b7243ba318aade66 : SUCCESS in 34m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0bf1f022831445d82d80beaed0647cd : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":11},{"id":"0b3238d3a757cbe86eb01dd1daeb535bad2a7a50","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:43:01.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d88df20f8afcddf358012829540d753c099d04da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:45:05.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"95a294670d3745fa28f35a72e8d80057c9c32832","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:47:27.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"04bef2bd124ec4d1c6b003252eb2d93d2cdbe1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:55:42.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f377d6ed01c4892a0a33642e9917339\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f56b99341441e696fab6d2c4bc48f7 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/08d982b5f7834f58b4e125d303f887cf : SUCCESS in 45m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2fdd17db3504454bf0c1313605cd2fd : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/559302b03c37419eb11e2dd7aaa3b484 : SUCCESS in 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/344988c323bf4fe5ae40bb9d1e334808 : SUCCESS in 34m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea598ed4b37f4617ae780d64d524c8a1 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/657e6fde3e7e4109a65a632366a24c72 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27b41d690c99415fa23bad833b0d092a : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8b97295238c4a3faa91af7317751374 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/57caa1b525c6408dbbd592e26e53d316 : SUCCESS in 25m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d99d13ca09bf4c1ea444a971987e717f : SUCCESS in 9m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6deae4dcff3142d09adbf4814fe36607 : SUCCESS in 29m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/89feed3886f547bc987432e5242a7b2e : SUCCESS in 20m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/19722a6a049b49a498fc5a7b8dcdd215 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/32e52adaebd9489594f4ab7083cc8c51 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c5df65ac5e274ded821da2e76a162ffd : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eccb42f75cf04f09a64e95d2d1e6f26f : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6433ad54ee0c419082ac5e6ea9acb670 : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aef351755fe24cd8bebf6596d8dcb5c7 : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b06e1ebc089465592249290496aa305 : SUCCESS in 57m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c79eb6c905d34cfe8e3d93a45b6a018f : SUCCESS in 54m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5b8a1b4b9894dc29c9e0e18b9ffdf11 : SUCCESS in 47m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dee8f48b00304f5d88f2bdda0bde917f : SUCCESS in 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e17e8eb4314459a363a60166b8a0c4 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/459e77f759e646bd922ce1d5be662283 : SUCCESS in 16m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/583e2774d3ab4867a0f53e35f6ddcea5 : SUCCESS in 54m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b065e03aafcd4047b7402a4463b6314b : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":12},{"id":"b1d38e273d93262bc2616ffc90414eb9ebc595c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:56:07.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"bd906459c53747480acf75cf5b7303fe87b4532e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:50:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39adea07ddfa4adfb55ef37e0e9f3477\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13aff47720f440f986fbe869550c95d4 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":12},{"id":"0dfd38668e00b8e2b5b38444ed8f579aea78249d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:59:46.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ccdfc0b142541f68d79f6b5f66ebfae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f092481833d46268ea79444ddeafde4 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a6432556485c47c7b7a9bb141615c47d : SUCCESS in 1h 00m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/586a7c84d6c4491cb5f49c6288fde6f6 : SUCCESS in 34m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f82ae4342ee34e3b993ee512723f28de : SUCCESS in 28m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78982e10b41e4cd5b4842c282271b576 : SUCCESS in 58m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bf2f67712f20414b96fc51ebe08466ca : POST_FAILURE in 6m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a421a449acd4c86ac7353e34c8979b9 : SUCCESS in 18m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46f8ccc0db27406d84583138ffc86ba9 : SUCCESS in 17m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ebdbd771cb9241bd9342f56a34613be5 : SUCCESS in 21m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2b4fdbed2eec4bbf9004a3a4cd15bf7d : SUCCESS in 27m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ced33d49d79549ffa4ba5854198cf94d : SUCCESS in 8m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/af22bf6acfd04527831a56d4da7653b5 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e1a732a124aa4596b6bd3cf96c254de7 : SUCCESS in 8m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/950530081259414295666dfb044db984 : SUCCESS in 7m 09s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a32649ace8564017bc5985b80af64299 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/210ffcc99d0d438781bd01db3eb4db2a : SUCCESS in 57m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66c86bbb51634107a0b8f0e94feaada6 : SUCCESS in 47m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c349fee62a8e45d88649c9de7fb9d93e : SUCCESS in 49m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e430fa88889541a48a9f58ede12b7699 : SUCCESS in 21m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa31c99b23334e4cb428177a84882a83 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10e161921e6e4afebc6e71e5972101fe : SUCCESS in 18m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c31615954ca435bafc44edde65cad21 : SUCCESS in 53m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83cb76d1a6df444cba6cae17ddad63f8 : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":12},{"id":"d2fa36145fec7dd5767b601aef75be41d05836bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:59:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"2668f96d9ea2c24fd0c7883975566849cce7670f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 21:00:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3da43a4091a42df8d90c53f9a306aeb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9941f15aa0c43f0b831984375de2589 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a8125f09ed01489e92ddcf76b100d42e : FAILURE in 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"19da05547b94b69679c0de253126c1bb96df8532","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-20 20:11:01.000000000","message":"Patch Set 12: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"64bde7e05c258322864703b55042455d4d3c48ad","revisions":{"f12623985c89df6ff695c5fee4acbb21197f2c2f":{"kind":"REWORK","_number":1,"created":"2024-12-17 14:59:36.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/84/937884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/1"}}},"commit":{"parents":[{"commit":"1af3d28b70a02ca2ed62f9252b527b1efb018b60","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1af3d28b70a02ca2ed62f9252b527b1efb018b60"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-17 14:36:34.000000000","tz":-300},"subject":"proxy-logging: Add client label if original request is from S3 client","message":"proxy-logging: Add client label if original request is from S3 client\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f12623985c89df6ff695c5fee4acbb21197f2c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f12623985c89df6ff695c5fee4acbb21197f2c2f"}]},"branch":"refs/heads/master"},"1a16a95a0b3c905075dbf7b1237ff21299bd9b6f":{"kind":"REWORK","_number":2,"created":"2025-10-31 13:59:17.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/84/937884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/2"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-10-30 21:26:18.000000000","tz":-240},"subject":"proxy-logging: Add client label if original request is from S3 client","message":"proxy-logging: Add client label if original request is from S3 client\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a16a95a0b3c905075dbf7b1237ff21299bd9b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a16a95a0b3c905075dbf7b1237ff21299bd9b6f"}]},"branch":"refs/heads/master"},"ae0a8bc24fb35ac2b17eea9ae027b43b8da15b00":{"kind":"REWORK","_number":3,"created":"2025-11-19 17:21:06.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/3"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-11-19 17:16:24.000000000","tz":-360},"subject":"proxy-logging: Add client label if original request is from S3 client","message":"proxy-logging: Add client label if original request is from S3 client\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae0a8bc24fb35ac2b17eea9ae027b43b8da15b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae0a8bc24fb35ac2b17eea9ae027b43b8da15b00"}]},"branch":"refs/heads/master"},"398a19d33ae09cb8abb0145a00f3a797080b1e98":{"kind":"REWORK","_number":4,"created":"2025-11-20 20:50:15.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/4"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-11-20 20:49:55.000000000","tz":-360},"subject":"proxy-logging: Add client label if original request is from S3 client","message":"proxy-logging: Add client label if original request is from S3 client\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/398a19d33ae09cb8abb0145a00f3a797080b1e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/398a19d33ae09cb8abb0145a00f3a797080b1e98"}]},"branch":"refs/heads/master"},"c5bfee1573707d83d7e3401f68aacc81cdd427fe":{"kind":"REWORK","_number":5,"created":"2025-11-24 17:04:15.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/5"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-11-24 16:56:41.000000000","tz":-360},"subject":"proxy-logging: Add \u0027api\u0027 label to determine if the original request is from S3 client or swift.","message":"proxy-logging: Add \u0027api\u0027 label to determine if the original\nrequest is from S3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5bfee1573707d83d7e3401f68aacc81cdd427fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5bfee1573707d83d7e3401f68aacc81cdd427fe"}]},"branch":"refs/heads/master"},"0562f32d72060794237da863d993b69076e1089e":{"kind":"REWORK","_number":6,"created":"2026-01-06 17:37:51.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/6"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-01-06 17:37:34.000000000","tz":-360},"subject":"proxy-logging: Add \u0027api\u0027 label to determine if the original request is from S3 client or swift.","message":"proxy-logging: Add \u0027api\u0027 label to determine if the original\nrequest is from S3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0562f32d72060794237da863d993b69076e1089e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0562f32d72060794237da863d993b69076e1089e"}]},"branch":"refs/heads/master"},"514328e2a7f4484b45129e37b121aa171f0ccb03":{"kind":"REWORK","_number":7,"created":"2026-01-06 17:44:47.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/7"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-01-06 17:44:29.000000000","tz":-360},"subject":"proxy-logging: Add \u0027api\u0027 label to determine if the original request is from S3 client or swift.","message":"proxy-logging: Add \u0027api\u0027 label to determine if the original\nrequest is from S3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/514328e2a7f4484b45129e37b121aa171f0ccb03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/514328e2a7f4484b45129e37b121aa171f0ccb03"}]},"branch":"refs/heads/master"},"a5040eadb1037a2f5d803d89bf91b035428e83a2":{"kind":"REWORK","_number":8,"created":"2026-01-14 23:14:33.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/8"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-01-14 23:08:58.000000000","tz":-360},"subject":"proxy-logging: Add \u0027api\u0027 labels","message":"proxy-logging: Add \u0027api\u0027 labels\n\nAdd api label to determine if the original request is from\nS3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5040eadb1037a2f5d803d89bf91b035428e83a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5040eadb1037a2f5d803d89bf91b035428e83a2"}]},"branch":"refs/heads/master"},"27a0b496119e7a38b3a47c4b69d32e60dfb284bc":{"kind":"REWORK","_number":9,"created":"2026-01-15 16:16:38.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/9"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-01-15 16:16:23.000000000","tz":-360},"subject":"proxy-logging: Add \u0027api\u0027 labels","message":"proxy-logging: Add \u0027api\u0027 labels\n\nAdd api label to determine if the original request is from\nS3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a0b496119e7a38b3a47c4b69d32e60dfb284bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27a0b496119e7a38b3a47c4b69d32e60dfb284bc"}]},"branch":"refs/heads/master"},"510200509cbb99c11a1e578543a42a11415eacbe":{"kind":"REWORK","_number":10,"created":"2026-01-15 16:21:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/84/937884/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/10"}}},"commit":{"parents":[{"commit":"dcd5a265f6f36551e0bc0feac0e08bfdd09301a3","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dcd5a265f6f36551e0bc0feac0e08bfdd09301a3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-01-15 16:21:23.000000000","tz":-360},"subject":"proxy-logging: Add \u0027api\u0027 labels","message":"proxy-logging: Add \u0027api\u0027 labels\n\nAdd api label to determine if the original request is from\nS3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/510200509cbb99c11a1e578543a42a11415eacbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/510200509cbb99c11a1e578543a42a11415eacbe"}]},"branch":"refs/heads/master"},"0f3e11176ae4264ab4a9b3134d2599c7ef92dcd6":{"kind":"REWORK","_number":11,"created":"2026-01-19 15:55:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/937884/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/11"}}},"commit":{"parents":[{"commit":"72191716d26291e17f46c9e2bd3c0e169a6f3004","subject":"Merge \"Recliam db_dir/*.tmp files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72191716d26291e17f46c9e2bd3c0e169a6f3004"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 15:54:36.000000000","tz":0},"subject":"proxy-logging: Add \u0027api\u0027 labels","message":"proxy-logging: Add \u0027api\u0027 labels\n\nAdd api label to determine if the original request is from\nS3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f3e11176ae4264ab4a9b3134d2599c7ef92dcd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f3e11176ae4264ab4a9b3134d2599c7ef92dcd6"}]},"branch":"refs/heads/master"},"64bde7e05c258322864703b55042455d4d3c48ad":{"kind":"REWORK","_number":12,"created":"2026-01-19 18:45:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/937884/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/937884/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/937884/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/937884/12"}}},"commit":{"parents":[{"commit":"72191716d26291e17f46c9e2bd3c0e169a6f3004","subject":"Merge \"Recliam db_dir/*.tmp files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72191716d26291e17f46c9e2bd3c0e169a6f3004"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-16 20:22:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 18:44:51.000000000","tz":0},"subject":"proxy-logging: Add \u0027api\u0027 labels","message":"proxy-logging: Add \u0027api\u0027 labels\n\nAdd api label to determine if the original request is from\nS3 client or swift.\n\nlabels[\u0027api\u0027] \u003d \u0027S3\u0027 or \u0027swift\u0027\n\nChange-Id: I26f75fb61ba0d13ca16810f13cc6cc193898211a\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bde7e05c258322864703b55042455d4d3c48ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bde7e05c258322864703b55042455d4d3c48ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
