)]}'
{"id":"openstack%2Fswift~937990","triplet_id":"openstack%2Fswift~master~If6248bd85892738c891c25ed5b2e5fb4a46a2193","project":"openstack/swift","branch":"master","topic":"p-probe-test-reconciler-versioning","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-01-06 11:35:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If6248bd85892738c891c25ed5b2e5fb4a46a2193","subject":"object-versioning: refactor helpers to be module functions","status":"ABANDONED","created":"2024-12-18 14:04:31.000000000","updated":"2025-01-06 11:35:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eacb4bb5f7c83730457cd7db143813f437641ea6","_number":937990,"virtual_id_number":937990,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-19 11:38:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-20 16:21:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-20 16:21:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-18 15:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-20 16:21:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"6d5f6f833b212d7c518b57f1d1c0811e52d2dbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 14:04:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc536028711d4300c10fb4d37892acf9cdc3d6ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-18 14:10:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a347e954ad4bf1873a0f1d71a8b1c5d06f751c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:36:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a66c5c79a7124545be7678250fac5cc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b39dcf63214bff99b4b416da8fcc8f : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e7166f5bcf324fce94671fab9b3c356b : SUCCESS in 1h 17m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38ccce93329a41109e34767d1f989138 : SUCCESS in 1h 05m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/846ce3b825a84f4baa9f6220e91d1f50 : SUCCESS in 1h 05m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8755a6f192d14371a68dbcaf89758b08 : SUCCESS in 1h 03m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98d90b5353f546749d6a3275874cb3a5 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81c6fcb8eadc41f7b5033e7314c91e7a : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/860bcfb3c85b436aadbb650a258f1fb2 : SUCCESS in 14m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e59f7eaeb1e649c4af29b048195f38a0 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f46376629ba43b9a3192024a220972b : SUCCESS in 24m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4fff0916ccb043c2865e7774cdc76081 : SUCCESS in 21m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46f2df3b2caf492f83d3932fb8253a9f : SUCCESS in 18m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cc197fcf59db433da684e5166ad592fe : SUCCESS in 31m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b745934eb65b421a94ffa5ab97cb0c09 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92e1a863d7f1498a97c869347c384786 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/66e517e28c0f457b9a39e0825963329e : SUCCESS in 22m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3553b5c8fd174b068ce045b31a57c875 : SUCCESS in 9m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5f922644b0f740d2b4505dcd22c24858 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7f1f28bfdbe34f0b8bca542c4fe03f6d : SUCCESS in 10m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b892e51f4f544505aa397a2caef6f137 : SUCCESS in 9m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3479f5e90acc40788734067140515c72 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3fa68dc21ee548929186fa0170e59f17 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/96b2e7fc46de49908246c303186f0990 : SUCCESS in 53m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c39e5c2a7e447fa8585bcedbffaa619 : SUCCESS in 54m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d19adf4ae3984e49acdd71de7e2893a9 : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5cccc4e71834702973f77c7214bf666 : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97155710f71f4cdc97ba20852eb2428d : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0a885054a0c4c06bb9f91cbf9577a98 : SUCCESS in 18m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9c63fa53bdb40f7bbd6496f6a1efc1e : SUCCESS in 52m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcb3a33369f04dbc8e2e69504eddd1ec : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ae88a24261f5bb8dc989152c62b9621f3264874c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 18:45:15.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97710ce90e224a15b80cff93c21fc6c5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f5e57de7c57d498f9fc8da7c332ae16b : TIMED_OUT in 3h 01m 02s","accounts_in_message":[],"_revision_number":2},{"id":"5563d0eef67035f30bd1f64e53a2986265b30812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-19 10:15:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ec2a67d412ec265bf6ab6b677d44548c849a60c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:36:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09226a99891141e0b46d0caf262c54c7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb79f580a1324de6bf20e7d0c898a7ce : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":3},{"id":"9ea5397ea96de82e4666d1ca2edb989265d82d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 11:38:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2cad61e6d31496db41b7f33709d58f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39bc8d9c4ef4a46bfdd45cc9e1d779a : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/86cc1a7e5f8847b6ae1e478c7c72f7cd : SUCCESS in 1h 13m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f2b035b8ed9491cbbb1f8023071161a : SUCCESS in 54m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74f018a9a2c944d69e4c16b84be96ab7 : SUCCESS in 43m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bc8cad93fb54dae9d42a4f69356bb2b : SUCCESS in 1h 05m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e773a22f04a4730b8fc5d6955b557f3 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb327831a6134e7daf46c323b38d2409 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6e7f539440949a7a4c2730605ad3cbe : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6611eda37be4417b0dafec2f7a59e4e : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fded5e2aca5a4f6fb0b41eca9cae2fa7 : SUCCESS in 20m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/683e9859a7004125bcf46ad74d975434 : SUCCESS in 15m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7ace4c4d1204dfdab84848e8280c47d : SUCCESS in 18m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fe0ebaf8bbf54d158c862da325fdb06c : SUCCESS in 32m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fa09cada0604349a110f755599d19b6 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e41c8244fb264a1e873df3554ac9ee6c : SUCCESS in 16m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4449461223842988ae2175b370d68c7 : SUCCESS in 15m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/40b916453fe049fba087afb12581f9bb : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e1e60169e4c419392a5b5de4ced16e7 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9b4aa2f26aba4f6691a0c9c0efe439ed : SUCCESS in 14m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/acbacc74729c402bb6a92d6b17f088bd : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22d7b1c799e14c1eadb03199b065a538 : SUCCESS in 14m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5041631216d0442d9b5b50a318bedbdf : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0284ac62a8704b61b2283f7a371e6b49 : SUCCESS in 1h 02m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43b61858b044415a9b7df08201020767 : SUCCESS in 33m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da1ac4c19bf7400f9545d2eee40bc0fb : SUCCESS in 50m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2699188dfdd047baae01731fcf6308d9 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c0e9bd2b534f55beb68a360b818473 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65d2ee470edd4d8f9595eab1a92409c1 : SUCCESS in 26m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9e2d6d717534a0299ce1b1fe3b969c5 : SUCCESS in 38m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74a8d78736074926abb762226737fe09 : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":3},{"id":"7c2bffdeb0dbeee5a9d39a3a5d01ac4a05b9c309","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-20 16:21:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eacb4bb5f7c83730457cd7db143813f437641ea6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-06 11:35:56.000000000","message":"Abandoned\n\nprogressing on feature/mpu instead","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d52f841b6d0f5993693df0f3d8319311d5374e5d","revisions":{"46c2b3e931647913bae21a47bbffd204ea850423":{"kind":"REWORK","_number":1,"created":"2024-12-18 14:04:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/937990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/937990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/937990/1"}}},"commit":{"parents":[{"commit":"e2e3360f7dc748cbbcf084a8dd84f555652e4f91","subject":"object_versioning: only create one version of each .data file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2e3360f7dc748cbbcf084a8dd84f555652e4f91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 13:52:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 14:03:51.000000000","tz":0},"subject":"object-versioning: refactor helpers to be module functions","message":"object-versioning: refactor helpers to be module functions\n\nMove the build_versions_object_name and build_versions_container_name\nmethods to be module functions so that they can be imported and used\nelsewhere.\n\nAdd unit tests for these helpers.\n\nRefactor only; no behavioral changes.\n\nChange-Id: If6248bd85892738c891c25ed5b2e5fb4a46a2193\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46c2b3e931647913bae21a47bbffd204ea850423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46c2b3e931647913bae21a47bbffd204ea850423"}]},"branch":"refs/heads/master"},"a0a26bbbb51a94ceecb7554c2bbd088a14ff553c":{"kind":"REWORK","_number":2,"created":"2024-12-18 14:10:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/937990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/937990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/937990/2"}}},"commit":{"parents":[{"commit":"e2e3360f7dc748cbbcf084a8dd84f555652e4f91","subject":"object_versioning: only create one version of each .data file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2e3360f7dc748cbbcf084a8dd84f555652e4f91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 13:52:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 14:09:57.000000000","tz":0},"subject":"object-versioning: refactor helpers to be module functions","message":"object-versioning: refactor helpers to be module functions\n\nMove the build_versions_object_name and build_versions_container_name\nmethods to be module functions so that they can be imported and used\nelsewhere.\n\nAdd unit tests for these helpers.\n\nRefactor only; no behavioral changes.\n\nChange-Id: If6248bd85892738c891c25ed5b2e5fb4a46a2193\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0a26bbbb51a94ceecb7554c2bbd088a14ff553c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0a26bbbb51a94ceecb7554c2bbd088a14ff553c"}]},"branch":"refs/heads/master"},"d52f841b6d0f5993693df0f3d8319311d5374e5d":{"kind":"REWORK","_number":3,"created":"2024-12-19 10:15:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/937990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/937990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/937990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/937990/3"}}},"commit":{"parents":[{"commit":"85d6f8a984dd0d7f075318a9d42f6d78e6bcc337","subject":"object_versioning: only create one version of each .data file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d6f8a984dd0d7f075318a9d42f6d78e6bcc337"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-18 13:52:30.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-19 10:13:30.000000000","tz":0},"subject":"object-versioning: refactor helpers to be module functions","message":"object-versioning: refactor helpers to be module functions\n\nMove the build_versions_object_name and build_versions_container_name\nmethods to be module functions so that they can be imported and used\nelsewhere.\n\nAdd unit tests for these helpers.\n\nRefactor only; no behavioural changes.\n\nChange-Id: If6248bd85892738c891c25ed5b2e5fb4a46a2193\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d52f841b6d0f5993693df0f3d8319311d5374e5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d52f841b6d0f5993693df0f3d8319311d5374e5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
