)]}'
{"id":"openstack%2Fswift~938111","triplet_id":"openstack%2Fswift~master~Ie1e815c0434255e77ef5e9103576f85d9d6490ae","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"37293":{"account":{"_account_id":37293,"name":"Junya Noguchi","email":"noguchi.junya@fujitsu.com","username":"ngcjny"},"last_update":"2024-12-20 16:28:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1e815c0434255e77ef5e9103576f85d9d6490ae","subject":"docs: Changed OS version to RHEL 9 and CentOS Stream 9.","status":"MERGED","created":"2024-12-20 08:29:39.000000000","updated":"2024-12-20 16:29:41.000000000","submitted":"2024-12-20 16:28:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938111","meta_rev_id":"41a3a9349dab2d0b5a23787427982b1e8a9059c9","_number":938111,"virtual_id_number":938111,"owner":{"_account_id":37293,"name":"Junya Noguchi","email":"noguchi.junya@fujitsu.com","username":"ngcjny"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-20 16:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-20 16:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-20 16:11:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-20 08:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-20 16:11:45.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"038dbb7bda424af23ee6b0c5aca1ad275033a21e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37293,"name":"Junya Noguchi","email":"noguchi.junya@fujitsu.com","username":"ngcjny"},"date":"2024-12-20 08:29:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe4f8d23b29b91e1fb491dc16100328d3fe4ace5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 08:48:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92da42fc0f734ca6bddc7a462e2b7043\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1a0bbd1a4b4262ae3f8c2985c39d4d : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"59afb1bb55770b379d2930c1ceed93019ecfc0a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-20 16:11:33.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b49e97620ff5c97850b8ca51c9990e8cfc929197","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-20 16:11:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4f9524960f4bfb939b9ee63b4d08a34f98788ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 16:20:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ebfe90a3c04bf394ba5fd85407fdf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f459fe2358485998e3fde3b6124498 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"13417cb5b09d2a9cb7a8e2272e4f1ded4bad9d98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 16:20:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d695fb3a6470fd2e81a94fa7cf3a1264b3bff5e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 16:28:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa408acdf7b4d8b96be9fccaae43e90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceadb3eed02c463da115c953412b59cb : SUCCESS in 7m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2836142c1884718bd3cb61df6e8dbf2 : SUCCESS in 5m 05s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0c160f8a6af24753869020624da54b66 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c67405083cc7b755ed640130788ffd62dcdfa409","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 16:28:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"41a3a9349dab2d0b5a23787427982b1e8a9059c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 16:29:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab8bbfb15ff454786435ebb20b5c60a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32432aab69b241778bd24cfb6c65e663 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dc8ae266590e47889834c40856d6a9d5 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"17f77b2d764985e0acd5d7d63c2042fbc40f636b","revisions":{"a2dd51932a21920168e9e48b2d0a7bd194a8d820":{"kind":"REWORK","_number":1,"created":"2024-12-20 08:29:39.000000000","uploader":{"_account_id":37293,"name":"Junya Noguchi","email":"noguchi.junya@fujitsu.com","username":"ngcjny"},"ref":"refs/changes/11/938111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/938111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/938111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/938111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/938111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/938111/1"}}},"commit":{"parents":[{"commit":"f9a3f142ab16594c6c43b14b258ee417a80eecbe","subject":"Merge \"Make OldestAsyncPendingTracker timestamp float\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9a3f142ab16594c6c43b14b258ee417a80eecbe"}]}],"author":{"name":"ngcjny","email":"noguchi.junya@fujitsu.com","date":"2024-12-20 08:27:19.000000000","tz":540},"committer":{"name":"ngcjny","email":"noguchi.junya@fujitsu.com","date":"2024-12-20 08:27:19.000000000","tz":540},"subject":"Changed OS version to RHEL 9 and CentOS Stream 9.","message":"Changed OS version to RHEL 9 and CentOS Stream 9.\n\nChanged OS version from RHEL 7 and CentOS 7 to RHEL 9 and\nCentOS Stream 9.\nChanged python to python3.\nChanged yum command to dnf command.\n\nChange-Id: Ie1e815c0434255e77ef5e9103576f85d9d6490ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2dd51932a21920168e9e48b2d0a7bd194a8d820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2dd51932a21920168e9e48b2d0a7bd194a8d820"}]},"branch":"refs/heads/master"},"17f77b2d764985e0acd5d7d63c2042fbc40f636b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-20 16:11:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/938111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/938111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/938111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/938111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/938111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/938111/2"}}},"commit":{"parents":[{"commit":"f9a3f142ab16594c6c43b14b258ee417a80eecbe","subject":"Merge \"Make OldestAsyncPendingTracker timestamp float\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9a3f142ab16594c6c43b14b258ee417a80eecbe"}]}],"author":{"name":"ngcjny","email":"noguchi.junya@fujitsu.com","date":"2024-12-20 08:27:19.000000000","tz":540},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-12-20 16:11:33.000000000","tz":0},"subject":"docs: Changed OS version to RHEL 9 and CentOS Stream 9.","message":"docs: Changed OS version to RHEL 9 and CentOS Stream 9.\n\nChanged OS version from RHEL 7 and CentOS 7 to RHEL 9 and\nCentOS Stream 9.\nChanged python to python3.\nChanged yum command to dnf command.\n\nChange-Id: Ie1e815c0434255e77ef5e9103576f85d9d6490ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17f77b2d764985e0acd5d7d63c2042fbc40f636b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17f77b2d764985e0acd5d7d63c2042fbc40f636b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
