)]}'
{"id":"openstack%2Fswift~938127","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ib18cf251e0f2a313abc894363ba478f00e1c17c0","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-no-symlink","hashtags":[],"change_id":"Ib18cf251e0f2a313abc894363ba478f00e1c17c0","subject":"mpu: no more symlinks any more","status":"MERGED","created":"2024-12-20 11:34:28.000000000","updated":"2024-12-23 11:10:41.000000000","submitted":"2024-12-23 11:09:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938127-p-mpu-no-symlink","meta_rev_id":"a026fd82e6af2fef0ddde309c4b022da47a423ec","_number":938127,"virtual_id_number":938127,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-23 11:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-12-23 10:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-12-23 10:13:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-20 12:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55bdffd4125715353f11db5f248cb156fd405665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-20 11:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb0798bd839d5b307bd39d79493df73d55ea8961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 12:28:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b08ad9e57c4c7db37ab4156b908751\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afac8457a61c46f08cfab3b9742ecb50 : SUCCESS in 6m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/069c457e463346d785c45f1502af7602 : SUCCESS in 25m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78e2450bf3cb4481b0e3a2aad4887af1 : SUCCESS in 13m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87c3c1c8f7a449a4a1f0456ae1df43e2 : SUCCESS in 28m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47b167a483ed468aaf550ab78f4a19a9 : FAILURE in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ceda26d2e47b439c92fb0330cfdffa1f : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9aa117725fef4e6f9ee755b573036852 : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c26bc33a190c4ba9af2503e01ded98f0 : SUCCESS in 18m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/653c7a3fe25f4a4dafcbe8e1af6e097f : SUCCESS in 28m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/607051e56062475d867deb7d47d9e003 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8786200deec44b3af8f6a34c119f494 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/295e3f439c7d4688ac9138f173e9f0de : SUCCESS in 18m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f62f5d2c2734481faae1e67678b1228f : SUCCESS in 12m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f1291281e894768a62321a1778b5457 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7c84e89079a1468b954ff093676f9e7f : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3e7d46f10b24190bf70f08265719da0 : SUCCESS in 8m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/37e412f7e06a4c64b6a1392f912588ed : SUCCESS in 13m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/41ab4232c1974a9bb3e0a4d617a7be58 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b162422062d4ecba3c2a987677e3695 : SUCCESS in 53m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61d572869dee4e41ad87488cbc43d657 : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cee0b8509e8460388c85a138e29d934 : SUCCESS in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9d037843af04581918b4a9c73b12e3e : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"50a690ad3254564bb1c4ac0adbee114ca6c863f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 13:35:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a9dab055c3541f090fabd10b5e04363\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef711aa457b64e8390d706f2c3f3ec80 : SUCCESS in 1h 59m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d62805ec783173bf9d2ce7c02b1378f59780647b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-23 10:13:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f96f71d2f21873efc33d58e62e239aa289b5c3aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 10:13:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2239f94bc532c41b4178b2d76f383182d9f4a5d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 11:09:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af640256ad224b81ab3b436c817379f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4680b0a34ec347b3afee1ab3eb96817f : SUCCESS in 8m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/76c1cdf594cb40b4b45559219fffbdcb : POST_FAILURE in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c06da33fd1240778814c26001598f65 : SUCCESS in 11m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73d450a49ece425485b1614f5480f9ee : SUCCESS in 19m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c3fd9db654a74e5d9106bda2c72672a5 : SUCCESS in 21m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/588cd29d3322453fa826debe66ed8bd4 : SUCCESS in 23m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19c59a46f64f4de3b8bdac199885aa2a : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ac12df09a8542898a7f941ae66baafd : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a490c669b1b4c709c4b37aa173736cd : SUCCESS in 16m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d27b77907b64d49abca0bab4f44f54b : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/04dc1fa9c7754bd7a19d083c567011bb : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/768c7b7203a445489f2a7721b6abf7fd : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7343ba58c724ea0a42d4afc64f7a7bf : SUCCESS in 8m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/254bd6b866e844b8b6b23f95889aea81 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/134aaf84bbc3456bbb0d80e97f2335bd : SUCCESS in 47m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cb5dacc2daf4a8fa82eb890b9ab66ab : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9cf3f1c3664c76b17e1cce738c9078 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a0e33c7b13547639648a72760819d62 : SUCCESS in 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ec6a660c0d75d5219d624d544ca065760c3044b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 11:09:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a026fd82e6af2fef0ddde309c4b022da47a423ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 11:10:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0eab6f61b094ff48ad6088a3e75c49c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4cc006608b4775af53c1f89c4e363f : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b94a8dec559a49508a151422c3cc1fa4 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd7b4349b3caa1596135672ffa3dc97d3cc0dd1e","revisions":{"dd7b4349b3caa1596135672ffa3dc97d3cc0dd1e":{"kind":"REWORK","_number":1,"created":"2024-12-20 11:34:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/938127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/938127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/938127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/938127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/938127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/938127/1"}}},"commit":{"parents":[{"commit":"e1e0d5cfdaa6f9a85cba711f200896e570cc1795","subject":"mpu: improve error handling and add tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1e0d5cfdaa6f9a85cba711f200896e570cc1795"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-19 09:10:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 11:33:59.000000000","tz":0},"subject":"mpu: no more symlinks any more","message":"mpu: no more symlinks any more\n\nChange-Id: Ib18cf251e0f2a313abc894363ba478f00e1c17c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd7b4349b3caa1596135672ffa3dc97d3cc0dd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd7b4349b3caa1596135672ffa3dc97d3cc0dd1e"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
