)]}'
{"id":"openstack%2Fswift~938141","triplet_id":"openstack%2Fswift~feature%2Fmpu~I587672442d08f4a6027b289fc07bd55b8c010be6","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-obj-server-detects-orphans","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-02 09:47:21.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I587672442d08f4a6027b289fc07bd55b8c010be6","subject":"WIP mpu: make object server discover potential orphans","status":"NEW","created":"2024-12-20 18:12:52.000000000","updated":"2025-09-02 09:47:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"990641c0662a1ba02bc54c01084d1ae46f0b0098","_number":938141,"virtual_id_number":938141,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-08 11:56:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2025-09-02 09:47:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-20 19:10:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d936807cce8cdaeaaad4c2d080202f149eae3e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-20 18:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0f64fd864c92ae3de56485391d89325b33ae6a4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-20 18:13:11.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d637e417b62a72b58b2150b6bbcfe4ee108643b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 19:10:45.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59c5991477ce4d419f925ace192f0145\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ccbcf32d9804162b79cd4ba797d9a21 : SUCCESS in 6m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2420f72015ff44908ee29bfc2fc44644 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef4a1d186ef8431b98c18514d1e33d88 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/868204b27d104e87811c790d4f6bdb2b : SUCCESS in 22m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0265ded41d1a4ae79725875529a49c58 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7c754a5d99d140c78121a06cfe1770de : SUCCESS in 15m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7fbbd485176649cca24d28062e942b18 : SUCCESS in 18m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb9a56a246d44a8cb0815245a367d1d8 : SUCCESS in 16m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c5750b5c42634576857167b948ecf0c7 : SUCCESS in 28m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/91c797c717d54291905fc97ba699be4c : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bada97bebecc40faa37160d2cba4d9c2 : SUCCESS in 10m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d4c505f3c10c4672ae68073175249f01 : SUCCESS in 18m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5470189bdecf47dc888bcc4a27a3fd28 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6665a21e55d34b5e89e362d2256684c7 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/974eaedd9da945dfa5a29a31201a7dc2 : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/16d712286aed423e9f4d2080922ea86e : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11c3971a48674dc3b3d9ab204d0a62a8 : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f1787dd2635402286633120392ab4e2 : SUCCESS in 4m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bb2a688ba6c45b1ab2fe2054f90ffcb : SUCCESS in 52m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a15c7404343445a1a8b4b4e3f5d10a0d : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0c8f89fdc24e998a994f92dd9282d3 : FAILURE in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8461fc05f0f34a6da85ff9b39a97aac8 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f5a81d29d57d08c17bdec3d823a77c54eb2eea8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 19:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddfa0fa6e93e4ff28cd4363e69ec9aa6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99ef446012474ae0aae7ab2da0bb4e57 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f0083e262f27ada0b991199b69b95767b41f45ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-12-23 18:09:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6af3b5703cc8c9f600c7c39bee3f92846318e9d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 19:12:12.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd103224c4f54b9a9fe273784fd77313\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72743610e53346b9876140b0f1723c4b : SUCCESS in 5m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/992769e039f14ccbae537625b2917a35 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b16bcb2df22414f9b7f37437cd320c4 : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad1a01f8b3a9485ba2e2ec2df658a0ba : SUCCESS in 22m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/674821929f694e599985b6e70577bde2 : SUCCESS in 4m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6eda87e25b9a4a8fbc206f36332f7b66 : SUCCESS in 18m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2572135ca0cb48f48b003d67b9970e35 : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4e46d675d3f4d4986a667211265ccd0 : SUCCESS in 13m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4da2cf2ab175490b88d3f1bac821b922 : SUCCESS in 28m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8a4eb9ae1ce4dc893a5041783067a00 : SUCCESS in 21m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/701edbcfdf8e4e70af5fcfba3a94c901 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60a95173003d4613b94375e22791fc75 : SUCCESS in 17m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a449e3f798ad4df98164ec5eefaf3ea7 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/563fd62b6b744553a61beb525b9b896a : SUCCESS in 15m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/328f0a996b284aaf8ab483180ab6821f : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b779b9132b64cc2a2b97cb355ae4019 : SUCCESS in 5m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a2088d245be9427f953dcd0b7152f318 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/632e455bd03241cfb0f52684c8133aee : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6689eaf5cdb04a7186df25753b0b0cbf : FAILURE in 56m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5388e4199d847568bfeaba33ee933f4 : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6288b38819194a779311299917e6a194 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3395e586903142fdbbed03cfecd030ad : SUCCESS in 18m 23s","accounts_in_message":[],"_revision_number":2},{"id":"420b826637dafdc6e425e40d4ba6e0218cba96c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-23 19:18:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e78d97985543bea04f338482d734a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5031a157beff49c09f10e7913e559b8f : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e37948f8b15bad32278d8a38d16e78c37174dfb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-03 13:00:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7645cc08c38df96e62b7a331be26a433c46f0265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-03 13:39:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"73fd4b3bd227415e840b8ad19a8f23aae8d22099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 14:51:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8da788c779346f8812afe9ef15337c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f14917c3bf64075b0d4d446a7e1dcac : SUCCESS in 9m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/131ce2a93e1741b59d6e1fed3bebe00f : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/235ecf397baa4e07b9766d4e031c56b5 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f794579836cd4d0586fc323356785830 : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f143d42fd94a42e1bbba71c2a9f2fc4b : SUCCESS in 14m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e6488e7ae8a4473b3c48eee89447faa : SUCCESS in 18m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42eb23283dc441ff9c92cba17028f0e8 : SUCCESS in 18m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/71daf4b313c94e978996dad3c1d8741e : SUCCESS in 21m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1cc9058952a84262a79ccd20570387ee : SUCCESS in 31m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10657ff2b05147cd8fbe5d54630d2d27 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d3eeaa085565473db0ba9d81bdd55f74 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bdcd886e8fb84fecb013df5270a68248 : SUCCESS in 17m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/59cb9c1643e74e6f9683492e30bbde21 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e695ff8db7f49739f4f24d9b5a115c1 : SUCCESS in 10m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/862d5740354a4c39ab89a6b75fd7c4be : SUCCESS in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d191a0a474140d087a55b8cbde6d0af : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb56fbde0ea0424a9c6fe302e4802682 : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a52d62a6ae074214a0c4618539ef6e30 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cc9fd00535f475b8ba901b7a80974c4 : SUCCESS in 54m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f2df81d795b4506ae3d3fad4fb90c19 : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43747559044499db0aac9fa577027e6 : SUCCESS in 15m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b456ba3ab3d0482990787b80256fdddb : SUCCESS in 23m 01s","accounts_in_message":[],"_revision_number":4},{"id":"8c86fe0191437de67f1bebfc11deb73e85e48370","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 16:09:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40792a9ef9dc4ffda5ee33e472bbc7fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b71e1db5b6c04a1fb918cb93cca1dce6 : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6cb4126483ea2c1b2aae8b1cbe5a712c355c577e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-07 14:57:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"48431dd266e6e1b6b3aa2ad89a78a53e647850d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 16:01:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc8d0066aee43c083866672fd0b26b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f664802e0e0b44fb8f887d8594fa4647 : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/344d138c719f46e492a943aa5dbb0f97 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0afd3d19e074326a49faa7c13c9a7f3 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5e2de3e5b6846b2925959448ec157c2 : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78a351af0d504ad1a7813ebc6483f826 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe3292c327e24cbf93bd1dbe2241c9ae : SUCCESS in 17m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d734e47b982b413f89fd1db2ee7f2924 : SUCCESS in 19m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d4b7682c010f4db1aecf60ad98968807 : SUCCESS in 32m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/998d635865d8473eba892300798efc50 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4476f3f9e9e345b1a587f50f960acd3e : SUCCESS in 15m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/acf3a9a62fe54f158427d0d6b9297585 : SUCCESS in 19m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5eca6b7efada4d8498d8cd3607230829 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4ca2f291f474e299790fbcc55c936fe : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a66911dd5d844fa2adf18b5382530a88 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e842c037d61e491d81a2fd26f180414b : SUCCESS in 50m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bf90676fb4b447182f6e3fd85bd6f8a : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f527879db345feb033fd6a49bee95c : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2156291681104c0981d3e200bf4f3d4a : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":5},{"id":"bc2a24017f057a98f22294d035b756d9495e2058","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 16:52:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab7c694c85d44fb999f47f1ff30ae8e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14800f3f714e40ee844d1035a7cbbfab : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":5},{"id":"199cd01d98f20166cc6c477b4aef39537a814230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-08 11:02:12.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a65a8829f9a46a466eb1b0d3d95603e00cb2f89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 11:56:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afba36b66684896a14448cc4f4c55ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3942a3cfe3484887c219b43e8060a9 : SUCCESS in 8m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b983db4bc4ff421faa67ca9a4012b7a2 : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b035fc3450974d179ecbb62f2584082d : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed9efaf6e1f94c29aa62d7893d389dcb : SUCCESS in 27m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d4af2697aa3489c9d8974595eca3f7b : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa8cb274998e4de69c22ba40b09e31c4 : SUCCESS in 24m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f972d09341a48059a4b01e127689123 : SUCCESS in 14m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/de396e361f534b43825a027539a15a5e : SUCCESS in 21m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2e2a8e75be0d4315a2d76df89846276f : SUCCESS in 22m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e0f48c0c628a46268f66fc9908a654ab : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14eef11eb45f472bb86665a15396769a : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1eaf972d1274aaf839ea3368d3e7975 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5b775ae6df94867939a7068b9d119aa : SUCCESS in 9m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9f7186937a442feb51e0df116dfde73 : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcb9c2b8a0594d1eb8753f000d66d65d : SUCCESS in 49m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b819372c3a384235abea7bb841734918 : SUCCESS in 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e5f9a5be37419da3bdd6e54ce06dc9 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee3685767b7a4a27bb060d3fd26b4037 : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":6},{"id":"6bcfe0ff75a77d7d65b0bd1eca946faaa41800ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 12:44:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43a4832714a2411298e6472395050439\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae4ecdec46f74706b7a5f3820f8d56e2 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":6},{"id":"990641c0662a1ba02bc54c01084d1ae46f0b0098","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-02 09:47:21.000000000","message":"Patch Set 6: Code-Review-2","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"80cba2bd44c95aa1c4c5a5b7e5437f52280cdfac","revisions":{"392443a5e6dafd3a1db222eb55fe3785e7eb0c94":{"kind":"REWORK","_number":1,"created":"2024-12-20 18:12:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/938141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/938141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/938141/1"}}},"commit":{"parents":[{"commit":"dd7b4349b3caa1596135672ffa3dc97d3cc0dd1e","subject":"mpu: no more symlinks any more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd7b4349b3caa1596135672ffa3dc97d3cc0dd1e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"subject":"WIP mpu: object-server generates async delete markers","message":"WIP mpu: object-server generates async delete markers\n\nChange-Id: I587672442d08f4a6027b289fc07bd55b8c010be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/392443a5e6dafd3a1db222eb55fe3785e7eb0c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/392443a5e6dafd3a1db222eb55fe3785e7eb0c94"}]},"branch":"refs/heads/feature/mpu"},"9fb12343598b1df15277509b578a78303ba41c44":{"kind":"REWORK","_number":2,"created":"2024-12-23 18:09:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/938141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/938141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/938141/2"}}},"commit":{"parents":[{"commit":"f33ad5b664fe215b3a64e7b4fc40233952b9299b","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f33ad5b664fe215b3a64e7b4fc40233952b9299b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-23 18:09:09.000000000","tz":0},"subject":"WIP mpu: generate async delete markers when data files unlinked","message":"WIP mpu: generate async delete markers when data files unlinked\n\nChange-Id: I587672442d08f4a6027b289fc07bd55b8c010be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb12343598b1df15277509b578a78303ba41c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb12343598b1df15277509b578a78303ba41c44"}]},"branch":"refs/heads/feature/mpu"},"1a8dde6e13eb5fa821f8bf345eea9195cc8a7533":{"kind":"REWORK","_number":3,"created":"2025-01-03 13:00:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/938141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/938141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/938141/3"}}},"commit":{"parents":[{"commit":"e0e6fcf98a0a517453b560446b1c2c08d9a63595","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0e6fcf98a0a517453b560446b1c2c08d9a63595"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-03 12:36:55.000000000","tz":0},"subject":"WIP mpu: generate async delete markers when data files unlinked","message":"WIP mpu: generate async delete markers when data files unlinked\n\nChange-Id: I587672442d08f4a6027b289fc07bd55b8c010be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a8dde6e13eb5fa821f8bf345eea9195cc8a7533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a8dde6e13eb5fa821f8bf345eea9195cc8a7533"}]},"branch":"refs/heads/feature/mpu"},"1a707ee0fbb4cceac55ce18db523023ca4ce5892":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-03 13:39:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/938141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/938141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/938141/4"}}},"commit":{"parents":[{"commit":"a6ffbfb39d1ca006a3cddc87a731e37215479d7d","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6ffbfb39d1ca006a3cddc87a731e37215479d7d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-03 13:39:34.000000000","tz":0},"subject":"WIP mpu: generate async delete markers when data files unlinked","message":"WIP mpu: generate async delete markers when data files unlinked\n\nChange-Id: I587672442d08f4a6027b289fc07bd55b8c010be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a707ee0fbb4cceac55ce18db523023ca4ce5892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a707ee0fbb4cceac55ce18db523023ca4ce5892"}]},"branch":"refs/heads/feature/mpu"},"65ade364b23373d0222779b6db93e5945fa3faee":{"kind":"REWORK","_number":5,"created":"2025-01-07 14:57:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/938141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/938141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/938141/5"}}},"commit":{"parents":[{"commit":"f2f0bd7feb041772fc862ebdfa114bc657a98f45","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2f0bd7feb041772fc862ebdfa114bc657a98f45"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-07 14:53:36.000000000","tz":0},"subject":"WIP mpu: make object server discover potential orphans","message":"WIP mpu: make object server discover potential orphans\n\nGenerate async delete markers when data files are overwritten on the\nobject server.\n\nChange-Id: I587672442d08f4a6027b289fc07bd55b8c010be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65ade364b23373d0222779b6db93e5945fa3faee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65ade364b23373d0222779b6db93e5945fa3faee"}]},"branch":"refs/heads/feature/mpu"},"80cba2bd44c95aa1c4c5a5b7e5437f52280cdfac":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-08 11:02:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/938141/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/938141/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/938141/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/938141/6"}}},"commit":{"parents":[{"commit":"99e7e15e10178c0d706175731b5dc612aeceb010","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e7e15e10178c0d706175731b5dc612aeceb010"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-12-20 18:11:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-08 10:59:58.000000000","tz":0},"subject":"WIP mpu: make object server discover potential orphans","message":"WIP mpu: make object server discover potential orphans\n\nGenerate async delete markers when data files are overwritten on the\nobject server.\n\nChange-Id: I587672442d08f4a6027b289fc07bd55b8c010be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80cba2bd44c95aa1c4c5a5b7e5437f52280cdfac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80cba2bd44c95aa1c4c5a5b7e5437f52280cdfac"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
