)]}'
{"id":"openstack%2Fswift~938162","triplet_id":"openstack%2Fswift~unmaintained%2Fwallaby~I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1","project":"openstack/swift","branch":"unmaintained/wallaby","topic":"zuul-config-error-swift","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-12-22 12:14:54.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1","subject":"CI: Move off CentOS 8","status":"MERGED","created":"2024-12-22 10:30:04.000000000","updated":"2024-12-22 13:23:32.000000000","submitted":"2024-12-22 13:21:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938162-zuul-config-error-swift","meta_rev_id":"33eb3e1ad01d0208619f79b03d9ecc7e9b567f37","_number":938162,"virtual_id_number":938162,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-22 13:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-12-22 12:14:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-12-22 12:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-22 10:30:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-12-22 10:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d3e8e97c1ee6f930f25907cc967fb72c88711df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 10:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"063cb11df394e7b25dd33cd26063dcb54887218f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 10:30:26.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch master.  The\nproblem was:\n\n  Job swift-multinode-rolling-upgrade-victoria not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - integrated-gate-object-storage\n        - swift-jobs-arm64\n      check:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@master\", line 631","accounts_in_message":[],"_revision_number":1},{"id":"96d3fdd3bc0d83efc059c6a2a9b565c938f69a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 11:11:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6012ecd38da412ef7e464776ac55c7e07bfd6d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 12:10:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/963f53a6caef44cbac060bc88478dd1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc53fc91fb14ade9a64869f950e0d2e : SUCCESS in 7m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07e8bce446104344b86fc558fa27d7b4 : FAILURE in 3m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cb421338cf534a7ea8c6274902f3d790 : FAILURE in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f5641813c664332ab26616aca02bf2b : SUCCESS in 13m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5ee53dc33d4373b1a11580e25c00ce : SUCCESS in 11m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b097b3f0a5624225aec46958c7c4f2db : SUCCESS in 14m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/04f84665ce83411cb99d5e92789d142a : SUCCESS in 15m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5acbb184f064860b4ade0a6f713479a : SUCCESS in 11m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d733f2247b594fa8b983e2f18e500313 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e06a1218db744638b33917db304ba3b : SUCCESS in 10m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b35b426edaf7475fa1b4a6e9c1abee2a : SUCCESS in 20m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a4bc057b1c3747188f86bfd5eda12998 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b679355137a3484fb2f27e369ff0c5e0 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/589e7673346d42e0872e004bbc6c5432 : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab28993599244fa3a71303ca08080c4e : FAILURE in 6m 15s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d20059c0491142c2821d3dca9f984338 : SUCCESS in 22m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bbccf03eed441c1832eca5806e886d7 : SUCCESS in 25m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fca925181d2471b869b6ed64c2dd5d8 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11499a49063d4fe1bed18d6d574b1f7b : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7fe3d8c7c2b648b9b6844b2e688c84bf : RETRY_LIMIT in 5m 09s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/27da60c327304196a368dc0c948c741f : SUCCESS in 27m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9589239a3c324d418a29682eb5c4e1a0 : SUCCESS in 41m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/959bbf55eb80491bb7afe84fe8e782db : FAILURE in 29m 39s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4a0972cea544fc8921f00ccba4b6e58 : SUCCESS in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e2c3bc37c5666827e37546f4f48204888d494ad","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 12:14:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c3dfc33e0fbd8700b56984422470dcad2fb399d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 12:15:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9e60115fe752a3b7c5b9790147d02806cab4089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 13:21:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76859d0035ff40d0843505a5de1bac49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f4329de6724453939531379b603cd3 : SUCCESS in 5m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0ac2734815914343a6007ffbe821446a : POST_FAILURE in 4m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c84af66c0bdd49e485abd37862567798 : POST_FAILURE in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc33667f684f4458ba3c4ee4f8342e31 : SUCCESS in 11m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/751a0bceb08d4b84b7734011bd74f3ab : SUCCESS in 11m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8f61bf5032fd4388b40b99fa580efa57 : SUCCESS in 11m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/22b6e5d248ec4412b2c3e74beae1ea74 : SUCCESS in 11m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3531cccf713466f832e0396dfb8b2e8 : SUCCESS in 10m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb2f39d82df844c5a9587ca8f413c807 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdab09991afb4507ab6ed603dcc82e6d : SUCCESS in 9m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f9f3aa617b347daa12920f8dc935c94 : SUCCESS in 15m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fe761daa369e4e4e9dc35ac14cb3c7ca : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/286e3a25e6b844b8807c685ad3f1f153 : SUCCESS in 10m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/960968d823cf4726b619c8edd3ec7ad4 : SUCCESS in 16m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/44c5c6d455ba4e41946b47752cee1902 : SUCCESS in 32m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1348c9d66e574c96bd30130a9a575a48 : SUCCESS in 47m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/370a01bb5e4a4211bbafb987772e73a6 : SUCCESS in 28m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2990acd07c8740c983b994e1cd1dd4e8 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e47bbe97824addbada4c47aba921c7 : SUCCESS in 4m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59cb7878309d42008fc4a2e306206db1 : SUCCESS in 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64d56803ffd04a1e9a1c12447d114fc1 : SUCCESS in 39m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06f08723acc84f27828ec6f3cdc2aff9 : SUCCESS in 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e79385e40886f3a4b345739807c0c5ea48dc3a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 13:21:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"33eb3e1ad01d0208619f79b03d9ecc7e9b567f37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 13:23:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a94d24e56f48a09fc95ac84acaf574\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a652ce22274a3e9b60b2788401cf32 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/84f5cb7705134e7cb85fa2a2c77318eb : FAILURE in 25s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d5b023d164ff459c9724da8128f758e9 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"783bc2a80f34c4e9612925c6479cab42305058e4","revisions":{"e582d17ee247961b421f5983ccf95f96601764f1":{"kind":"REWORK","_number":1,"created":"2024-12-22 10:30:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/62/938162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/938162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/938162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/938162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/938162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/938162/1"}}},"commit":{"parents":[{"commit":"c4e943b7d4c581aefc3b2f272d45aaa1930845b2","subject":"Clean up centos-7 based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e943b7d4c581aefc3b2f272d45aaa1930845b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 15:58:19.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-12-22 10:28:46.000000000","tz":60},"subject":"CI: Move off CentOS 8","message":"CI: Move off CentOS 8\n\nThis is a combination of 3 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move off CentOS 8\n\nRemove swift-tox-py36-centos-8-stream job entirely.\n\nMove the following jobs to CentOS 9:\n\n  - swift-tox-func-s3api-ceph-s3tests-tempauth\n  - swift-tox-func-s3api-tests-tempauth\n  - swift-multinode-rolling-upgrade, as well as the other rolling\n    upgrade jobs\n\nRemove the swift-multinode-rolling-upgrade-victoria job, as py39\nsupport (required for CentOS 9) was not added until wallaby.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Migrate CentOS Stream 8 FIPS job to CentOS Stream 9\n(cherry picked from commit 6fd031055c951dc84a8f80252985b80edb5a6971)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nConflicts:\n  .zuul.yaml\n\nNOTE(elod.illes): this patch is backported to unmaintained/wallaby as\nCentOS Stream 8 is EOL and centos-8-stream nodeset was removed from\nZuul and it gives a config error: \u0027The nodeset \"centos-8-stream\" was\nnot found.\u0027 Official Wallaby runtimes [1] only contains CentOS Stream 8\nand python 3.6 \u0026 3.8.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html\n\nRelated-Change: I596415d17f77f48a6e8a63a61b734a8ca0865847\nChange-Id: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\n(cherry picked from commit 2032fb214eeb41cbb30a0c71525d0db9e4899760)\n(cherry picked from commit 9fa1b7edec54f855ae559709982a22d4aa57be79)\n(cherry picked from commit 1a18d3c16bbec95da54aa6b2a8c3e1469181d799)\n(cherry picked from commit eadb2acde6849aa1e78cfb0a12a9396ff5ccd40b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e582d17ee247961b421f5983ccf95f96601764f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e582d17ee247961b421f5983ccf95f96601764f1"}]},"branch":"refs/heads/unmaintained/wallaby"},"783bc2a80f34c4e9612925c6479cab42305058e4":{"kind":"REWORK","_number":2,"created":"2024-12-22 11:11:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/62/938162/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/938162/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/938162/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/938162/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/938162/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/938162/2"}}},"commit":{"parents":[{"commit":"c4e943b7d4c581aefc3b2f272d45aaa1930845b2","subject":"Clean up centos-7 based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e943b7d4c581aefc3b2f272d45aaa1930845b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 15:58:19.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-12-22 11:08:39.000000000","tz":60},"subject":"CI: Move off CentOS 8","message":"CI: Move off CentOS 8\n\nThis is a combination of 3 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move off CentOS 8\n\nRemove swift-tox-py36-centos-8-stream job entirely.\n\nMove the following jobs to CentOS 9:\n\n  - swift-tox-func-s3api-ceph-s3tests-tempauth\n  - swift-tox-func-s3api-tests-tempauth\n  - swift-multinode-rolling-upgrade, as well as the other rolling\n    upgrade jobs\n\nRemove the swift-multinode-rolling-upgrade-victoria job, as py39\nsupport (required for CentOS 9) was not added until wallaby.\n\nNOTE(elod.illes): swift-multinode-rolling-upgrade-victoria job cannot\nbe removed here as it is defined only here and still not removed on\nmaster branch.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Migrate CentOS Stream 8 FIPS job to CentOS Stream 9\n(cherry picked from commit 6fd031055c951dc84a8f80252985b80edb5a6971)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nConflicts:\n  .zuul.yaml\n\nNOTE(elod.illes): this patch is backported to unmaintained/wallaby as\nCentOS Stream 8 is EOL and centos-8-stream nodeset was removed from\nZuul and it gives a config error: \u0027The nodeset \"centos-8-stream\" was\nnot found.\u0027 Official Wallaby runtimes [1] only contains CentOS Stream 8\nand python 3.6 \u0026 3.8.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html\n\nRelated-Change: I596415d17f77f48a6e8a63a61b734a8ca0865847\nChange-Id: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\n(cherry picked from commit 2032fb214eeb41cbb30a0c71525d0db9e4899760)\n(cherry picked from commit 9fa1b7edec54f855ae559709982a22d4aa57be79)\n(cherry picked from commit 1a18d3c16bbec95da54aa6b2a8c3e1469181d799)\n(cherry picked from commit eadb2acde6849aa1e78cfb0a12a9396ff5ccd40b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/783bc2a80f34c4e9612925c6479cab42305058e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/783bc2a80f34c4e9612925c6479cab42305058e4"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
