)]}'
{"id":"openstack%2Fswift~938163","triplet_id":"openstack%2Fswift~master~I3d6a679e6553534e937303b5210125a6ef8365af","project":"openstack/swift","branch":"master","topic":"swift-remove-old-zuul-job","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-12-29 18:32:15.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-29 17:24:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I3d6a679e6553534e937303b5210125a6ef8365af","subject":"[CI] Remove old experimental rolling upgrade job","status":"MERGED","created":"2024-12-22 11:06:42.000000000","updated":"2024-12-29 18:34:21.000000000","submitted":"2024-12-29 18:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938163-swift-remove-old-zuul-job","meta_rev_id":"5184dd833e520dede75387857dcf6ac404dd1215","_number":938163,"virtual_id_number":938163,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-29 18:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-29 17:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-29 17:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-22 12:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-22 12:17:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"8f894030a78c0da9968db612638d3516d2ad7b96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 11:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcfe0380e101c9635277db14563eb1b5b9491dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 12:07:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b2532b80904252b33d947c98e8c259\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e3ce8c60514a798079736fc788bb04 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f4e1883f807d4d81a3570e5b6f4d3d50 : SUCCESS in 45m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2cfe5b9a58e44439b3f804fe77412308 : SUCCESS in 44m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b0f1e8b9c694cd39c9235b56555246c : SUCCESS in 41m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c4681da383b41db91a12a0da78f9bab : SUCCESS in 48m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d602362406c24c1190579118754a0534 : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3c5567d459c486fbd096b250a36882e : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57286dae7c58426aa1d350f4e49e58d0 : SUCCESS in 24m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc99fdc312cc48c98bbefa61a2bae8dc : SUCCESS in 4m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c515e517cd84cc38ae36208063055b5 : SUCCESS in 20m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0569780abd4012a197241d3b8ce722 : SUCCESS in 9m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9fc8defdff3d402ba3b03c50940f5f83 : SUCCESS in 20m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/858a452dcf93423c80705bda72836dd9 : SUCCESS in 16m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce9a6446b3754dc2939a0c6983548ed4 : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32beca7cdfb448f49e47063f06536266 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e831846a89e746dda51fefd55180da15 : SUCCESS in 21m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6b124d971b7141259734eee06be815aa : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d1f0d9e9fe542c089efd9aaf25ded1a : SUCCESS in 9m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/63cb5e0f8ef9460299a631df9bc43417 : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb422c79ee9a4906a1a91368ae039617 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/602d2ef9627b445d8aa889dc506ab2f4 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/572af17614134a38a708353f473ed7c3 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb3c3b5584da4730984fc784fdc5d499 : SUCCESS in 54m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fff5ea2cb7624b4794f5117e80ecea43 : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4bb4106313b49b7b865fb121c14cea8 : SUCCESS in 27m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e08fd79c0bdc483b9fad13c123342802 : SUCCESS in 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e78d71938414a31983fdc1ff925f03f : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7df2f5bdbf7346d19dd3b66e3b2b953a : SUCCESS in 14m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff15c2c5651e4cffbd365970ff5f9dfb : SUCCESS in 40m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b110718360d43b9bfc9f20e71da91da : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fbd57d79f8d52377c8ed2cafdfe65402cabff31b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 12:24:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb3dd639f404b16b413cd0457c30980\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3944f977ea414c22b4dfda224719b06b : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e5e51e8731690c4d594c4b210f63ba339024e6c7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 17:16:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"917cc7d70a49519b8d2614e416767732ef79349c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 18:36:14.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85943fa889a84c7d851c9d32f9f47c2a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/30c87741f5394d148cd3407e87b758d7 : SUCCESS in 1h 18m 04s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/d19e7409c2f046558189573071090d4a : FAILURE in 1h 09m 30s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/110400133e5648e8abc5f4a5b298d175 : FAILURE in 1h 10m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6ef01fcc5d694507b1034c43d1609544 : SUCCESS in 10m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1659a4dbb23b4b76ba0e43d83a8ecd88 : SUCCESS in 10m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/129b6b3f2006480aa7919291f905e952 : SUCCESS in 15m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e8f51d8776da4d1ab071e25b3fd977ae : SUCCESS in 23m 09s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/dac13cdf0fdf4b22ba10587cad15578d : SUCCESS in 10m 21s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a17be1764f84d5783cb28e8d5c045d2 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f8d86c7de364a62b40cb7085323ecb7 : SUCCESS in 18m 05s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/f1067e198e074da2a5a5c8a8415f8055 : SUCCESS in 10m 38s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/98221aeee810460b97d9125731ead1fc : SUCCESS in 17m 20s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/8247c4305a174db9a30ccb24fe264dcc : SUCCESS in 18m 41s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/11ce4f92cc0f4a9fbde9481bd1140ba3 : SUCCESS in 23m 57s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/9992f093a7fa4e6c9bfbd6692b37ad8a : SUCCESS in 15m 27s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/1bfff8d312034b1b913c6473bfdcd776 : SUCCESS in 14m 04s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/97d65c98aeb74122b4b9e3fd365f4e8a : SUCCESS in 17m 10s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/b35a18ecb0d9431986f3ce619244ddd8 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7a4b4d9bb02840acb7ea56936ea1da54998b7188","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-29 17:24:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4fcfdac176de1574f4e89adf62c2910f34d4915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 17:25:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc7fc28a31aa01b38e362000a0bd0c7a87416b23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 18:32:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9008709b7bf14f9cb573684f55dc41bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39d5acab8a648ce836d6c1da853bafc : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/815777208e534610bc187ffec3efb38e : SUCCESS in 45m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d0c9ddc94484d0595a4034ba0f216a7 : SUCCESS in 49m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2602f335af0f4e8eabfa5e34b3dbb826 : SUCCESS in 40m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06f558cbc7b6419ebe0a752e3bd60e07 : SUCCESS in 51m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a90775bf6dac4233897daa447ce8ece2 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51d37449310c4c2ab246c2b5ea743692 : SUCCESS in 19m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c9dafc44444f49812d09a17604331f : SUCCESS in 18m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ecb8688d7994426f892295c326205bc7 : SUCCESS in 19m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ec18265a05e44788c85a0973c08ecb4 : SUCCESS in 19m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d452048deba74bd48fc9a6b5de724398 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/157c5b20de7d46f2b8b606ae1edb6d80 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0e655930e46144ffbe9aa76c9457bd9f : SUCCESS in 19m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f6132e8662c54dcab8e98a1931a9423d : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22ea3cb3aec4486ea1611de652b2a06f : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/292c329f004a4a95879eb4b249271d85 : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f22604c58fb74eedacff8b822911e56f : SUCCESS in 5m 34s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c834f276244f499d98c6d19820214937 : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a85cbd514925447682111b2fccc6be26 : SUCCESS in 49m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57a0e753585b4052b6c3625469eb6256 : SUCCESS in 38m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4585076ba9dc4b088d4bd470d6e75452 : SUCCESS in 54m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72e03a56ecd945b5a1a3ce76432e8755 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07c4ad8b01c4642bbf695ce4d97d26b : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b52ade3a1484de597fefaa366c07723 : SUCCESS in 17m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6f4639d5f7747dbb985360228c99881 : SUCCESS in 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0821bc730c2e473cb6dc6b1373221617 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ac9b49940889240f21b13d6d486e56d8da739e26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 18:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5184dd833e520dede75387857dcf6ac404dd1215","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 18:34:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d34f670aded4a50a19cf61a0a52b128\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4528ab35e24c43b2b8ba5d27a80e7d8e : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/eedf0074ea1b466784db311673dec381 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c9838101a1ae847d637f7f3f6720825c3d6d1d5","revisions":{"3c9838101a1ae847d637f7f3f6720825c3d6d1d5":{"kind":"REWORK","_number":1,"created":"2024-12-22 11:06:42.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/63/938163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/938163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/938163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/938163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/938163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/938163/1"}}},"commit":{"parents":[{"commit":"5d1dbbccbee04bf7c703538dbab7c903ea146778","subject":"Merge \"docs: Changed OS version to RHEL 9 and CentOS Stream 9.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d1dbbccbee04bf7c703538dbab7c903ea146778"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-12-22 10:59:14.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-12-22 11:06:41.000000000","tz":60},"subject":"[CI] Remove old experimental rolling upgrade job","message":"[CI] Remove old experimental rolling upgrade job\n\nThis patch removes swift-multinode-rolling-upgrade-victoria job for\nmultiple reason:\n- victoria is very old and unmaintained\n- job is defined only on unmaintained branches\n- py38 + CentOS Stream 8 are EOL\u0027d and the job is based on these\n\nChange-Id: I3d6a679e6553534e937303b5210125a6ef8365af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c9838101a1ae847d637f7f3f6720825c3d6d1d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c9838101a1ae847d637f7f3f6720825c3d6d1d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
