)]}'
{"id":"openstack%2Fswift~938309","triplet_id":"openstack%2Fswift~master~Ia028624bded221c3bf03a8d3dac94183d4388431","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ia028624bded221c3bf03a8d3dac94183d4388431","subject":"CI: Add Dalmatian upgrade job","status":"MERGED","created":"2024-12-29 17:27:35.000000000","updated":"2025-01-03 19:11:04.000000000","submitted":"2025-01-03 19:08:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938309","meta_rev_id":"12e6f297d3ac59e3cde1a818cdb55274ae994fba","_number":938309,"virtual_id_number":938309,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-03 19:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-03 18:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-03 18:07:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-29 18:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a4dae514e16adde011fed5964d3d07d2d313e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-29 17:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dc4beac3665f8264f0e1683c6e45255d1143827","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-29 17:27:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bae9b7ae7a0390dc7596a5fefc5ffd428db51d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 18:44:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eefa738c68b4629bfa2c0e3bf436486\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90274e5ba93f408884ff3664c20083d3 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ea18a67b6ef941c08d893ebb058f1fa0 : SUCCESS in 44m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aceab04925b64499beb1f375232419a9 : SUCCESS in 44m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e316b5524ca4be8909ded97ceaf3f4e : SUCCESS in 40m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe8fe2b7d34b439d83c4e5730842a53a : SUCCESS in 47m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c23a4380c61a499f81c16867524c866d : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49f318108e5b4201ab71ff679592e512 : SUCCESS in 14m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fe896f035fc41c793aecb0cef7b35a7 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b07ba235c47409ea0ed87dde84a71d3 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/16f815a187ca4112a2d6affd0daa978d : SUCCESS in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57664a4e871d475ab51343bb16a5ea1d : SUCCESS in 12m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e955262e02ff4545be0e357358e2e3b9 : SUCCESS in 18m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1b37aa79984e4c26ad39f68c2813db57 : SUCCESS in 28m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/923aab2c72ed4faba8a45836fff960dd : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6bbc0a9cd4f944fda652d50b1da36bf6 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a80222be9ea54bba8c77b75fea2a0b06 : SUCCESS in 20m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb5e989108774760b8564d5bada6022b : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e5f5e9a6cbbc4b22a46d2387292d4daf : SUCCESS in 16m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/78d9e24cefa848f78d65023112e79055 : SUCCESS in 13m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ddd8020f21247229368d31496cfc607 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1014bd67e141487283db509c8f03e16f : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1913332ccbfb445794710d02197e67a3 : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a5aa18f7bce47c58d7b67c4062a060d : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d9943d79f5c41438657caabe06f8875 : SUCCESS in 49m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6630b381c5c04f3cbd94b1c87b691803 : SUCCESS in 35m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34a2fc68fa2e45ee844044f0e1fe19bf : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd8bb5cbb8b4c57b3b2ce227935ca65 : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1292fd4f24854ff1b8496eff1388cd83 : SUCCESS in 17m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60926a2e8bd4438d9e619a672d6665e5 : SUCCESS in 33m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/689d20dc33e244379a57f68100a128b6 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c4bc6753fe4541504a542b8bc931bef435ee6ca0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 19:18:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4387b2945c8417792fbd9e8f6a19e3b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d71b64b686e42b59e47b0b69b4db2b4 : SUCCESS in 1h 24m 10s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/06491443dd1045feb487d86c0ca8f292 : SUCCESS in 1h 48m 56s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/4a393ae8f4d74fda95053ad6f727450b : SUCCESS in 1h 30m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf66884c768e43f2912580cca23619f2 : SUCCESS in 19m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1752e32779ce4a6595960f1ae711d4ac : SUCCESS in 20m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7048d551da46490fa5dd7a32ff4fd63b : SUCCESS in 15m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b8b597ffb2bb46989285ecc946ab9b7d : SUCCESS in 19m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/04e17e3a41e4458ab2957be6c00b660e : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/56851a924f844994abe01ce214065b6f : SUCCESS in 18m 04s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2804daac95d84c778aa0cf1c003562c3 : SUCCESS in 16m 53s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/f92bd0d15fad4874b798fba0c92794d3 : SUCCESS in 14m 29s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/f29346e2e62647608bd460ef7a8babff : SUCCESS in 19m 22s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/705bdcafb6064fc3a87b590691a39659 : SUCCESS in 14m 45s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/b1d822ae2f1a4daebb3bd6da337d0eb3 : SUCCESS in 19m 27s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/902ee1b6c06e416f845273c947a779f2 : SUCCESS in 17m 37s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/8df5a2c37ab54cd4877871612108503b : SUCCESS in 17m 41s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/2008b24d5bf44fd185ebc5f60561a690 : SUCCESS in 15m 20s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/e71678a4d0ac4e389ceafc3c5c0c8c87 : SUCCESS in 19m 40s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/58071d713fe34da999e30933e7a65433 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4f150a4fbfbe086b44070586510f65a4c0feac10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-29 19:24:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9199dcec3d340ae9f60eafebf5da728\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3e8f86ea2654fefb43bb5365c761944 : SUCCESS in 1h 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7fd53821c52360219f3c60afdaaab86ea119863c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-03 18:07:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b6cbf1d7c9e57c6428cbc2f6162fbb71a421b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 18:08:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2457c1d522d4062e2c14f1e42c37921e5dad97e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:08:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb942da3db1466b8a85685430a64e04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69e2eba9bb547fcb13025b6a860c907 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/14ba1a7e0de343fa9b5afd74e98c0295 : SUCCESS in 43m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1b0706186514e34a78e36b34415361b : SUCCESS in 42m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd31ee4f900f4504910cd856f8da7cdf : SUCCESS in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5cdea37441045999a363fb600192214 : SUCCESS in 48m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/25368c75631448779e879103ff6cde73 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9428b881fdfb4b97be37734b114c13e5 : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84c9962c833f4715b170342c7ec1fa2f : SUCCESS in 19m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3134d21112f7411e95c3a458c65aa33b : SUCCESS in 12m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f31f70f625924253af7a4193ca5b9a5a : SUCCESS in 10m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c4cc24a4e3b6437996c4bdf1867745fa : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2431dac256684f2eb18b73bd45f3d644 : SUCCESS in 14m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0949ed405b99455c862fda63c17f239f : SUCCESS in 5m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa878249bc7046ae9572afa6f0e236ca : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b99798bc19e3464db262139b6b6c3a32 : SUCCESS in 50m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ab013123e9444c582fb46fcbd7a91df : SUCCESS in 34m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fadbb227cf864d12b1ace1fcc9c70b9a : SUCCESS in 54m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ec7d5c4b42241beb9837adc8f08bf0b : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8aa778aa9f49ee85d980fc64f14b23 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d33f8826d0874f78a205ec3c581b6c9d : SUCCESS in 19m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e84c02dad414ba59d992ba369178306 : SUCCESS in 36m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c17360ee00a04b8f80c53a089fed9e85 : SUCCESS in 44m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c15e5e1103aea37eda3949b895967d7669fe0cb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:08:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12e6f297d3ac59e3cde1a818cdb55274ae994fba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:11:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b40e8c1b88a4b53815034d7c8e2c4e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c08338687e64036bcd911ad6f689565 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d4a51bcb516f44d39f096b86495c0912 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f35e0c10f5044150f9c2e685e7de43b0ad85807","revisions":{"1f35e0c10f5044150f9c2e685e7de43b0ad85807":{"kind":"REWORK","_number":1,"created":"2024-12-29 17:27:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/09/938309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/938309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/938309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/938309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/938309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/938309/1"}}},"commit":{"parents":[{"commit":"5d1dbbccbee04bf7c703538dbab7c903ea146778","subject":"Merge \"docs: Changed OS version to RHEL 9 and CentOS Stream 9.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d1dbbccbee04bf7c703538dbab7c903ea146778"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-29 17:27:11.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-29 17:27:11.000000000","tz":-480},"subject":"CI: Add Dalmatian upgrade job","message":"CI: Add Dalmatian upgrade job\n\nChange-Id: Ia028624bded221c3bf03a8d3dac94183d4388431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f35e0c10f5044150f9c2e685e7de43b0ad85807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f35e0c10f5044150f9c2e685e7de43b0ad85807"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
