)]}'
{"id":"openstack%2Fswift~938330","triplet_id":"openstack%2Fswift~master~I2f38436b6bb46e03a2617d7ab710c0e8e7eced95","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-30 19:41:13.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2f38436b6bb46e03a2617d7ab710c0e8e7eced95","subject":"CI: Add no-XFS-tempdir job","status":"NEW","created":"2024-12-30 18:19:26.000000000","updated":"2025-01-03 17:48:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8db2f52bf94352e6caefda94af22e1ca3c4fb439","_number":938330,"virtual_id_number":938330,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-31 07:08:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-30 19:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 17:48:26.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"4d74f75cc3bb6651615506f0cd676d12370a8acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-30 18:19:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a76835af03de12b6cc06abb4087ad5b51c27d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 19:41:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/392b1bef4e74456fbd98e8f396b81a5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45ae38369354d4eb70e2639a080e9e0 : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/aaa42d21b1804bbe9a5413f7f948de2b : SUCCESS in 1h 05m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3763d4f7b1994f9c9d9ea1ec31003395 : SUCCESS in 35m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/435e1a058eb847f685d090fe7367aade : SUCCESS in 36m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec7baf35ef2c431aa38ad67f02d8942a : SUCCESS in 48m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c657fbaa091449c2b67ad4374cf213a2 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3893d38bda04435a0e3abcd37015f66 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8d63610e3274567ac1033f674575d57 : SUCCESS in 18m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f74e89497cb24a819006851918f4cab6 : FAILURE in 3m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/932f772701ba4128bf36774fec9c8781 : SUCCESS in 13m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29e1f9bc24794afcbfdfa3bff16d6457 : SUCCESS in 14m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/79fe7c249b8547abb43022cfe36717db : SUCCESS in 13m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/133572bf33194ed2b383f589222a530d : SUCCESS in 27m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47844d5c91904f1382779c88d9e68f97 : FAILURE in 19m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/125556c647b4463ea89712462ebe329c : SUCCESS in 11m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9b6b94178a2424b89571753b6ddb9d3 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/043c72cf106b4c8cbb3f2449eb77f644 : SUCCESS in 18m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0cd114b68c864203945eb0dd87701303 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/befaf6bcfa654a3dac588e417cabd440 : SUCCESS in 14m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0800a16a949d4e3ab49d714b08d23541 : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55c601b49b4b447fa74d77ed1063f0ba : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d05f9a98f684e0782481210b82fe30d : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f60ecd93ef144f4f92c7e84399717d7a : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/027f7f21e72148da8e74de4abf8ebada : SUCCESS in 56m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5cfa8230a0346b799d1243e20f2e6af : SUCCESS in 23m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f58d689b0f1a4735beed83d073d60960 : SUCCESS in 47m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ad44560f0c3462ea538b03ef9d1d6d7 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89094acbc6b41c1a73587f0356d7b42 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/212ad2ccf2b44042861be34f330d91c3 : SUCCESS in 9m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0d0663b33c94908b1fd77c8d0104aa8 : SUCCESS in 46m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6139ef03ed564bb79475f8c17c936a50 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"72bb057b9fbd2d826755976ce45d79db69e981f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 19:41:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a50b62643245b68edf29cf830e7b0b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd8046b835b048e1926e24a6095c3847 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"eae3288a8867579acf0d0f5ca85395a9720020f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-31 01:00:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c83f0822e9d06ffa70315ae02774077141b4afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 02:09:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96b7b14fc4d340739663546e0ae78c76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56fadf9eee2c411cac0d53b7671a949f : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/1eeba4d7824a4031881b1a40d1fbc03c : SUCCESS in 1h 00m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/338cccf8ce3e421dafec21eb300a7f65 : SUCCESS in 40m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6b27f7a16264a109eaafa216f2df4e9 : SUCCESS in 35m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c07d6dd509140a68644f0e237b6fbb7 : SUCCESS in 45m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7f467431e464993a371a2a7301e7657 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbf7dee8180746c5a6f71fdc871562b1 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/199000987afc4d6da1f9268066a040de : SUCCESS in 15m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/512b3bf00f7c454cba4844266cab3eef : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47981e4d7473430ca47f2a11d9cbe028 : SUCCESS in 18m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d16485b992b4ce68c11ccf8e4ea4fbd : SUCCESS in 8m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/594e3eb31ed84812a9d67479c1a00724 : SUCCESS in 21m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9bed105ef5ac4584a8acf2ae29fe47a3 : SUCCESS in 26m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/311d021f93814e9ca9c02f0db39c2d1b : SUCCESS in 11m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/129ca65f0f954a0b88bb1f796bb0ac7b : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17feee01eb4d48d59acb12db27d4c5f7 : SUCCESS in 9m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e66989ff9d8e42278f5dea9353faba28 : SUCCESS in 15m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8a9f1f131454762ae2b115370a41d6b : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4746e70f092d43dcbf2aa1efd28b6479 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76e7e6e9369d4e75a1eb1a263f8cbefd : SUCCESS in 13m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/443c5cd738da4261847c64b03da165eb : SUCCESS in 4m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1aa7391f4e4578bf21a7762bf35a86 : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff7eebe35b104416b0f4e654ef5180b9 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48c1e262501a49b2ba12ff2219677ada : SUCCESS in 53m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c16f796ef4f4f1a995556767de00cf6 : SUCCESS in 30m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33d2cb82c2cb45bf827d4616af42b9e6 : SUCCESS in 48m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e30f7f034f5b4c7a8a44018ba1d956fd : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a5068dbc394d1cbe9cd822e5e7257d : FAILURE in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60a43bc3305d4b289e8059fd911490aa : SUCCESS in 24m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98cd8a06436b41a9b64cb06d6c4162a8 : SUCCESS in 54m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a4bd5139767473d8e291507a4ee2ae2 : SUCCESS in 1h 03m 27s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_updater.py","accounts_in_message":[],"_revision_number":2},{"id":"b0abb727665a8387ec14910cd901dbc27817e87a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 02:39:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0137e170984d00b8681ce8dc18de19\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a3591e1a9814b8a906fad666e7415a4 : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":2},{"id":"02d892b5a4baf6eaf65982c6917d4e6f8e766857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-31 05:54:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"323136a8e9aa4e73b6a463fbd6728aebb7073287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 07:08:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b092ec2e86a4c0fade4b9d8959d3354\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67951ef01c7946da99b26b0e893a9925 : SUCCESS in 5m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1e05afada0ac4bae9cd46ae21a09271c : SUCCESS in 33m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6fdcf25b46564cc7b91fa177c3d6ad4c : SUCCESS in 1h 09m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00b51dd00bfd4da7b71f40d0b28811ec : SUCCESS in 1h 03m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38c4d5e6f0e1431994287a6c596eb35e : SUCCESS in 1h 05m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a7cca9f4ad245c8b45ab0357508c3f7 : SUCCESS in 11m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b96f4195a8a4a53b24d6eb6de547a82 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c755e5d351144603b8a52491d9346de2 : SUCCESS in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb42c485821a41aa8ed068c6a6096ef8 : FAILURE in 3m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c1f430d6dabf4f4daeccd3cf8fd1660e : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14f36dc969e84474ad04595853d6ca1e : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4af9ed4fcab943a99dcbb4c49faee0db : SUCCESS in 13m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4fcd70c2c3a04de8820344c911d276f7 : SUCCESS in 16m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b81a2d310e9460aaf7908d8b16b4e7b : SUCCESS in 11m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb99e9becc7b4e0d84072924bdcaf8d6 : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb362eedbb7847f9855b01a907490db4 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/58c65cad790044babe70fa18be5c7208 : SUCCESS in 13m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/490ad7033e084d43afe30dc8c0c90f4e : SUCCESS in 12m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fdc38b86729a4e64aabfb3f06314ac37 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b27244c735df4df9abdbaf4c37094e56 : SUCCESS in 11m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5cd2ea9b7ad54214987587247adaebf5 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/141e2e4d4e214596aba41bc5298bfc6c : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c175269f98a4268811aa148457831d9 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/72790655ed0048119655731d074bfe5c : SUCCESS in 57m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/455dee02add844a882941589f49465fd : SUCCESS in 48m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/067ebadcebbe43dabf40a327c1f24a4a : SUCCESS in 51m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10d978fa970e4f9888b7048ca596f07e : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62477bc2bd0435d8826b4c724fe6991 : SUCCESS in 2m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/057016c318d740bcb3c68fcb874f2306 : SUCCESS in 10m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aace42d17c7c4aa6ad75bbd8b44b0958 : SUCCESS in 46m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd577528c598472a916f8d32ccc862bb : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":3},{"id":"4580ab6b4cb26f06cbaa351413098bfb9d9119c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 07:12:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e2f5394efab473ba69dbf15e79d4304\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62ac830ddd5a4c7ca48ca42e9b22c95d : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44d1b8a8d02b5b12bc900d4390117a18e2e1931c","revisions":{"9ad22e39cd2213cd5269deb0b9740548fd6647b4":{"kind":"REWORK","_number":1,"created":"2024-12-30 18:19:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/30/938330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/938330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/938330/1"}}},"commit":{"parents":[{"commit":"3c9838101a1ae847d637f7f3f6720825c3d6d1d5","subject":"[CI] Remove old experimental rolling upgrade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c9838101a1ae847d637f7f3f6720825c3d6d1d5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-30 18:19:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-30 18:19:19.000000000","tz":-480},"subject":"CI: Add no-XFS-tempdir job","message":"CI: Add no-XFS-tempdir job\n\nChange-Id: I2f38436b6bb46e03a2617d7ab710c0e8e7eced95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ad22e39cd2213cd5269deb0b9740548fd6647b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ad22e39cd2213cd5269deb0b9740548fd6647b4"}]},"branch":"refs/heads/master"},"8560f4362db7fca6e08e2c72028b1bad0af829a4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-31 01:00:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/30/938330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/938330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/938330/2"}}},"commit":{"parents":[{"commit":"282e6c151d5a8a98be05e35b38b37bffc72ead8f","subject":"tests: Enforce sorted listdir results in test_updater","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/282e6c151d5a8a98be05e35b38b37bffc72ead8f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-30 18:19:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-31 01:00:42.000000000","tz":-480},"subject":"CI: Add no-XFS-tempdir job","message":"CI: Add no-XFS-tempdir job\n\nChange-Id: I2f38436b6bb46e03a2617d7ab710c0e8e7eced95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8560f4362db7fca6e08e2c72028b1bad0af829a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8560f4362db7fca6e08e2c72028b1bad0af829a4"}]},"branch":"refs/heads/master"},"44d1b8a8d02b5b12bc900d4390117a18e2e1931c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-31 05:54:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/30/938330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/938330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/938330/3"}}},"commit":{"parents":[{"commit":"1f0777d96c8074290bd8572635fb576931fe0a53","subject":"tests: Enforce sorted listdir results in test_updater","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0777d96c8074290bd8572635fb576931fe0a53"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-30 18:19:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-31 05:54:37.000000000","tz":-480},"subject":"CI: Add no-XFS-tempdir job","message":"CI: Add no-XFS-tempdir job\n\nChange-Id: I2f38436b6bb46e03a2617d7ab710c0e8e7eced95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44d1b8a8d02b5b12bc900d4390117a18e2e1931c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44d1b8a8d02b5b12bc900d4390117a18e2e1931c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
