)]}'
{"id":"openstack%2Fswift~938333","triplet_id":"openstack%2Fswift~master~If9a0fdb3749a18a9479f20fb174e0c1908a783bb","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-12-31 16:25:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9a0fdb3749a18a9479f20fb174e0c1908a783bb","subject":"tests: Enforce sorted listdir results in test_updater","status":"MERGED","created":"2024-12-31 01:00:54.000000000","updated":"2024-12-31 16:26:13.000000000","submitted":"2024-12-31 16:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938333","meta_rev_id":"1687c66b4ccbb19a3fcfa9b5fd5a5b53ab79f4b1","_number":938333,"virtual_id_number":938333,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-31 16:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-12-31 15:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-12-31 15:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-31 01:57:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-31 15:28:44.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"34de08cd2381a5f99fdaa34510f1cdafdfe962d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-31 01:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fe124dfb0934e070f4ab5263f8dc91ce4d3384b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 01:57:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/392df04382f9432cb0920e7d4c875646\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37304a7155584b5a8137a60c83dcc9da : SUCCESS in 4m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40b1a8d6bd474f9aa5de03fe1248da23 : SUCCESS in 33m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1425d56ee5624c1285612c0a7a6e8e39 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efc70cb8e3234056aa4bdfa2353ea4c1 : SUCCESS in 21m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2709fdf7534047c49b44c3bc5f9b1757 : SUCCESS in 11m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d85c7bcb2ff4d7eb17f4c49dca802b6 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2691e96cedd4bf1bacdba814d6e9a03 : SUCCESS in 21m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/186d758821174f7c89a64870c9843c72 : SUCCESS in 9m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da1a485a18da4f96ab1e30b7d4d5d3cf : SUCCESS in 15m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/64dc0e921b8b4ce49458c79c00b7c25e : SUCCESS in 30m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fc8bc43f3764572baf3b7c2940ff682 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29a2883d33394af09762896f99ded801 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e412c5cfaea486b9a2c38d74d8e23b7 : SUCCESS in 9m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/40fd34c300a64196b60dcfa5aa725488 : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/673faa2162c74a308224a3db52b11a38 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/64626b015280449c8630dca603a4f622 : SUCCESS in 8m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4022522c5625439e8dec213575d62d34 : SUCCESS in 30m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/298930ccb1af4684a1765a83ee792302 : SUCCESS in 49m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e946a42eaacb4e59a2886b4b9994790c : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b74cf5c61b411b8a432e63085df224 : FAILURE in 3m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74d8be7bc1b947b6a251cc873d0c4a08 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6214929a70b15c2331417d46b76374dd870296e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 02:59:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c54467e20e4dd5b2aa8512ff57da2f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29729de3a0ea4bafa375fff929f3ddb2 : SUCCESS in 1h 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ce8c7390928acf3e42d49fae562c66084440a514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-31 05:54:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1eca329a556376a67adc54884375861c9276f418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 06:49:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/696cda3a91e941b6924ae9cef2a694d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d17b30a8f14a2ab6ed6a25f3b216f4 : SUCCESS in 7m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f470d0c6174242bcbeed95019dab00c4 : SUCCESS in 41m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56aec4caa14842588f514c1163dde94a : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b485cfed887d42f490069c5fc1234567 : SUCCESS in 16m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98ecb86782e24b31996f5520686c5e01 : SUCCESS in 17m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/509debe61a1a492e880ac5c5983ac906 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7bb395bd60fd4d57a77e705858c7f936 : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/532b0b9cfd914a278183f0b03f7fe16d : SUCCESS in 14m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8526966541634b7d9e6603d8d6f2836d : SUCCESS in 15m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50c858166ce24f27a9c55a11b9a67088 : SUCCESS in 12m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a41e05c97bd84db9a1c36e202bc2b61a : SUCCESS in 9m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a6c73ca6fd04a38b3af9cea18d0766b : SUCCESS in 9m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04f24dfa9a064397a068658231aa673e : SUCCESS in 18m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f3abda1980d4410bf03583cdb659797 : SUCCESS in 8m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f1316e5972034b98a0849fdbcd8473e3 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/21874867d5e444d29c2322d49f394820 : SUCCESS in 9m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce68b67acc1e4721bab912dcec7441a0 : SUCCESS in 54m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce235f652d7d42efaf85166bbe42d676 : SUCCESS in 24m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b176ec9931b64c2b9acc08ab4325d54c : SUCCESS in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5c492f69c149a3a868c837f985339a : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/817eadfd12e842db8031c3224afc2d04 : SUCCESS in 19m 04s","accounts_in_message":[],"_revision_number":2},{"id":"84de4552505ab0d0e7035c6e35cddde489cf04db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 07:25:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc05a43ea53540bca44f9b0f93c0ac90\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b943b5be38c4b6c8ba8e367e66fe558 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":2},{"id":"35b94f1b15048ddb4880482b523e18867b427f0f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-12-31 15:28:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f9744fc21baf588cc168e8544df283be11ccf3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 15:29:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd9af039699f92c9171b106d5e343b05aedeb804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 16:25:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df2eede7f5db44b3828e549ddd71af67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f89901686048e6b54b99c298b5f5ff : SUCCESS in 6m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a16aadd559f649a1be0c3ed1099282f9 : SUCCESS in 55m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/76c96d69b3634f70802bb4d0380b93a1 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63fc435701b94f1198bed75785bbf561 : SUCCESS in 21m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/098449526e9c48e88462118def888b60 : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0df7209f23be4bbf9c2f833f33dc995b : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/faa46dfc2c42420fa3b9102c51b362ec : SUCCESS in 22m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec95e534fdd645e4bad614716fa31401 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bcf30d2f231b4c8287914fde1713ef25 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/06e7347299a2491baba19cffd63cddea : SUCCESS in 16m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/20022b2260ef468db2151093bfcaf533 : SUCCESS in 9m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3f85952f97f14de494d4616d90a6e6ff : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b8ec65e01683465c85b3bd98620fc7dc : SUCCESS in 14m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ec67bdeeb0b4b8db529b779f2debe71 : SUCCESS in 4m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4dfd7d021b87453aa228c5a59b7944af : SUCCESS in 6m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2447d9d432cf4b3dad733ae0715a4770 : SUCCESS in 31m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/622e45215a4e427e9fa8adf24e5c355d : SUCCESS in 30m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a89fc47d0c45472abcdd7fe20932537c : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6180cf67adfa41c3af18c8f94ed4cf47 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45e44ee48b854edcab4c541b9ace5376 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":2},{"id":"bd997e4fc394b848d6d8ac939ce3383404f66396","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 16:25:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1687c66b4ccbb19a3fcfa9b5fd5a5b53ab79f4b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 16:26:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d8a8a434d045e3aef4aed582ea741d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c61dcbb8d5846dc9cb0cc1ffef3821e : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f9c7aee2ed434d159277feef29922128 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f0777d96c8074290bd8572635fb576931fe0a53","revisions":{"282e6c151d5a8a98be05e35b38b37bffc72ead8f":{"kind":"REWORK","_number":1,"created":"2024-12-31 01:00:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/938333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/938333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/938333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/938333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/938333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/938333/1"}}},"commit":{"parents":[{"commit":"3c9838101a1ae847d637f7f3f6720825c3d6d1d5","subject":"[CI] Remove old experimental rolling upgrade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c9838101a1ae847d637f7f3f6720825c3d6d1d5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-31 01:00:05.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-31 01:00:42.000000000","tz":-480},"subject":"tests: Enforce sorted listdir results in test_updater","message":"tests: Enforce sorted listdir results in test_updater\n\nPreviously, we were relying on some xfs-specific return order.\n\nChange-Id: If9a0fdb3749a18a9479f20fb174e0c1908a783bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/282e6c151d5a8a98be05e35b38b37bffc72ead8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/282e6c151d5a8a98be05e35b38b37bffc72ead8f"}]},"branch":"refs/heads/master"},"1f0777d96c8074290bd8572635fb576931fe0a53":{"kind":"REWORK","_number":2,"created":"2024-12-31 05:54:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/938333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/938333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/938333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/938333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/938333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/938333/2"}}},"commit":{"parents":[{"commit":"3c9838101a1ae847d637f7f3f6720825c3d6d1d5","subject":"[CI] Remove old experimental rolling upgrade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c9838101a1ae847d637f7f3f6720825c3d6d1d5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-31 01:00:05.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-31 05:54:37.000000000","tz":-480},"subject":"tests: Enforce sorted listdir results in test_updater","message":"tests: Enforce sorted listdir results in test_updater\n\nPreviously, we were relying on some xfs-specific return order.\n\nChange-Id: If9a0fdb3749a18a9479f20fb174e0c1908a783bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0777d96c8074290bd8572635fb576931fe0a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f0777d96c8074290bd8572635fb576931fe0a53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
