)]}'
{"id":"openstack%2Fswift~938414","triplet_id":"openstack%2Fswift~master~I22785806f2375217dd86c49fd8e554b8a887fa95","project":"openstack/swift","branch":"master","topic":"remove-xattr-dep","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-01-31 17:29:21.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-01-31 18:37:21.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-31 19:06:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I22785806f2375217dd86c49fd8e554b8a887fa95","subject":"Remove xattr dependency on linux","status":"NEW","created":"2025-01-03 18:33:46.000000000","updated":"2025-01-31 19:06:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"12c74fde7c68524525e773e7299ece78bfc05530","_number":938414,"virtual_id_number":938414,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-31 19:02:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-01-31 18:37:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-03 19:24:41.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-01-03 19:40:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-06 14:10:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-01-13 17:22:29.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"0a30ec9efc6a0292d0e58f39ce6538ce5340c734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-03 18:33:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cecbb574c355e348c79733c614ba45d73edd7936","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-01-03 19:24:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f29d39dfe7603ec7d02e5268cd3083653035272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f69926dc9a456887c5f86139823fe3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3751c6f119cd45e0a20098e158355a00 : SUCCESS in 6m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21b34dc99caf4e5d8d38f8bb55e8ab5f : SUCCESS in 2m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e5e26d1e819f48c097d1ef49b321ce6b : SUCCESS in 43m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70e7768080f742df889f190cdbd9c041 : SUCCESS in 42m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca33354b71d44c2cbe764d45cdb30f04 : SUCCESS in 35m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dd20a277d944ce0ba862174e0aadce8 : SUCCESS in 44m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c772ba59b4244d9cb84ce5f0ff9704f6 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/166de5a3528349fc8e5f8536668f8d79 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ee063e88e774724997779f5cfea099a : SUCCESS in 22m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3bab5abe3ad54d7083572de9d45e2447 : FAILURE in 4m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/378357bf9d2446278a5944dcda336ec3 : SUCCESS in 19m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/58762950445a4f70a42a8900eedf3d71 : SUCCESS in 20m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50b51226994e412d9afab473a52c7706 : SUCCESS in 31m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a31708ac5f7840f6bdaac3646e0a8a5b : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2262499b8fd84c88a87c140edbbc2bd2 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c94c934a5d60462f85af366692a3ba8a : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f238c002b25c4e298bdee37e15a5c9bb : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03000931947c46ceba8c5144e07ac281 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ecfd80eac4c448f98d862cee7512536 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/92b14f05ea404143b9f022f6b0d6a6ba : SUCCESS in 51m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37075f5c80e64c469bcdd40bd3d99d76 : SUCCESS in 52m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbef18a1f8a341188a9b739f6d2de541 : SUCCESS in 48m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70ab27268e2c4c479f5725d9dc85756c : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1028f2631524395b083927b1af052c0 : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1473c8c09975477bad329292fb89b2cb : SUCCESS in 17m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5c4ac781ce448b3a9b47545972cf873 : SUCCESS in 41m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/108fe469f76749e2b2d792ffe0a9fe0e : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a5f9c24e5d5d3a7f0604cd71dc203780346d557e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:45:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f33002611d5b4281b8c277d0cc802553\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88254039826141788b880c5ebc8bffdf : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b2195c9bbe7db3bd8103161e102c9a0810710400","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-06 14:10:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c7eebb07ce28e3a3459d27969624bd56b51b988","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-06 19:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3082415210cd5572775e0786f72b21cb270afd0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-06 19:42:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"799df4c85dfe21dab817f69b72f2548356f418ee","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-01-06 19:58:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8a28dcbe42bb85468ea5c1b3c0785bfd12a9713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-31 17:29:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"188ca30743eee0399265ac59750f59bb9b850796","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-31 17:30:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61ebd485e0a97fbc73dacd2f4f146f3bec5b554e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-01-31 18:37:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5513f9e409436afc4a878979a85f03ddf2de6138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 19:02:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89197ba6bca548659f407c5f74a730e2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0725d03656fb4099a021fc6420e0ac96 : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":2},{"id":"bb2a861508aeb3a89b5de6eac011d17f8717b5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 19:02:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36fc0aa9b8c5482d829714df5fa24175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8543d8a04a864107af9665b048952947 : SUCCESS in 21m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a94fb7bc01d445a4875bf1840d1cb503 : FAILURE in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/66a4246e1197419b816bff563b638fc4 : SUCCESS in 1h 00m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/252891d5cb954e4483ea4f0e81a1f5a9 : SUCCESS in 1h 27m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d588b01210d4d1c8e836ccbd51e950a : SUCCESS in 42m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dece10df8f440eab2adf83b2af837f9 : SUCCESS in 1h 04m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe5e0e7b5f2f498eb0659815047afcb5 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4f1dd8b814643cdaeb1ed02e1d8533d : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83acddc79b1d46c083b75c2fa266c117 : SUCCESS in 25m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2225ccc64714b12866f2347dd209c9e : FAILURE in 4m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/454a4eb926694f9eb6cbdedb5405ce0c : SUCCESS in 18m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc59a7326d414bc6aca23e3ad656263f : SUCCESS in 13m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/269261ad03a4487881878c53c06ecf63 : SUCCESS in 27m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83cc06a167f04ba2a86077f9cfe8d793 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0d5ae07d8de247df89bd169449f5ee5c : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0429dde7f43741f3bb6e90944cbad934 : SUCCESS in 11m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2e6bf0872314189a44603b886886b73 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a4f24e47b7a14d2aab4489b33fcd5fc8 : FAILURE in 20m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/414de8cb7f024375bb821960b416e6ca : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2602c25adb3c440bb11adf42580a5e3b : SUCCESS in 1h 00m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9b7ee9da30b4822b1f6a34af815fd6e : SUCCESS in 52m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bdfe9d3481d4d1abc1f88f12c31fa53 : SUCCESS in 53m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74b82b14111f4b649bf10fa7ae2f2d9b : SUCCESS in 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88002c07f994819bddc9d7b34e58467 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4103696e75f3405c805db21f0f821781 : SUCCESS in 18m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7992d23067824aa9ad12f2b1f40f06b0 : SUCCESS in 38m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d3243ee162f415d91df7d68dd6db4e5 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":2},{"id":"12c74fde7c68524525e773e7299ece78bfc05530","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-31 19:06:06.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b7854bbe0c6e905eea1052753c4f3a16dc720251","revisions":{"90fcffaa80b9bf88aa502fcdb756cfe8394de3e7":{"kind":"REWORK","_number":1,"created":"2025-01-03 18:33:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/938414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/938414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/938414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/938414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/938414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/938414/1"}}},"commit":{"parents":[{"commit":"06e09b9eced6e53e85916126e0b65f56decfac77","subject":"Merge \"Drop py2 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06e09b9eced6e53e85916126e0b65f56decfac77"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-24 17:06:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-03 18:33:38.000000000","tz":-480},"subject":"Remove xattr dependency","message":"Remove xattr dependency\n\nThe standard library functions in py3 suffice.\n\nSee https://docs.python.org/3/library/os.html#linux-extended-attributes\n\nChange-Id: I22785806f2375217dd86c49fd8e554b8a887fa95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90fcffaa80b9bf88aa502fcdb756cfe8394de3e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90fcffaa80b9bf88aa502fcdb756cfe8394de3e7"}]},"branch":"refs/heads/master"},"b7854bbe0c6e905eea1052753c4f3a16dc720251":{"kind":"REWORK","_number":2,"created":"2025-01-31 17:29:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/938414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/938414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/938414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/938414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/938414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/938414/2"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-24 17:06:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-31 17:29:09.000000000","tz":-480},"subject":"Remove xattr dependency on linux","message":"Remove xattr dependency on linux\n\nThe standard library functions in py3 suffice.\n\nSee https://docs.python.org/3/library/os.html#linux-extended-attributes\n\nChange-Id: I22785806f2375217dd86c49fd8e554b8a887fa95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7854bbe0c6e905eea1052753c4f3a16dc720251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7854bbe0c6e905eea1052753c4f3a16dc720251"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
