)]}'
{"id":"openstack%2Fswift~938431","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ie829f0c1467c2e030e690649688569212d7abb3a","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-06 14:14:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie829f0c1467c2e030e690649688569212d7abb3a","subject":"Merge branch \u0027master\u0027 into feature/mpu","status":"MERGED","created":"2025-01-06 06:09:17.000000000","updated":"2025-01-06 14:16:49.000000000","submitted":"2025-01-06 14:14:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938431-merge-master","meta_rev_id":"1801b9fe9e63f07d08141c89d8c3fdce1b9d0756","_number":938431,"virtual_id_number":938431,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-06 14:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-01-06 13:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-01-06 13:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 06:47:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-06 13:16:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"1a10c9f88071d368b3fa9933f8af81c89ec17e0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-06 06:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c98ce7bc934af7e211663beb5067a5f5b2c3a9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 06:47:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2072413c954dbf8f18d88d01d3a66a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1434d5fa7d2f4c118f2a4384cc82324b : SUCCESS in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2932ff65544e46889802c3637df8c855 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cc2e682154d401b833dd2d25b92c8d8 : SUCCESS in 5m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c80016818aab4c9b9fe33ef20f966e33 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fecd52cabf864d9483684f0f94ab9e0f : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7eed8d2801bf4a8d8402c2df538af673 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9020184d8d974a09b189b40bd8c5bae7 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e40252d8f7d42ecb5636ecb4a2ac135 : SUCCESS in 18m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/43f20094ea574c528b6a45e1d41b8528 : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e11436cbf94f4ecabecf6f4382904eed : SUCCESS in 26m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0945edd889df4586b49f24c0bba85eae : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/279ea008715f419994fa0e27bb90b629 : SUCCESS in 9m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6bb6ffcf89a446e97ac28bc23d16e84 : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/054a2835c5d043abaa0733edce0d5332 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe817d5cc4dd4143bd2b244e8e5ee5c3 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04acbc421b10455fb82d6f93d2b14af2 : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68bd5bfbf943467aaa391bc5d06e1a00 : SUCCESS in 36m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79e62b4caeff4443a02566b93ec0f57e : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddb78bb470f40d0aa6b43a8f9a41d16 : SUCCESS in 3m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09cce461fbc049af87aa5ce2d40c85d2 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ddc6c570aca92081bb34be6f8a5c5525201798a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 07:24:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ee9162b52048e097edcec2bb442cab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/098bf0757de0491fa2344cc5f63ac4d6 : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"78e629b800b08cbbac505394d1b80e63cf807af5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-06 13:16:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"672d2b82cf50d05fd7373d088056c262355b2892","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 13:17:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"554644aba9b4c23a0373996f7a90069ee509c838","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:14:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/387f9149fc6640e2b7ade4340e37b4dc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/616fa643f16345db905217a9ec74b195 : SUCCESS in 1m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c503f360854dbcb64fbe4043d2a70d : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b6516c58e4b4075939d56999f91f324 : SUCCESS in 7m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b5c5098090cb472fa061647e850c2681 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6fae0d8c5bc94a1a9a59ef8471f4f967 : SUCCESS in 13m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/593c8bf363b3409baaef28adb561ddf6 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5fc686e25bf7480398a18d87e1baca8c : SUCCESS in 21m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/099ac0fce7df4068b3e225fc20a80a3e : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d7229e22e529478a938f59f862c0c1c8 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a7517159a1c34e1dbc2b8309eb89a177 : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5539f08333494e4b9a481ca59ad406b9 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7174fe4047eb4a4080c7820d7259f74f : SUCCESS in 5m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/75cd6956205042d282a02cb1a71ad074 : SUCCESS in 56m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8a93d33b49046d8a467b3bc9941c063 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10158b24d4374ffe943f33a53bb8d796 : SUCCESS in 2m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/923f77761f7447a09539c3e2d971c752 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b4f78ee4bd9589f7fd72bafeb6fc72f4fed27c00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:14:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1801b9fe9e63f07d08141c89d8c3fdce1b9d0756","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:16:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6be2df7b9af433ea256b848c9097807\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a105e491e3849d8a047b258017abcbc : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71abda945d314072a27b9dedee1a88ba : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/102ab9bbfc0947ad972e799ed4a2215d : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8836d81f9600d085e2c0a500c05be2eb4dee853c","revisions":{"8836d81f9600d085e2c0a500c05be2eb4dee853c":{"kind":"REWORK","_number":1,"created":"2025-01-06 06:09:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938431/1"}}},"commit":{"parents":[{"commit":"638596e0a3814c7ab26df9cef5b3384e857f1163","subject":"mpu: repeated completeUploads get unique timestamp offsets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/638596e0a3814c7ab26df9cef5b3384e857f1163"}]},{"commit":"95b9e6e335878c8bc97088543a630cb9f67e7262","subject":"Merge \"CI: Add Dalmatian upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95b9e6e335878c8bc97088543a630cb9f67e7262"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-06 06:08:47.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-06 06:08:47.000000000","tz":-480},"subject":"Merge branch \u0027master\u0027 into feature/mpu","message":"Merge branch \u0027master\u0027 into feature/mpu\n\nChange-Id: Ie829f0c1467c2e030e690649688569212d7abb3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8836d81f9600d085e2c0a500c05be2eb4dee853c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8836d81f9600d085e2c0a500c05be2eb4dee853c"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
