)]}'
{"id":"openstack%2Fswift~938543","triplet_id":"openstack%2Fswift~feature%2Fmpu~I2ac25ffd752c9c689c8b79cb39a413de4200a46c","project":"openstack/swift","branch":"feature/mpu","topic":"mpu-py2-drop","hashtags":[],"change_id":"I2ac25ffd752c9c689c8b79cb39a413de4200a46c","subject":"mpu: remove some py2 specific code","status":"MERGED","created":"2025-01-07 10:22:51.000000000","updated":"2025-01-07 12:32:07.000000000","submitted":"2025-01-07 12:30:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938543-mpu-py2-drop","meta_rev_id":"6ddd9ee857d47f6b6f5cf939761f56797fb17bfa","_number":938543,"virtual_id_number":938543,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-07 12:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-01-07 10:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-01-07 10:23:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 11:26:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c57ab5061d9626daf63bd4eee5314b7780de3bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-07 10:22:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50dc72b1ead138de5130dc48ec91ad9bc92bd1e9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-07 10:23:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7eebeebd083a712a6440f4d4f8dafff8457ba40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 11:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977a0ae9d52d47f88db06c4219895d3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89402eebaf3b418bb5e5f2a38dffa7f1 : SUCCESS in 8m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa67bf762d2245579e5bff45f594ca2d : SUCCESS in 22m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed6f916e2ad94191bb1c1caa5239e28d : SUCCESS in 25m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cd6122f07804e859c65aaa345b8e5bb : SUCCESS in 12m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6b5cca54aeb4539bb85e98a02a680e9 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c74c614a01304462b263316dabf856cc : SUCCESS in 21m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f9ca450b9c149a3aefdb8d932d74fa5 : SUCCESS in 9m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4bf96e4386cc4406beab4508ebb91b52 : SUCCESS in 35m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3fb7156bdb7d4296b78c1f8f7794cc16 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/20ed88f66cc34b0caa49adbd8d679017 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d9a6508257b431884ecf84421ca9184 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c98e75752a8f4b069e0c3de8f9c09ed2 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6095fab8cc444bd0b8ca29f0d3a917b7 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/af53afdaad1d404a97b493471c24f63b : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6fd02a0a3fa4e589073490cf2caa184 : SUCCESS in 56m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/106380009ed441b8a02a552f2ef3770c : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c460b17d9d594edb82e64079b93ec8fd : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/785370351aaa4116bbe0268bca728770 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a58fc8f5f3ffdf4ccd8546583cc0ad5b8063a947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 11:26:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe8770e182fba795caa8dbab56114a03fc9bca6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 12:09:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a0a52675004177bfdee8bf22a0fe44\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04bb9a4b3a584e3cbe7ea19390078c03 : SUCCESS in 1h 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6ceb41aa5b597bffaa900037c4042d9927ae677e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 12:30:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf946a855abc4f52a5f75575df23b4fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990e3b6c2f15428a9f4bf178fa83fe5d : SUCCESS in 7m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/daca3e8569754e148df52aeff14c66b4 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a7b761fd544d18a4b81b93ed554c22 : SUCCESS in 19m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6ffad3cf2084302bbaa7a50837b58d0 : SUCCESS in 20m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eee9607079e64ec5a305f1bf577102ba : SUCCESS in 26m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9911f727fa7f4e219f1d6dec10358bee : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f66b76ec961411e9170c15dbde557e5 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/92ea1639be274f899ff85f351f0bff6e : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4119eb4106b343388397c72c0ed39a33 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b99fd981d14946e284f0055835836ce7 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/05e4afc638fa4d12b0c6282f293825c8 : SUCCESS in 53m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b76e607ebbc406b80360d3c4533dd0e : SUCCESS in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ca3e91bcc44b4abe793af3ee53a429 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21240dd952f940acbe59e488ea377165 : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":1},{"id":"47364c2d8cfe17d155ca4d423edac61e08fffde9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 12:30:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6ddd9ee857d47f6b6f5cf939761f56797fb17bfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 12:32:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99685958be84586951b93f2c9be921e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae280ac0c8a4abd8057a45629c80ea3 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6fcc1d74a22b4100ae97e162b8d0ca03 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a974bfe72f37ba50b0c9efd51238dd76eeee675","revisions":{"3a974bfe72f37ba50b0c9efd51238dd76eeee675":{"kind":"REWORK","_number":1,"created":"2025-01-07 10:22:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/938543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/938543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/938543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/938543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/938543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/938543/1"}}},"commit":{"parents":[{"commit":"8836d81f9600d085e2c0a500c05be2eb4dee853c","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8836d81f9600d085e2c0a500c05be2eb4dee853c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-07 10:18:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-07 10:18:41.000000000","tz":0},"subject":"mpu: remove some py2 specific code","message":"mpu: remove some py2 specific code\n\nRemoves some obvious py2 specific code and comments.\n\nChange-Id: I2ac25ffd752c9c689c8b79cb39a413de4200a46c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a974bfe72f37ba50b0c9efd51238dd76eeee675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a974bfe72f37ba50b0c9efd51238dd76eeee675"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
