)]}'
{"id":"openstack%2Fswift~938550","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ia72d476b6b8d281516c0afec28ce19b06cda6cf9","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-timestamp-offsets","hashtags":[],"change_id":"Ia72d476b6b8d281516c0afec28ce19b06cda6cf9","subject":"mpu-auditor: pass user container to constructor","status":"MERGED","created":"2025-01-07 12:30:16.000000000","updated":"2025-01-07 14:44:27.000000000","submitted":"2025-01-07 14:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938550-p-mpu-timestamp-offsets","meta_rev_id":"8ebed3eed8c4bb7e7eacde602df05d92e8f42ac2","_number":938550,"virtual_id_number":938550,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-07 14:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-01-07 13:46:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-01-07 13:46:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 13:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fda1056c7a05fb77fd13d4020efe3c33fa74dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-07 12:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a325013e8e13e07079dbc92782dc03fad1074fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 13:33:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34a7c82568344e4ea6b09f651f7279a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c96091e5fba4d00965f23268b927719 : SUCCESS in 8m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a2f2b9cee88484a88e358ed950708aa : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/415a02e959f34464b9df6dea501e914c : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35fd8d54fbea4358b08c1948b0c089a4 : SUCCESS in 24m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b5238524ef53477daff0628cb74fc0d2 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f9257694ff0848019ae7d8b02aad6662 : SUCCESS in 20m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4da65be658824a329923e61ef7574c62 : SUCCESS in 9m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2a33c9aff01548699ed87095c813876e : SUCCESS in 32m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/62be62424b744ce19f4b617ee0b2aa00 : SUCCESS in 9m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d59e9e7287394eecb2c03169370b9e82 : SUCCESS in 11m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/afdddf3f2dd748a095aa914377825158 : SUCCESS in 12m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c05f05458f1243d498a7348ea6fea47b : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a33000f496b443eadb3c730ea35657e : SUCCESS in 13m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa01dd2f72674cbdab376611f33373a7 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb1d9daad39e4e3699387d27638e2c63 : SUCCESS in 56m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d582031a185442bbfaf084d2fe9a1df : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a796b66020584c0b9af82f42e5a322fa : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ca844f6712c40568d4ed5edb31b87c9 : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ac5e351d470a31bdb4433f09474df1de8532516f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 13:39:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79605d72b3eb44bbba7b5d0d7dbdf29f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76a9f36dc22d476b8d767a210f7a5a58 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e08f9eb0c1228c9c6441b74f0c03b5fc185860f4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-07 13:46:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1fe79909898482c38436dcfa16460038468cd88d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 13:46:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a144b4023fd183028d165e8896524d1a497445f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:43:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9938988f6c4699bcd49676658f449c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0ee020ed744ac580f933da4e5b162b : SUCCESS in 9m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7c218fcd591d41f9aa993faeb343b20e : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07c4429a73134787ad9183e1e3b76645 : SUCCESS in 19m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/601163f9f3d144f395d4cc8d020f8069 : SUCCESS in 20m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/41b496035ace4e3faba3215ce3e588bb : SUCCESS in 30m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/54cbe1c74336499194d3e243c8f21921 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/405cefcae1a3436eb0f888f47acd3213 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab3853efcc9f4237ad7f8d42142cffc5 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d61deffda5842ccb095abbe5451e305 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0797022bf964b36aa966599825f26ce : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25c31fdd6b884e5d8d31fa6d0dc1fa22 : SUCCESS in 47m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64eab6ac01364977ac3c9e0cb8459e5a : SUCCESS in 21m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b8a273440c45e1994e65f21a050a30 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/009e3f755ca141b49a9b752e1d7781ff : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"14642d452026e825feb8597778985dbe87e1293e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:43:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8ebed3eed8c4bb7e7eacde602df05d92e8f42ac2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 14:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c863d805b02f4f3396eead43f9e842c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908cee6ab4384c1b928033bc58e057cf : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/afdaddabcc5f480790cb6ffc9b21a779 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23aca1ec9adb69b4277ca95929c49827c020cdd9","revisions":{"23aca1ec9adb69b4277ca95929c49827c020cdd9":{"kind":"REWORK","_number":1,"created":"2025-01-07 12:30:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/938550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/938550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/938550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/938550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/938550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/938550/1"}}},"commit":{"parents":[{"commit":"3a974bfe72f37ba50b0c9efd51238dd76eeee675","subject":"mpu: remove some py2 specific code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a974bfe72f37ba50b0c9efd51238dd76eeee675"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-07 11:10:06.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-07 11:52:20.000000000","tz":0},"subject":"mpu-auditor: pass user container to constructor","message":"mpu-auditor: pass user container to constructor\n\nInfer the name of the user container when choosing which MpuAuditor\nsubclass to use, where the audit container prefix/suffix is\nparsed. Then pass the user container to the constructor.\n\nChange-Id: Ia72d476b6b8d281516c0afec28ce19b06cda6cf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23aca1ec9adb69b4277ca95929c49827c020cdd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23aca1ec9adb69b4277ca95929c49827c020cdd9"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
