)]}'
{"id":"openstack%2Fswift~938630","triplet_id":"openstack%2Fswift~master~I14ea6bbd55512e8798e5cadf4f6e68f95961206b","project":"openstack/swift","branch":"master","topic":"sqStatsdClientSource-patch2","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-08 10:56:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14ea6bbd55512e8798e5cadf4f6e68f95961206b","subject":"tests: use a method to test a method","status":"MERGED","created":"2025-01-07 22:38:40.000000000","updated":"2025-01-08 10:58:54.000000000","submitted":"2025-01-08 10:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938630-sqStatsdClientSource-patch2","meta_rev_id":"e5c11062eebe268f8f338d17c7288f43e928629b","_number":938630,"virtual_id_number":938630,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-08 10:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-01-08 03:21:13.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2025-01-08 09:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2025-01-08 09:24:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 23:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 03:21:13.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-01-08 09:24:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"e1c7c5f06d1bbdfaa639cdd3ec82649650e20f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-07 22:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3b56760c9d1cadb84c9881b643f80e1c68f24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 23:14:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f248fb5e094c87b448049c450defac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2bb44a118c4c21874e2e06fe4d0a4c : SUCCESS in 4m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dffdae0b00184fbf8cd59efce81c836d : SUCCESS in 35m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7cd581c6d3047ad88d6d90bd501a3fb : SUCCESS in 11m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b856e6b60804ef692fbbb6283450887 : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cd9a9ae38ed4242a081e1fc3f9da65e : SUCCESS in 12m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a05a72f8079458e9c27f34d33995c00 : SUCCESS in 4m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/31e87dbdcc67446bb95abad2a6607423 : SUCCESS in 9m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a1d6934fda946758c72b7810d6ef6b6 : SUCCESS in 22m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/55508d1203a84bbbb0a50260fa4cca22 : SUCCESS in 18m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/005c4595808f486fa7160aeae65f7d4a : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/952096ef2d444c6f972ea31d31a1f7c9 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee676ce3712d4e5b94c17fa6da7510f2 : SUCCESS in 14m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03c8c5fe602d498394ae27cf098fbadd : SUCCESS in 31m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40f8a28c5f684f1e98e24a7f80ff085d : SUCCESS in 31m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61a62d4f363e439c959febfb977369a2 : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4140be159b92458083d5a93327b82c9b : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/788144f317134f9e85ead11dab04083b : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8adace04dc6724094a502c9d14464970af638f69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 00:29:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e49ae7f5b06443ba4dcaa712fbc43c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2a856fe8c0549ba8db2b615f5291fc1 : SUCCESS in 1h 48m 59s","accounts_in_message":[],"_revision_number":1},{"id":"25942e7425c7e0ee5e4bb50b9eef4d281e6dd4ae","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-01-08 03:21:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb33102a6db3f5da3fb2259520fb261c698e9e6c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-08 09:24:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87467a6b5954e09eb762dcbe115d4f12f05fdaf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 09:25:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ace35eab95d74d20ff140494e58f577218fc730","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 10:56:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972040ac764f4335ae61d0c1fe8006a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50595153295d4d3891146d98861829d5 : SUCCESS in 6m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dc391b57f1f4086becd991bed998baa : SUCCESS in 1h 01m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3cb327cc053043cd804eceff67a51ab0 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9743187f92374589a6d5ad0963a536bb : SUCCESS in 19m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5a0940d0d4e34e98a284900a572d17ed : SUCCESS in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9b219f7b8cf4e7da0c99140b2d2c0da : SUCCESS in 30m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/90390e47295b452a8f03246050022ab0 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9df9724077eb4649a18b5c35682a6d9d : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7a8402488f7a46c6af687941c863a6cb : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f2de1e620bf4d8d9b997c30812e9271 : SUCCESS in 5m 39s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59fd321f61e34ba0a3eeb01d3d4eeeef : SUCCESS in 6m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c797254694144f37ae67d82fffaaa9b4 : SUCCESS in 56m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd62265c048f44218cde61d8b7f7b5e8 : SUCCESS in 49m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ce8df315fe4656aec021dcb402b982 : SUCCESS in 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14288f1ebcd94213a4c8f55d7f2e3816 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36055093e2f24ae68d5bcd356a3baf99 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7ca701bb434341b3ec9912e8f321b02dfd1dc2e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 10:56:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5c11062eebe268f8f338d17c7288f43e928629b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 10:58:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce72de56a4849eaa1f8933715f0067f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e167cc3a05ce4cbc8a46611629a2a016 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1db0a1d3140d4f23b5b16749bdf12ac6 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a633d06fb287e279729c681027aa8e44205da6e","revisions":{"6a633d06fb287e279729c681027aa8e44205da6e":{"kind":"REWORK","_number":1,"created":"2025-01-07 22:38:40.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/30/938630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/938630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/938630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/938630/1"}}},"commit":{"parents":[{"commit":"7d3a32d107997c7cf1b6df590d57c0ccc4d13a42","subject":"tests: relocate some logging related unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d3a32d107997c7cf1b6df590d57c0ccc4d13a42"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-07 22:38:08.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-07 22:38:08.000000000","tz":-360},"subject":"tests: use a method to test a method","message":"tests: use a method to test a method\n\nChange-Id: I14ea6bbd55512e8798e5cadf4f6e68f95961206b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a633d06fb287e279729c681027aa8e44205da6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a633d06fb287e279729c681027aa8e44205da6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
