)]}'
{"id":"openstack%2Fswift~938631","triplet_id":"openstack%2Fswift~master~I6aa16fda9285c9fc3816da6fbff2615bd14a020c","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-03-03 20:08:13.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-03 22:10:26.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-03-03 22:10:26.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-03 22:10:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6aa16fda9285c9fc3816da6fbff2615bd14a020c","subject":"py3: Fix unicode-header-name handling in bufferedhttp","status":"MERGED","created":"2025-01-08 01:44:53.000000000","updated":"2025-03-03 22:11:37.000000000","submitted":"2025-03-03 22:10:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":110,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"938631","meta_rev_id":"6f8dc1008985e7ba97966638b64527ed2bbe4dab","_number":938631,"virtual_id_number":938631,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 22:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2025-03-03 20:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-03-03 20:08:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-08 03:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-08 03:46:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 06:28:19.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2025-01-10 10:35:49.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-01-10 17:07:10.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2025-01-28 17:24:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2025-01-29 20:29:26.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"662af752ec2241d61913edba716f8f4f3f04d2d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-08 01:44:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac253888c236963106d8a1f6de55723fea12546f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 03:35:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cea4e260fb11434489c5e740ad45a5a7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90d765cb556d44f293999086d7650433 : SUCCESS in 1h 48m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3dc7b418694c8e5a649f731e78908d7cc0bd3fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 03:46:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11aca9e2c57481987ed9b2268aee108\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9cdd6df61d845fdab467ffe0659b21d : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/40df8a83b7ce4ef6805592533f8b3a4a : SUCCESS in 51m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/884098d27534407c801aaa7a9bd11cff : SUCCESS in 45m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6fefb0a74c7041f28618b8bf11dbf0e7 : SUCCESS in 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/269ac05f92794fc88217bab90dd2e72b : SUCCESS in 1h 22m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df248e6fae77400b9d2618124715a5fa : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d2e211c1b1c4115b528de8a3000fc03 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d2f381453634549899ba22ecd8ba7a2 : SUCCESS in 16m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c338f67f33041f6afb800df4d4076e1 : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6422b09f1a4347bc9df73d7ace78ef53 : SUCCESS in 12m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c65a75137c8424b897f1ea64f638f8a : SUCCESS in 18m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/57a0ca3bfb90485a8242acb843362713 : SUCCESS in 18m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdcc9e3bd2514c3c9a59a78fadb207b0 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9adbad396ec44100856642525968b908 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/106f2eda3c724048836069de474b0a61 : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10d160f33cc546a5900fd0591922d57d : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5e6b29184a84903b37ae4e4f86472ca : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfe9818f733243a5adbcc2a6650217e1 : SUCCESS in 4m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e0fd389b75d41d698e562d18ec75fd0 : SUCCESS in 1h 02m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/88b0f15af5dc42fa994748783c291c20 : SUCCESS in 54m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cd0f91c40a94bf48ea227277fdc4305 : SUCCESS in 25m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13c2701aebec4f72bac30c3c261f4e78 : SUCCESS in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ef5b3280e2489aaf75af1e0e508449 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b97568805f74ddbb208b42408f83819 : SUCCESS in 18m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea48b1787d4f47c5bb22b9fb006d34e1 : SUCCESS in 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bffd1b3bed34b9992db0447ab299fc1 : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b9415a6c44012c359acb86b765c4fd070c642ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-10 01:44:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"013483bacd059460296b7a5bdf90ab1666672bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 03:14:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efbf4e99018948edbb42cc0f38c765c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b978345c334f759f4c8512f2858291 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/510c05e3513247919ccb0a89ff84f6bf : SUCCESS in 1h 22m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/564f3642391348e8b6b798238ebea8b1 : SUCCESS in 35m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b4cc266e3904e5596bf134e7584b02a : SUCCESS in 57m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c686b1eb6cb41358846380aae2e5681 : SUCCESS in 1h 11m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0372b7d16c71423db1f904f4eb768503 : SUCCESS in 14m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55f33c6594a84720b28a54e172ff77d4 : SUCCESS in 23m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f87ae9df7d0d485b9d83de10c430e3ee : SUCCESS in 21m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7abceea72034675936a0bd435bfaf0a : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b3e640a66d8471989e62dde1d8ac4f4 : SUCCESS in 19m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5a59f1948e814efa8a95c29abc3da6e0 : SUCCESS in 22m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ee1ef8e1251f46e99fff584166df7d21 : SUCCESS in 28m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/46100c89c8ad4e60b393aed031806bf6 : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/be5761bd35c9470a8a86499b07a9340e : SUCCESS in 8m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1375422cc06a42f99f0254acdea3336d : SUCCESS in 20m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a7e7300e683407b8a1e738aebb6b560 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e076f8f9bcf41daabe1dbdbf2b960e3 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a924df55bee4df8a0c7f3510eab3bea : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/809fed16a62f4abc8aa46d9866d1769c : SUCCESS in 56m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f15ae6731d54cf6b9a6edb526dbf9c5 : SUCCESS in 49m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0efd521e7c6a4ceb9c50be8ae0e8dd66 : SUCCESS in 40m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6ead774473641d786847fe9848619ef : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3981d664587342b395632c99bd585649 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26bec0ca8f514b00a7a402c1d7516f45 : SUCCESS in 20m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43ab4c66d98e4901bb0ddd1daaad493b : SUCCESS in 59m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95182dd81b3b4cd6ac082e77c5768af1 : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1b0bebf45f4a668c4fb7bd0fbaa742ffc0f1d6ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 04:08:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73cf66f401fd4613a8df074b33596dd6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d1fbfdc00e5c4a629c2fa38017f7ba8a : SUCCESS in 2h 21m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3b999ad0a4c8aa9202ce84ac317540366054b5a2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-10 06:28:19.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4de5ab051d58a889d0d2305a9a1fff6f6361ba3d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-10 10:35:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"57195f23c6ec2ea220c5827961294ae0eada0e90","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-10 17:07:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"306b321c2e134c7afe0221a4ea9adf8998142beb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-10 17:13:25.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee6fbe4be118d9a4c48540dc560d8d3276efe0bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-10 18:37:41.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"520dbed0bb038db8bf3c82d7d6eaa00dce4c8221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:56:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb5807a79254acfaa3fe7d46bed3e4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b68605eead42629ea046d1f91bf183 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/2919de3bfb4f48f3ae0670a82457adaf : SUCCESS in 1h 09m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3ee723da1164ab28b4282ce18d20e7b : SUCCESS in 1h 05m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f345f9e7e5d44528df8ee54bf642855 : SUCCESS in 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e78b908a2254cd9844efd31a270974d : SUCCESS in 1h 03m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c104980116f45539c51de1898ed62ed : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7e3b4a69f71413dadab4341b1ebc7ed : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f31504f23c804a9fb6ba13f5f67c338d : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d86bd693f70746069efd1ebe979fb21d : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21aaaafdfe874e04b9a939e50b2497c7 : SUCCESS in 9m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/155ddc6cd56b488bbc66235becb36dc9 : SUCCESS in 21m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ef2cc56560224c2789ad2aa72ec61989 : SUCCESS in 14m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/962d2eb58f2843a5a80adf9330cf6d78 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40764c7b19d249d8ba05b30765b47027 : SUCCESS in 9m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c832d5ca31e94e5d9c044c5c8869bcc4 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b333ce31a3947bca1356bf5a633b635 : SUCCESS in 4m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8eaa4d788697400eaa41e520a5ef9019 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/484c35acf0e842aa91c7046b3f9fd8d8 : SUCCESS in 7m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa364af7b4dd48c382746ef39d6472d0 : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74f3b4f01e7143e78e7a90d8d31ab485 : SUCCESS in 35m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7467dcf5dc14f4b8f6c7afc5126f575 : SUCCESS in 35m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db845f388cde422cb2cbbcf30aa8836c : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3a0dbb54eb47ae97fb22d7a080f53d : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d59b4289e9b4b8788551fd1e5c27b46 : SUCCESS in 20m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38be6c8dbe8340d9af5b794e48cbe1e3 : SUCCESS in 50m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1276dba7b6564e1f8f70559bcad47e5b : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":3},{"id":"c74df0629fbacb188c6329755291af51e144a0aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 20:36:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fec993d66c194230a61eb488a31a8e49\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4da1f5aa258f4748b6c236ca2f9b222c : SUCCESS in 1h 56m 58s","accounts_in_message":[],"_revision_number":3},{"id":"41fbcfed286fcfaa8cf275c38c2af1777803d176","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-13 06:03:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dcfaf0395d8e7db6affa0cef90dc692d24461c09","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-13 17:28:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eaff9bc83af13f55f9978998898f5eb1076d8ffe","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-13 22:00:10.000000000","message":"Patch Set 3: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a4f09cd4a8c2a24c2e8ccbe8d8e6ea30c4602c4c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-13 23:10:14.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"68d077197be46f2b46a7a5934c76406d4acb414d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-13 23:10:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a7de2d2dff7326413c2c14de435d03a6185f1c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 00:35:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15da2201bde849cbbd7be7f9405db741\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570457065071486b8715438f3c9a840d : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/b7da06cc77ae4286ae903223531fdd5f : SUCCESS in 45m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf409fb7d0db42c78565cf5c6b0c0304 : SUCCESS in 46m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcd2719fd9274bb28040fa0d152f2734 : SUCCESS in 56m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/179b70068e1946e2b65235922618a1e9 : SUCCESS in 1h 18m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b3b439ba6824c6e904595a155878de6 : SUCCESS in 21m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f74b6f09e9a4ad7a59d0cdcd3b7965e : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4a3a1e4b6284f309e582f145fdb5d76 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5cf9ecdf4e4436a819fbaca34984ad4 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/236d1916dabd484f9d3a9f3e6ef919f8 : SUCCESS in 11m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57619d29967644b3ba8e2ec789b7910e : SUCCESS in 20m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a75b159dfedb403585e664045d256ccf : SUCCESS in 26m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83a933e67e134cbf8c3a21c6bd4aad24 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/18eef5420a864ffa999495a0747b34f7 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d6a7bafd98164d04a5b6ec5b01a53c95 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1fb17b054ad84593a19bc9b8a4e9b077 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae9da199308a4ad3ab7bec7ddacd1277 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fef9829d6ed74df3840e009e33d01d83 : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/97fbb294394c49fd91f2a71483e38fc8 : SUCCESS in 59m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99be84252e51415ca01fe9688366b88d : SUCCESS in 50m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eafe5e4c34214e00b6b559718a82aa87 : SUCCESS in 49m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d9f23afe95a4f1cac9c1b0e03709b37 : SUCCESS in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b2ab24e1fe4b3092d0827e384de52e : FAILURE in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81889ffe439b4ac89ee09bb99074778c : SUCCESS in 18m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f0c099aa98d48c7863cc4ad5e8371aa : SUCCESS in 53m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/894cb5406e684b4796a5c0e70ca12b39 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":4},{"id":"2a397c50971c414f7a5a3a9d2e222d2c24ff8e57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 00:36:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a44f831097c4a25b9246134550d09c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cf7b08e9f434d98a965b6be280e5a4b : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":4},{"id":"15db50003e08db2235373ac24c9cc0457d9519d8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-14 05:15:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7032c783f3f3b84a5baa27c8d62a8750d327c133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 06:19:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c476afdef6cc4f44a9e413b6ffdce23e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60f18926703544029ad665b955a50be0 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c1d6fff983994f589c2df51586e54732 : SUCCESS in 1h 02m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2620cdef336402a96ef4875f65518af : SUCCESS in 53m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0efc40c1643425e903c04efcfe825f4 : SUCCESS in 53m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95e1b00264594f0bba3ecf96fb16bb5e : SUCCESS in 1h 03m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97501cff18d64632ad5adc309d19b192 : SUCCESS in 27m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/284d77b02b4a42108b8a4b411de2bc8d : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f432cb56ab44cce8ad0ed488a7d6140 : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48c027f3c4c14b3ebc56d60b056e2bd0 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/815bd244a6f845d2ab69d4b7f5739c2c : SUCCESS in 8m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fda6dac161e44c5abc863b2d205288ce : SUCCESS in 19m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d66f03fc0934f58957dec0c31df697e : SUCCESS in 27m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4340bf9914794596890203773e3d9357 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bd8c14d16d4547d98318a57ad8378838 : SUCCESS in 9m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2245d230f034468aa9698f03cca23a7f : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06f2de5bda0d4306a4681c078a9c4934 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1237b6f3f03e46c88130ee88022e73d1 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/df28f657e0194a809be05f12650d0b14 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1879f040262f4bb0be4313ca99764fd6 : SUCCESS in 49m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2991171649b64526b68bab694a62a7ea : SUCCESS in 30m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c4126f66e1e4cefbc730bbba9108f6c : SUCCESS in 35m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a38b75b53be1480e90103e1b591dc0c9 : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b7dff80630c4bb5a692294669d2879d : FAILURE in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8688bd6587343ba9d5b7d90cbbc31e7 : SUCCESS in 19m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e20be11f75147d687b81689b297fdec : SUCCESS in 38m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b43e82e8095c4795801dfcd036e1756c : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":4},{"id":"36cbb094732819cabd07232c171caba92a619a7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 07:37:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d883655732421396038c1e2b8e0c06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1bf6b7aa819b4805a5ee74d031f7cd82 : SUCCESS in 2h 20m 57s","accounts_in_message":[],"_revision_number":4},{"id":"8188937c43b54ee0d5c3c984bef7915f2194bcbd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-14 16:40:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9ac7c93a121555b07339b095b48cc881ca67cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 17:58:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99dce00d79364fe68d1e2bfcd3a1dff6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80ffd0af1384adf8887d4daa47bb541 : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/83c22bd7a10b40969924a224c6d968c0 : SUCCESS in 1h 11m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4769e7e33a264562b42c7482771e55af : SUCCESS in 1h 08m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad6439151ff04570a97d66b2cb66fb39 : SUCCESS in 35m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af06f9b37bdf43f6b57c7f811e7e29e3 : SUCCESS in 48m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cba79228836d4771ad1a87a0bf18805e : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14113b2c335d4fc1a0e4a669de93dde0 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14971b8e354a4e9d8be14608a2b61d5c : SUCCESS in 17m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40a6224db4f049a1a1928237359db67a : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c347026c97a64a568c2324f04a14dfa6 : SUCCESS in 16m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f670e59923e4a238ab30567afa8534d : SUCCESS in 16m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aec55d53c09d435d8a7c279eb3a4ba0c : SUCCESS in 20m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7d721da0f0d4df4bbd04e9541cdfedc : SUCCESS in 11m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/23727e9c784e48b0aacfccc7509a43a0 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0bdc831f9e7f4ec48e1d5d15b134acf0 : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d2c0abb9d92492c85138263b78feb29 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fccdd76271041389864b819cc86da80 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d75a7b712478477d94dd379bb0269091 : SUCCESS in 9m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0540046d022340dfa37e0e987f5b67af : SUCCESS in 59m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77e84285f2fc48a19140f9655866b8f6 : SUCCESS in 46m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/566d46c59ffb485fbebff539d92185f8 : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdeb4704ccfb484b8a18d782bb6a2c09 : SUCCESS in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020601bee2c842b8a976f772560cfa83 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65504e7fc9954798a5d475873f56e4ba : SUCCESS in 20m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9692cc743dee4d858d5e18691793f6bb : SUCCESS in 57m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ecc420881c0442d9976c37362334d16 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":4},{"id":"a152a0d5ab03f73e094feb85c5e7e30153233f0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 18:47:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4694b6e4fbc3413f8366245ab1af2457\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/063e53176e364aeba4bc046eb51e1aee : SUCCESS in 2h 01m 37s","accounts_in_message":[],"_revision_number":4},{"id":"7fd640607d9ec2fc742f9c4fd93a47b40bff58dd","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-14 20:04:50.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7ef9d468237f10e87a9f9c4753e662ac9deda578","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-28 14:43:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"01eba537eb0f82590cf575ec3e7c2e43b5ea02b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-28 17:03:00.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"eb086c878748cd101fd73e2c951aa616789fe816","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-28 17:23:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"105c57a60d4418aaa3047564445534b8729c152f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-28 17:24:28.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5fb03989da4e77b99c0a9ff883461e8b067bbfb8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-28 17:26:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ad4f57ac4b88ab670fb15ee8284bb711ed37bd91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 18:18:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70bdcd2ea214e1bb9c150222fa87920\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a499fcaa5f54650bffe18546b8c95c4 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":5},{"id":"8469928ffe7098671e1cc95638cc16e3ec1c4477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 18:20:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ff210202004f46a380019325c9527d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939fe809379944328003fd095e32e9e6 : SUCCESS in 5m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/df0d0a1014be45d8a2893e005758a8dc : SUCCESS in 1h 03m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62b049df51b648e9bc6283b923eaa588 : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43ca1e43c3df48d0a0790385c640189b : SUCCESS in 1h 03m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40ebed1230de4c06b3abe3cfbeb7fb5e : SUCCESS in 1h 03m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/504cea361df94d1d8f051c456bf1046a : SUCCESS in 23m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76a33b1d1f0046ebb8d91a7b935ff115 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9871c66465464d0a8fed4d8cbc133b88 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/02dad23f7e34488bbabd437dca432e69 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7bd017e38864594a41332816499596a : SUCCESS in 17m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa0d337b74ac4ae883d5427c566f4b7e : SUCCESS in 24m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/92a31b3d40c24c529b8c77e52bd8793f : SUCCESS in 35m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/88a9082119804d63a9547a3e705e40a3 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b18bdae795040ad915d9d9a85974e2a : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b55b3465b787452783b08e8339ee9104 : SUCCESS in 14m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83c05f5b347d4e648ee3c40da88bfacb : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/343ec5b022ae435dbbdc73590b778e91 : FAILURE in 20m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ebd55a8c05bc4cde9a0ea8c9c9107eae : SUCCESS in 9m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e373e1fd584f4a24ae2c9c0f49263084 : SUCCESS in 1h 02m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c419b5aaea7443e8b8f35f3e63aea76 : SUCCESS in 24m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0aaf9d46fbf0481497cdd1db6a5b6fc7 : SUCCESS in 48m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90f22c7e937249e794dced07d72e24fc : SUCCESS in 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d738c155347c42979988b0297249977d : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01ea1ccebdc34e3391b7e5828cca0cf0 : SUCCESS in 25m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9eded43857844a08ad7a14f9a5286531 : SUCCESS in 49m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e676dbb13371411d9f60a025777f09d0 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":5},{"id":"857c5fe60416d1136c4bbcba1ef6069212d02446","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-28 23:28:16.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c7d38806b45a652b389d482fb011ce64fe86ab85","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-29 01:08:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ab1e20ee1183649ac19d04367be14d9564a8afa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-29 20:29:26.000000000","message":"Patch Set 5: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ff6da46c3e3df5f0d9a754576f5a233237aabe68","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-30 11:17:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fc9f54b66e6c1b538a3a6ed36add8ff210d943a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-30 21:22:34.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ad9c25291a384d7640825f8fbea3c12b57c711d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-30 21:22:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"06bedb293021d638c52394a8ea8cb3cc597aca25","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-30 22:04:46.000000000","message":"Patch Set 5: Code-Review+2\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"454eb58e6118f0fcd643d8d38977fff9c5180113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 22:30:49.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0db6df3c617345bc94d6039f673f3b85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a78a1e9a23a4081a0f1a8e437b3c480 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/40a32a47a5ec460cbc1655a4588ecfe0 : SUCCESS in 57m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be9e4add64d042e3af1be7b639dabff5 : SUCCESS in 59m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3bbe332a770442b84e3bd58f6244010 : SUCCESS in 55m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4477d705e7304d3ea54158c595c8b464 : SUCCESS in 1h 04m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f409ab5bd77a43a191f7646401ac96bd : SUCCESS in 12m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75957223c0ea4b2284700d508d619a1d : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff3e9a459b964e20b224311ee459df8f : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf3c16426fff408180726b2d4cbe01e9 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67fda827eaab4263b88e9c6b9d61a2b1 : SUCCESS in 17m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f84e6bb20c47441b867ee5dd51fe015a : SUCCESS in 20m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c17ec179e8794e6ea8c816c7e0121e27 : SUCCESS in 13m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/498ec8f493f34f989b785bcc70a37edc : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2eeefc54b79942f19d404179db2f7fac : SUCCESS in 16m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1b25066b770141acb206c7d5e59027f0 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fa95bead2f248a1b5f73bdbbc236b89 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/001562ef86dc41478f8be9385228c401 : FAILURE in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a96df8d02fc64d86a2a3372170703e51 : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4f19e7208e14fc6856f8e803ed289c1 : SUCCESS in 52m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce6af7439d194a018fbd580d996cf9ec : SUCCESS in 35m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79ae5d56f2af45bc80f747275c888200 : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3b18563835b4dd69faa194c3ae16a0d : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3e92673bfc4327aac6eac976fecd80 : FAILURE in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa59f95b34fb40238aefce53de10329c : SUCCESS in 22m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89c9c1be230e457793dd79752ee5d359 : SUCCESS in 51m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5e07285194f4fff978604f6d944e6d6 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":6},{"id":"554977d551e7c157098e9e7729806425ef4690d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 22:33:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b1cb315ce643d594b737e534c94767\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/600b5857aab2455e9a4e32c4e23e80dc : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":6},{"id":"7f1e8d0599f9289c07bef798491de6d2dc3c3269","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-31 01:16:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f6d86fd0001d720feaf16383468c74c6fb081ad8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-01-31 05:41:18.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"758d1994ae8f49304f6d9f3394022d224378c89e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-31 15:29:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9333f47f6ff12e51cffca473dce0dcd721e93ed7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-31 16:43:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab3f7c26771c8ab2c257381eb182bae0f143e9d4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-31 18:37:55.000000000","message":"Patch Set 6: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1d92b175e3c607b209c0189672d8bff4cfa90ab2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-31 21:19:50.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b451cff95f3b67fc34fc6c3d111746b259e4c160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-04 20:08:21.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b1894fc1c4ddd3762531954450d638133cf4c9b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 21:15:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d24e86010bac43cb9af797691fff1a46\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/196cd753ee4f4147bb52dabe205043bc : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":7},{"id":"a81e35678cd6884a0e64fc1b274b2db0d2a67d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 21:21:17.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd3e607ce1bf4f7bbe7b2f971107db9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f8927154e944e0805938aabaece2ac : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/df80fc3f1e7a4fa48e35f6b8ca314563 : SUCCESS in 1h 03m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4df8ed56a034611aafa3aeeef6cb5ee : SUCCESS in 1h 02m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b99b5904a73d4c71a812ae1aef327550 : SUCCESS in 56m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b017224b7bae40fbb1be5cd26c54d2a0 : SUCCESS in 39m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eef666930cbb489481b9cec17ef5d50b : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c848afa47e284a73800b341c46c79135 : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7237746274da4a59aa593f1583099f59 : SUCCESS in 23m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe5e37d66c304bc687b0b257ae4eeca5 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b46b6b5fcdc4089bd0678adf25ea446 : SUCCESS in 18m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52b96eb2e85c4344bf9d2337b8e414ad : SUCCESS in 20m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cf47b5e6da334bd7a4bf335f754e4e67 : SUCCESS in 12m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/76b10f9c1c3947fea0000c97af0050e8 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/235c3a15b1e941bb852ed8e75cddb401 : SUCCESS in 8m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/266fb0b81b134c2e98d25a91d6d555d7 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de2873966feb4f8db35ef13fcdc71f76 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5311aeba071347d6a9f82634d087c710 : FAILURE in 14m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f060a831b7564fc6ab36e27110b12e85 : SUCCESS in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/834fee01e5ac47788faf36216b999a64 : SUCCESS in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/639886fc66574fe2b5d7caa7c01a8004 : SUCCESS in 51m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cbe9e4f8a85470ab9d86de382348f69 : SUCCESS in 25m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/179757b49c7844c5ba1ce2ec06a2859d : SUCCESS in 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778f74c6df3e4ecea1fbec3a02eb2149 : FAILURE in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebe675d9950e4f0b93c7c58a44563147 : SUCCESS in 10m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca5da7d3e08e436da8569ce02ebaedc6 : SUCCESS in 54m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f041a981f384bf79e86832b92534d32 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":7},{"id":"aa36e5b853a91404611c49b7642194e795e54819","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 21:58:01.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f12665b29460e96c025ce5ba8663688735dc4bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 20:51:39.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a9ad21c69a2874fa85c944b92db4466108a917ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 22:12:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a2ab8b56a1408f994c3ec33c89ed43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5def46b2033449aabd5d0b5cd278bf7 : SUCCESS in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/3345c2921a4c4122a9f3089d37bbefbf : SUCCESS in 54m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94c0748b20d34b288235f5e09cca429a : SUCCESS in 1h 06m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa7698256ba74eaebd4eea7de250713b : SUCCESS in 56m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e0563659bd9404290156939b6ee6d79 : SUCCESS in 58m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd64e72ab3d944cc8e0da6c5e1f004db : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a8053fa204d42e4b0a2c83b93481079 : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4d0f101d45e4ea3ada26abfa515da96 : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19933cfbecaa43f1932b730dc983e5a1 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8df9a8f464074b5d9349048487334cb5 : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36b445ab91dd4c799a689f4a5d4e92d2 : SUCCESS in 22m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/97af244d896140c49be489be0b0b8279 : SUCCESS in 26m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aafbb243a34c48868db08b64f18a1f4e : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d01e41d0093f4176a3c1d577e6142e1a : SUCCESS in 13m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1339d4b2eeb246c29594ecf214b41728 : SUCCESS in 12m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbe31b58603a4bd5bb23cbec35f34e11 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/74e9ac80ebbd47059617fed35543cb0a : FAILURE in 12m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62ee3ebd96354813b10b080e8c913b14 : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed9d969aee4a447e974040b785266f7c : SUCCESS in 53m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da6e2438af3c466ca523d87c79577097 : SUCCESS in 47m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b2d46d8459043889465023b3d5bc5f3 : SUCCESS in 49m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4e6ca517a7d4bee8ef6461435028a34 : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26cd2bb13c88499fa9ba8c1828ee6a59 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/978ae8d6f37242078d041804b67be474 : SUCCESS in 19m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a08b8f39702466c8939cc77e1b6a9c9 : SUCCESS in 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50010876d6024e08af4e949c987847f5 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":8},{"id":"5c59dbdf44318c9a4e6092e9a8b9a5d8d3659b38","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-05 22:51:33.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a60610743d29c1292f0620faad4b7e219057b7cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 23:22:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f9c500361e4447bf07c1b45bfd2cf7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f51f770e23da49c3b81b54323d978206 : SUCCESS in 2h 28m 34s","accounts_in_message":[],"_revision_number":8},{"id":"9772f65ebb96e9ae0bd6c07370464fac9277d65f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-06 02:04:02.000000000","message":"Patch Set 8: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"17608359b4d1a10b301c837bb3e35ea669510479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-06 16:31:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"3489ea96fc9b461fff5ddb0304b9497680d2b743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:47:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8c3ca39eb04b8c88bcc8974db28156\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c83d1fe13545f3bcc2705616ea8fe7 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/4aecdd9bccdb4ec2b3633b30879430f3 : SUCCESS in 1h 01m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/916b070063f3433caf057788d2e369ab : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08a524e920aa460f8266e751eef659c0 : SUCCESS in 50m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f1d601953794483a86bad848decfd99 : SUCCESS in 1h 06m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9e028c605a7414bb50694b1441ce391 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a1a85683a344011bbd4673e29ff77a5 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12e75ea47fe14891a45b15b4f3a3d7eb : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c528572df7514eb5b71b7c2fbfbb29ba : FAILURE in 4m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dee31199e552416a92805e1df601e446 : SUCCESS in 17m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c9389b1808674be0abb82a158d015453 : SUCCESS in 19m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17df4775ef9e4312b432e110b17e7855 : SUCCESS in 24m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb534a1ea6784404a456840d010d1577 : SUCCESS in 9m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c16d88d7f2984382b28b471c90fe505a : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/128671b5785a4194a23a156b5c7fe3f5 : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3ef063fdd7f44f7956b84a8b80aa035 : SUCCESS in 8m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af20012461024ad68cf8223780510a93 : FAILURE in 15m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c353c97585348f6bb097002c7b87634 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe80003a3ca04005977a58b30ac1f220 : SUCCESS in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c9d20913f134ca39071b780734c9f55 : SUCCESS in 42m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6623d9674c724f2da52bd52972cdad62 : SUCCESS in 43m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6140b288de5b4a059c665b564df2f724 : SUCCESS in 23m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bf01f25d4a4e89a9153ccdb21a2fdc : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b32384fc6c245579a52a25ce84cea8e : SUCCESS in 16m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4055ad1991345b5892d742a155466e0 : SUCCESS in 1h 02m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c2c16d346914cae865bdec9f29748d3 : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":9},{"id":"4dba47d94ddcffb9628e1739931de468a4e0b247","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 19:51:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de5be8c53084f4d8f74c546386085a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2beb964a6329486fbe7e5c1ef8bb4c4f : SUCCESS in 1h 39m 26s","accounts_in_message":[],"_revision_number":9},{"id":"80b8543eefbba9413b28eddcfee02c6c21535a83","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-06 22:12:40.000000000","message":"Patch Set 9: Code-Review+2\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"df0e07facdc94d058194be1e5f16056e04820096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-03 18:18:40.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"599a258d7286583137a36ee3576828b4d9d9287d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 19:09:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6342e88374c49638d81b192b21b62c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ecc8206a9e2452ab7517c04e602fc09 : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":10},{"id":"2b799d3048fe9a0da96e437a8eb0dfd3d3c33fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-03 19:22:22.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"d541ac448005ea43ea3c2c4ce6d249aa0c2eb075","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-03-03 20:08:13.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"206b13cd30f72625d7a65210db0a791644cf539f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:16:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0644db0b78b24a319720c0d05518c482\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4d83066fcc64da7b1662dc1d1ca12df : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":11},{"id":"4d10808dd20cb065d060f3a9e403b1409e7e02f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:37:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6110600a1ca4dbdbe8e405b355a32a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52634c6f5c2c439c868e3406e91afa6e : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/5da018de0bbd4ce690e06e5be9905eb7 : SUCCESS in 1h 03m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/587e22ab36fd47e99cfa9336d1c355b9 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ffa48c1ad0f84ba48caa2d0f3a69f61a : SUCCESS in 51m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91f5afd90ad2452fbbe17099fc3c07ff : SUCCESS in 1h 02m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d996618d693a41419578a69c865475ed : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58590da738d3404995cca846ade9e6e4 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e15336fc028947379535bbe80db8047d : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4b45243f8c94369801a40273f92f5c1 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/040e866cab064b0aad13864311d12509 : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dfe153cdff744970986a982b74dad642 : SUCCESS in 22m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b18242239e3c474ba695e74a44868836 : SUCCESS in 23m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4743e9edd2714e81a447fca544402223 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/07ca50808655492da5b8dfc8ae650f6e : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0842907ff6ab437eb48116ea23818e25 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ca050769070415e8fc13c3f4fd73852 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b5bb01ad0f843b0bf89ee2e17dc4a5d : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c8d7cc0f9254520a963994f7729c60f : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16696bd7b4e24de5b6fd2ca598e3303e : SUCCESS in 57m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d40633fbd5a44c39009eb9eccbfbc4d : SUCCESS in 45m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f2ef98222c7417ab732dc566ba097a0 : SUCCESS in 49m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a1e2710d3cb438c8fda366a44460388 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c279398fd64b414ea52e2bdeb4b91e00 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3168a0a24094f958145597e981f1b55 : SUCCESS in 25m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97859ce740c2435aa7b3ecfb0531b843 : SUCCESS in 46m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e99c049f8cb4125946d279c271fdbee : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":11},{"id":"8f7c4258952e7131edcd485445b367b43a7230e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:37:54.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5000c08b6573a282ad49a3aab4a86b979878d85d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 22:10:25.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f5b8f7603ef44fcb39086f5c1947001\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3638d6fcc92445a5b61cf496a1c1d332 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/548a1af4c66c4b368498f18c76df6f68 : SUCCESS in 1h 03m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/319186c86be240748721ee6a710c8349 : SUCCESS in 45m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d54867dcb1b40d9890dfd8c94c7fd8f : SUCCESS in 1h 08m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c44ad542cab8429e9b735ef4bbf2c2e2 : SUCCESS in 1h 15m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f87cae15262b44098975dec3bcc4021f : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/faa630413d7a4c4bbf00a52442961b81 : SUCCESS in 22m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e06f3dc5067e4985bfa796a4cde833d0 : SUCCESS in 20m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a2b4be4a9ec1445b83b2d12aab7ce042 : SUCCESS in 29m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ee2ed1adfbcd4d748b88631990755e98 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8bc0f3297ca649ad8c1ef870434aa1f7 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f9ee175925784b1f8030882022d53815 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb07b37ce8ca4439989d91bf11eb8dfc : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/671c759ed2ed44f7819f607e1d8b214e : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc8be776da8446d999c1800185f7254d : SUCCESS in 54m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46e51c3790104eabac2a0ca66fc40a94 : SUCCESS in 51m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee80449f698c4889ae7a5684da25ed85 : SUCCESS in 26m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf17e806194b4e94ae50018a5a580a3e : SUCCESS in 20m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b22ce06e78476da84e1e7d4786bd82 : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03b095f65d324cd8bc0b617ae3542f6a : SUCCESS in 18m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5eb19748aadd42039d38122ae9dc0cab : SUCCESS in 41m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a7c21054e804d8395bce7e98ecd3f9e : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":11},{"id":"711297bae7265a9c15c147973a30f937b1484e71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 22:10:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"6f8dc1008985e7ba97966638b64527ed2bbe4dab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 22:11:37.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443ed99a39b149e2bf5041ea82a0f064\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83720456da5b415ba6a2f5b67957de2e : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1da352309e3b4b4c9313b67ee8540a12 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0a8ecbc554b811d35713a9486774ca93251d0c9f","revisions":{"fe0d157ccd48975cb74fa5becb4f1109aba56557":{"kind":"REWORK","_number":1,"created":"2025-01-08 01:44:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/1"}}},"commit":{"parents":[{"commit":"95b9e6e335878c8bc97088543a630cb9f67e7262","subject":"Merge \"CI: Add Dalmatian upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95b9e6e335878c8bc97088543a630cb9f67e7262"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"subject":"Fix unicode-header-name handling","message":"Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to re-assess message framing if/when it needed to\n  parse additional headers (in case they included transfer-encoding /\n  content-length)\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3 (though maybe not\nreliably??)\n\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe0d157ccd48975cb74fa5becb4f1109aba56557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe0d157ccd48975cb74fa5becb4f1109aba56557"}]},"branch":"refs/heads/master"},"42ed14f1e130f90d26059fff448add2f2528842a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-10 01:44:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/2"}}},"commit":{"parents":[{"commit":"94d3a5dee8122e14be0b00694d65c448ce47ea31","subject":"Merge \"obj: Add option to tune down etag validation in object-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94d3a5dee8122e14be0b00694d65c448ce47ea31"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-10 01:37:21.000000000","tz":-480},"subject":"Fix unicode-header-name handling","message":"Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to re-assess message framing if/when it needed to\n  parse additional headers (in case they included transfer-encoding /\n  content-length)\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3 (though maybe not\nreliably??)\n\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42ed14f1e130f90d26059fff448add2f2528842a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42ed14f1e130f90d26059fff448add2f2528842a"}]},"branch":"refs/heads/master"},"06644f4e05339247e08f924867e652e9a17c2d21":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-01-10 18:37:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/3"}}},"commit":{"parents":[{"commit":"94d3a5dee8122e14be0b00694d65c448ce47ea31","subject":"Merge \"obj: Add option to tune down etag validation in object-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94d3a5dee8122e14be0b00694d65c448ce47ea31"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-10 17:29:21.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to re-assess message framing if/when it needed to\n  parse additional headers (in case they included transfer-encoding /\n  content-length)\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06644f4e05339247e08f924867e652e9a17c2d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06644f4e05339247e08f924867e652e9a17c2d21"}]},"branch":"refs/heads/master"},"29bc7664204eca797affb58f248306b1028977ba":{"kind":"REWORK","_number":4,"created":"2025-01-13 23:10:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/4"}}},"commit":{"parents":[{"commit":"128124cdd8ca09136d4988affd1bb8c5c1361fc1","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128124cdd8ca09136d4988affd1bb8c5c1361fc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-13 23:03:04.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to re-assess message framing if/when it needed to\n  parse additional headers (in case they included transfer-encoding /\n  content-length)\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29bc7664204eca797affb58f248306b1028977ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29bc7664204eca797affb58f248306b1028977ba"}]},"branch":"refs/heads/master"},"0f6ba89b5336f4d9adfbc4b3e2f5f9c4ed8f7482":{"kind":"REWORK","_number":5,"created":"2025-01-28 17:03:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/5"}}},"commit":{"parents":[{"commit":"128124cdd8ca09136d4988affd1bb8c5c1361fc1","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128124cdd8ca09136d4988affd1bb8c5c1361fc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-28 17:02:43.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to re-assess message framing if/when it needed to\n  parse additional headers (in case they included transfer-encoding /\n  content-length)\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f6ba89b5336f4d9adfbc4b3e2f5f9c4ed8f7482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f6ba89b5336f4d9adfbc4b3e2f5f9c4ed8f7482"}]},"branch":"refs/heads/master"},"047cd2a570c65c0b2761cfc329f20db5493a15b8":{"kind":"REWORK","_number":6,"created":"2025-01-30 21:22:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/6"}}},"commit":{"parents":[{"commit":"128124cdd8ca09136d4988affd1bb8c5c1361fc1","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128124cdd8ca09136d4988affd1bb8c5c1361fc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-30 21:21:47.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to parse additional headers earlier, before stdlib\n  tries to establish message framing\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/047cd2a570c65c0b2761cfc329f20db5493a15b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/047cd2a570c65c0b2761cfc329f20db5493a15b8"}]},"branch":"refs/heads/master"},"b7f8263d1db2fbdf96178f6d5f76cd05c6f37b3c":{"kind":"REWORK","_number":7,"created":"2025-02-04 20:08:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/7"}}},"commit":{"parents":[{"commit":"128124cdd8ca09136d4988affd1bb8c5c1361fc1","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128124cdd8ca09136d4988affd1bb8c5c1361fc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-04 20:08:08.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to parse additional headers earlier, before stdlib\n  tries to establish message framing\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7f8263d1db2fbdf96178f6d5f76cd05c6f37b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7f8263d1db2fbdf96178f6d5f76cd05c6f37b3c"}]},"branch":"refs/heads/master"},"d125634b3ee8795b3ad9a2c017c60753d94ec08c":{"kind":"REWORK","_number":8,"created":"2025-02-05 20:51:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/8"}}},"commit":{"parents":[{"commit":"128124cdd8ca09136d4988affd1bb8c5c1361fc1","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128124cdd8ca09136d4988affd1bb8c5c1361fc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-05 20:50:36.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to parse additional headers earlier, before stdlib\n  tries to establish message framing\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nCloses-Bug: #2097030\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d125634b3ee8795b3ad9a2c017c60753d94ec08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d125634b3ee8795b3ad9a2c017c60753d94ec08c"}]},"branch":"refs/heads/master"},"8261712d603084f92f6d460743566580d58192fc":{"kind":"REWORK","_number":9,"created":"2025-02-06 16:31:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/9"}}},"commit":{"parents":[{"commit":"128124cdd8ca09136d4988affd1bb8c5c1361fc1","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/128124cdd8ca09136d4988affd1bb8c5c1361fc1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-06 16:31:48.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to parse additional headers earlier, before stdlib\n  tries to establish message framing\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nCloses-Bug: #2097030\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8261712d603084f92f6d460743566580d58192fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8261712d603084f92f6d460743566580d58192fc"}]},"branch":"refs/heads/master"},"6e7b8e975834d012064cd82c6702d375b7c7f0a4":{"kind":"REWORK","_number":10,"created":"2025-03-03 18:18:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/10"}}},"commit":{"parents":[{"commit":"e4cc228ed09eb30a4e186515f801d9cde0140c4d","subject":"Refactor some file-like iters as utils.InputProxy subclasses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4cc228ed09eb30a4e186515f801d9cde0140c4d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-03 18:18:27.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling","message":"py3: Fix unicode-header-name handling\n\nThere were two fixes needed:\n\n- bufferedhttp needed to parse additional headers earlier, before stdlib\n  tries to establish message framing\n- internal_client needed to encode to bytes before .lower()ing metadata\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nCloses-Bug: #2097030\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e7b8e975834d012064cd82c6702d375b7c7f0a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e7b8e975834d012064cd82c6702d375b7c7f0a4"}]},"branch":"refs/heads/master"},"0a8ecbc554b811d35713a9486774ca93251d0c9f":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-03-03 19:22:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/938631/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/938631/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/938631/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/938631/11"}}},"commit":{"parents":[{"commit":"e4cc228ed09eb30a4e186515f801d9cde0140c4d","subject":"Refactor some file-like iters as utils.InputProxy subclasses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4cc228ed09eb30a4e186515f801d9cde0140c4d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-08 01:39:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-03 18:41:34.000000000","tz":-480},"subject":"py3: Fix unicode-header-name handling in bufferedhttp","message":"py3: Fix unicode-header-name handling in bufferedhttp\n\nWe need to parse additional headers earlier, before stdlib tries to\nestablish message framing.\n\nNow, TestReconstructorRebuildUTF8 can pass on py3.\n\nCloses-Bug: #2097030\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/662546\nChange-Id: I6aa16fda9285c9fc3816da6fbff2615bd14a020c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a8ecbc554b811d35713a9486774ca93251d0c9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
