)]}'
{"id":"openstack%2Fswift~938966","triplet_id":"openstack%2Fswift~master~I52f40bf0468236dc7ba78e08f2f9aea9336faae9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-19 09:56:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-09 23:36:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I52f40bf0468236dc7ba78e08f2f9aea9336faae9","subject":"tests: Use subTest","status":"MERGED","created":"2025-01-10 22:12:52.000000000","updated":"2025-05-15 07:42:12.000000000","submitted":"2025-05-15 07:40:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"938966","meta_rev_id":"31a077d12a58bbc308e41bdcc3d480b20a3651ad","_number":938966,"virtual_id_number":938966,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 07:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-02-19 09:56:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-09 23:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 23:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 11:09:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-02-19 09:56:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"3d62050c4fb36a3bb82fdf42a383e356f40819d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-10 22:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d45ca8d1199aa14a1d63899f745f2c3e85ce261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 23:35:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a920ee75787744a4b9cdcd080a619f90\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59b26ef50928451fbd0732ca1730e450 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf8b60623ab49aa9de590da068aeb20 : SUCCESS in 5m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a981f71b30514203ab32a6f84cfb2f94 : FAILURE in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df6af1fb605241bb8fff1bb469c25b4c : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/4d67476c450b40a5869f640af8b87e3e : SUCCESS in 1h 01m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd135a2b290c4edf97964796493d4f79 : SUCCESS in 53m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf085c1fd34946aebbf0f631cae20ea9 : SUCCESS in 51m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c272012cf2748fa921fdc6fb66d2db5 : SUCCESS in 1h 06m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a68b4f4bc3154dc6a0f75555d09ae996 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6defb8a926c845d6a6e024628635bdc6 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31152e6dcdb44864acd87ddd68a76828 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b79695745dc42b78ee91eee9474fd98 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07b95980a5754db6bb300e3017849da8 : SUCCESS in 19m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3cdb2b4396445fbb5fd916dfa90759b : SUCCESS in 19m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3f10022b44c14870aa825e9375549094 : SUCCESS in 28m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/03fd00ca113f4a9a90f9d8b465fc582f : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fc7b709bcc3b47098ab62fcfc08ad928 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2f0627be780e4323ad922a7a086c10d5 : SUCCESS in 13m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f01c3fddb4194bd1ad8111e2359541f3 : SUCCESS in 5m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab363540e4e844a694d477e80ebc52e0 : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e876ec2a5e3c4d60a9712259c816255c : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/77d4b307bd454979918f6bc822ff93ee : SUCCESS in 48m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a09ac3252ac4f9ba5f73c47060dc7b7 : SUCCESS in 28m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93abb956ae7b4f5d86104c3f07339755 : SUCCESS in 48m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d68d1adfaa84e8da4a3717a5f7e194d : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c68add6b28456d987c6a766ebc345e : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ce916cba6ab46d1a84c8d7712e18661 : SUCCESS in 15m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9491b62f9efe48d79700b2ab37755eca : SUCCESS in 35m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d8093efc7c444c88cc618a46069338 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b9405b29dd52ce5ece61316cb920a09d4bc44d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 23:39:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad9f52dd44648c9aae881d5223dfe4f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13ee45426da84549929f96fcae3ef0bb : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c63b4813ae0e67539c00949d12ca9f4dbe9c85b4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-14 11:09:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3c4827597bcbe822585e071de179079254375da7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 18:41:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c883a0409ffe02cfb663b5e8642de2016bb635f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 18:41:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01cac1ebde724d1daf57a4e05405c294\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1433be9d4e274b84b3e0ec0a498e4927 : NODE_FAILURE Node(set) request 300-0026341732 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"577a0b2fe7e1f9e1f4c624b97b87419d329afc27","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 18:42:29.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e99bd410958d948fd9665456d4f3fe0ccf4268b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 20:03:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aa40afcc4b64100a618dd6d05bebaad\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30d6867b9b844a3386f7ecab8c1f866b : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b4061ab71a423886ba4aff23d3f472 : SUCCESS in 8m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3422f9ad194e4af4984fdbe3442f4f46 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60c315130cb34738a18aa26e5641d4ea : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/e50d27f116c4468e823af9c9465b8f44 : SUCCESS in 1h 00m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/97f636aab4c0417581faa595f15ae132 : SUCCESS in 55m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/13acf3073a674f0db2f4fb753bac07c8 : SUCCESS in 45m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/543b268fe87244959850c65288902da0 : SUCCESS in 53m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e0b5d18856848c186ae4b43eea55229 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fda358f94af4b37be9d40c74979f8f8 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e0f6cae8d934532a69db857c186e135 : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12e287f67cb246b1b15a42dbf6063c1a : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3cb0c4e4b194e118264044d422eb2ac : SUCCESS in 18m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/931a1f6d4e714b46ac815354a5218eed : SUCCESS in 17m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6cc8bf7ec5884a08a9842f467aa466b1 : SUCCESS in 26m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4d590b8a0b28469d9c56cbc9c186f956 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4f20bd97ed5d4ae1a0e9697ba84f49eb : SUCCESS in 9m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e822f42e26a4794b946ee745ddd4737 : SUCCESS in 13m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3acc96e225f14989a5e8c3b440c53f33 : SUCCESS in 9m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/feaa107d94f145ee86edd2bf24eb2cfc : FAILURE in 21m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c54d2d7297f42969566765939cc9f31 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e58f96e462e472091e1124cc5bfa682 : SUCCESS in 1h 09m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91d50fb4a7b841f7b6cbc68dba266900 : SUCCESS in 47m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c378f8efc7d146a38abab03ac2364c5f : SUCCESS in 48m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301ddea7d19a4177bb1f8f6424ada1f8 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501d2c258cfe465b92b3d808b7ffbd63 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe93a0201810478abd7af85214eb8a65 : SUCCESS in 17m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44a9f3c2edd34ac4b2fee145897dbdbb : SUCCESS in 51m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/060bd3ff7a884fcdb7d47c90b872aa98 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7e63603a0670e093ad9cfb683083d81a00d2e36f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-19 09:56:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46a1f8eded61299fef4fca2ccbd53323f3629859","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-09 23:36:57.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0390c4b061b01cb80eb855714e37fe1d37c83e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 05:49:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9c1bd1fb66dd3a0732923e966baffd4b631712a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 07:40:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b2df86fb0de421eb26be8ad967bdab8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f95c8e821c8847908675b52815329758 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a722737721447f2b5b1fb1e2627faf9 : SUCCESS in 10m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d232512e72324a10806ce922b3477cd5 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/021772b7a74b4f098dad5ba0c58bd2ee : SUCCESS in 2m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/19a98d81feba47f89d6878f869349cdd : SUCCESS in 48m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/419148430e924070891a8edb7fb0d92f : SUCCESS in 33m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6380698613414322906152c9893df3a0 : SUCCESS in 43m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51d2c90daa2f4a848087acda9e4752e8 : SUCCESS in 53m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/beaedd18ce274f958d2fb9933ca24220 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f0efb329b9744663a609ce6b4c960620 : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fec721636bd34fe1907ebe3fb5d98ddb : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/73b92b3dd90c4fed830401690ec1f591 : SUCCESS in 10m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c5bd85bacd643e8ab61b005a148c7ef : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/26cab9ab3f7b489994859f6552008c05 : SUCCESS in 8m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/830b1fdb129c4906ba67893df01bf81a : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de8749dd1b7b41c9b7229cd9ded77f1a : SUCCESS in 6m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ce14afb09884f0b88faf7461110d787 : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ecceb081ae04764ad17c7f725c62a39 : SUCCESS in 33m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d2bac71581b487a8e9e81dbda9e09d0 : SUCCESS in 42m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b509c3c59ebc4be698309348bae1c31c : SUCCESS in 22m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fb7d829d6e6404494539fc45330e4c1 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786aee1343aa4ffd8d4828c3a0d7b7f5 : SUCCESS in 2m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35751839709540e9a844035a47af2bd0 : SUCCESS in 18m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbd7ea48357c42c38f73ff4c88541613 : SUCCESS in 38m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/350f48f1b570490cadf6e825a6e0ca82 : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c02193f6609c19c86fb216d32c0588188b692e93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 07:40:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"31a077d12a58bbc308e41bdcc3d480b20a3651ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 07:42:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/610437184606400383371de42bafef10\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/56e585c3208e42beb92cb04e7314d47d : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2e7c0e1f4a4ef2939f10c8724f27a5 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0336ca12abb84e50a907b49117c76b97 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cd6564005ea8413185ca32d5a14aabbd : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff455194477e502f1ddc1652209dfba8bacac698","revisions":{"a31f2c5488ba24af18a84c89648568b5ff8a6726":{"kind":"REWORK","_number":1,"created":"2025-01-10 22:12:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/938966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/938966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/938966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/938966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/938966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/938966/1"}}},"commit":{"parents":[{"commit":"31ccf24770b36663e7fec302ade97231c35f5d1e","subject":"Remove py2-only code paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31ccf24770b36663e7fec302ade97231c35f5d1e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-10 22:08:35.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-10 22:12:44.000000000","tz":-480},"subject":"tests: Use subTest","message":"tests: Use subTest\n\nChange-Id: I52f40bf0468236dc7ba78e08f2f9aea9336faae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a31f2c5488ba24af18a84c89648568b5ff8a6726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a31f2c5488ba24af18a84c89648568b5ff8a6726"}]},"branch":"refs/heads/master"},"ff455194477e502f1ddc1652209dfba8bacac698":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-02-14 18:41:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/938966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/938966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/938966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/938966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/938966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/938966/2"}}},"commit":{"parents":[{"commit":"fc88b0e81644da2ac368526a9a5a25b0ee590fe4","subject":"Merge \"Add NotificationServer to fix swift-reload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc88b0e81644da2ac368526a9a5a25b0ee590fe4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-10 22:08:35.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 18:41:21.000000000","tz":-480},"subject":"tests: Use subTest","message":"tests: Use subTest\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/941708\nChange-Id: I52f40bf0468236dc7ba78e08f2f9aea9336faae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff455194477e502f1ddc1652209dfba8bacac698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff455194477e502f1ddc1652209dfba8bacac698"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
