)]}'
{"id":"openstack%2Fswift~939184","triplet_id":"openstack%2Fswift~master~Ie668d49a56c0a6542d28128656cfd44f7c089ec4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-14 12:40:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie668d49a56c0a6542d28128656cfd44f7c089ec4","subject":"CI: Remove B320 and B410 bandit skips","status":"MERGED","created":"2025-01-13 23:20:36.000000000","updated":"2025-01-14 12:42:57.000000000","submitted":"2025-01-14 12:40:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939184","meta_rev_id":"f04c98f6f8305ad8c183c12f20245759e5a532dd","_number":939184,"virtual_id_number":939184,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 12:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-01-14 09:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-01-14 09:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 00:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 09:28:37.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"7f28d312df09a02a37a49a27921e84595ca65eeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-13 23:20:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4d9c92722b35cbb986176a6dc3f96bd12298a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 00:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02cef02333964a09939616e25c648b17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48cd01e79254e789e8be20d47c233d3 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/85eb5fdafb1b4ecf85e25a4950fdb3d0 : SUCCESS in 58m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/126292e13c5b402e985092e62a9dbda5 : SUCCESS in 52m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9823d3089b2d4ceb93da31fad104e789 : SUCCESS in 40m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ead17bb5b224193a16c3938098a19b5 : SUCCESS in 48m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42f32bf400644527a15c99083fea93a1 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63ee562f64ba46e6a0116cbec8494796 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d826ccabd704b648e12a696cf38f1b9 : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f230410fbdc1406287d0e765f24297fa : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9911b86f923d40fe946b4b179098fd05 : SUCCESS in 16m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce21756e5ae74cefaee3196c36cbb9fe : SUCCESS in 19m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0929c81b03b1476cbd58606197c54423 : SUCCESS in 35m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e814e4d2f60640e0a7a6840e76d88bdf : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e050cf92e8e2496a945d01a42dc469ee : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/56b9cb78b9514f19bc3a03329ed6856d : SUCCESS in 14m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37f7a1f7d8314e9aa10f1ea7f4fc1990 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c1087d8b0eb43d5b0b739bdcd305c05 : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e682b81678ae40ce8f86443fc992f5fd : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d408591d29e4e9a8982eba7f6a27053 : SUCCESS in 50m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbb0b6b173cc4755ab65e8551053c21d : SUCCESS in 34m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d74addfbc6754b26a481b4513251b8ac : SUCCESS in 48m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/631148ab5fe64afdb2c0b2a28ee03df3 : SUCCESS in 21m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60596cb08f8d42dd86c80cdca86a9d22 : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7db67f16000b4d0196ad2d785391f6fa : SUCCESS in 18m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d955315ce3a54236bcecaad7107471dd : SUCCESS in 55m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/732818d22347406db78223a13097b10b : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d8afea3f7f7c4eb263dbaba5af4ca3610995f6ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 00:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32eaf25828b24be58e3a9d8594a0d1e2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4dbc01acf50340cf833e0bf03d4f6942 : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ab1608eb737ddbd8ef52e657298ee74d6f1b567d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-14 09:28:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c97166c165f2499e637c69db279ed2fac0db831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 09:29:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4edc782cdf25f218059412584077b964159ab387","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-14 09:35:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"629e7643ed81ee2f084435bb77b7a421b845b0be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:40:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee19ada9a6e4cec93079ccdfb46c2a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0e9360ebd74760bc4969c5615e6de7 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/b5531c4eb2884725a5600f2c0b1d4cbc : SUCCESS in 56m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16d10205d6bc45b1980e74a54ea2f033 : SUCCESS in 58m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65cd2c4f32d443dda72f295d1580d8c3 : SUCCESS in 57m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f89fd54d445431789a534332bd86597 : SUCCESS in 40m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7c6f5ac489104487b240a584e56bb037 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4faba98fed314a7288f6dd2214e00aab : SUCCESS in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dba66699cd2e47cdb2f6d4b068b18948 : SUCCESS in 23m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9c5475e068049729cd2df4224cb6b35 : SUCCESS in 12m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc5ff8ef73a440e58c192cebb25068ac : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44bd19aeb99b46e38c2be7317ee2caf9 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14cda313199c44179c065f6468bb933d : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce32f4d3ee99458ca58267cdb0c119f5 : SUCCESS in 8m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef29580685d24ac7ae28cf024333577a : SUCCESS in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b093d01718434d9c8c9078b7a24809f8 : SUCCESS in 51m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f0ce381291440ebb8aee16cf4061e32 : SUCCESS in 45m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/499d775a6b224753bd9bcb4bb0178c2f : SUCCESS in 52m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96e90c21ac0e4346955d29efb3d82364 : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fd401a013341fb8e96300152946866 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56a7e052ea484c4b9f72e6f2db9755a1 : SUCCESS in 20m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1769989de4ca4d84a9f52eba5360fdd5 : SUCCESS in 45m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a2253eda60045c9a0d3517e712c50d3 : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2d287869e1cee3bf06774296446dd4991e8605e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:40:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f04c98f6f8305ad8c183c12f20245759e5a532dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8112a211b0bd4291979461cb7b93fac2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40577e983f78418996a2c34f72b9796a : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/30e9e1d2b280442ca2828f5264cbcb82 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f95315b7111746fc54e4ab17c96ee1c1909ee256","revisions":{"f95315b7111746fc54e4ab17c96ee1c1909ee256":{"kind":"REWORK","_number":1,"created":"2025-01-13 23:20:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/939184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/939184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/939184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/939184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/939184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/939184/1"}}},"commit":{"parents":[{"commit":"94d3a5dee8122e14be0b00694d65c448ce47ea31","subject":"Merge \"obj: Add option to tune down etag validation in object-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94d3a5dee8122e14be0b00694d65c448ce47ea31"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-13 23:20:13.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-13 23:20:16.000000000","tz":-480},"subject":"CI: Remove B320 and B410 bandit skips","message":"CI: Remove B320 and B410 bandit skips\n\nThey were removed upstream recently, so now Bandit is complaining about\nthe unknown test.\n\nSee https://github.com/PyCQA/bandit/pull/1212\n\nChange-Id: Ie668d49a56c0a6542d28128656cfd44f7c089ec4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f95315b7111746fc54e4ab17c96ee1c1909ee256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f95315b7111746fc54e4ab17c96ee1c1909ee256"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
