)]}'
{"id":"openstack%2Fswift~939209","triplet_id":"openstack%2Fswift~master~I6eb106abbac28ffbb9a3f64e8aa60218cbe75682","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-01-15 22:04:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6eb106abbac28ffbb9a3f64e8aa60218cbe75682","subject":"CI: document bandit tests by reference","status":"MERGED","created":"2025-01-14 09:35:04.000000000","updated":"2025-01-15 22:06:11.000000000","submitted":"2025-01-15 22:04:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939209","meta_rev_id":"2c642974d949446e6a2910f9229816c44940f0a0","_number":939209,"virtual_id_number":939209,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-15 22:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-15 20:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-15 20:48:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 10:42:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 20:48:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"2586a3fa69765b209bf4b68171cdd7268e80d72e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-14 09:35:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95f87458f3a768b44f8baf4db5becf86ba101e32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 10:42:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e749cdb3a514538921015c55ce81736\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158dcc5f24e242a78e15a9f22c9b849c : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/26ccb5e12764421a96974b194724838b : SUCCESS in 1h 03m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/623bea77b6e7462a93f5e934306d9678 : SUCCESS in 49m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0411cbc1e5984715ae7375d23f41bf61 : SUCCESS in 54m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/841727df15d9447b869d2c3e983bfbdd : SUCCESS in 1h 05m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1a747190a5e4097abf21afb496eb5c6 : SUCCESS in 30m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e8228d48211454499db46761c62bd3e : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/297924a994534590927058f789ef8855 : SUCCESS in 32m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/202975e838e4409980797c16c6536256 : SUCCESS in 4m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af2285538ddb4c7182d95d8ac26932d8 : SUCCESS in 18m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9438b5d07a7c49329c6ae83f367a20d3 : SUCCESS in 18m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c1bfa80a6d8444ebfe3693fc8bb64b9 : SUCCESS in 26m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9c0162b057ec4d08ac37b44ea7f239db : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/feaa14a5bb83425db6dc3282788317c7 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/15a61ec6fcde4bd8a8959936e04b8864 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b371bb844f1d435a95ccd617aae26a16 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea088d80df96487bb4199bc6ccbb7073 : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2fd1049d983a4a6a979fbb37589e5ad1 : SUCCESS in 7m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d663383e24b842829e3d6b6f429b5416 : SUCCESS in 57m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b0d7b70c3c24ec6abb8990e969b69b7 : SUCCESS in 36m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea83521d98854a1cb2d1f59785a9f889 : SUCCESS in 47m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c26271346d8a4307bed461267980bf59 : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116ba7c716d84c16938df26a6278ff49 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a67b763604734b72b2cc294ff6dd36d6 : SUCCESS in 20m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dcae4b7fd9842bd9c0afc8b918d3fdf : SUCCESS in 45m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7e877006984cddb4ff696776bea481 : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7775ee61bf213b502fb7663212df134005b456b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 11:30:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1b2b0b38bc43d3805e10b0785a1a60\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d50a0f8906a4d83a00b4c577ddf982e : SUCCESS in 1h 52m 39s","accounts_in_message":[],"_revision_number":1},{"id":"90116b4cca0ff02dc19a835617549f1ba2ec8207","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-15 20:48:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66097b919cc0d525994a62e068725c7ae7bf9a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 20:49:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54966948a805595b95ff57b5c376135aefc27f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 22:04:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ced761bc6454833be708b5e235584dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8958f95b920b47a68b93d4fbf5845f30 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/79678fc7d5a54b44afb7677b8c074ba8 : SUCCESS in 57m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c707f05a2b7140dda9d0f51f894ffcc5 : SUCCESS in 1h 00m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24ba241afc114752b441071ff91c27ca : SUCCESS in 56m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/affb6ff3fb684f06b901d12f76e12662 : SUCCESS in 1h 07m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/baba6b44559747ab8a2444526a620442 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fb7828fce4d45a78c8583519d57fd8e : SUCCESS in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/999aad5bd3ae48f086be893a2022e80d : SUCCESS in 20m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9029a6b9baa24cc9a87ff7dbb6703295 : SUCCESS in 28m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4a35f06f1f454dfea57bca76fc7a624f : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/567c2eeb7db04865b6a486de8aa979a2 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aedfd85af27b4a4f9e427fccc7fbc3e0 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec6aec4b5cfc4c6f82daf60067b254c0 : SUCCESS in 7m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb79d1f018174983aa5c6b3762dae538 : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/acecccc4bd2b400ba741f8f6f5a23b8e : SUCCESS in 1h 06m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b355c5c304c44f94bd05c787aad5b337 : SUCCESS in 56m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65aab6c9e39946d0aed2abcccaa5f9c5 : SUCCESS in 52m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f0100a668a24aa6b43532f51304237e : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee0493bfdaa44deb0cf577000e4ea47 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/deb4ccd595ad4098b00a7621976c7877 : SUCCESS in 19m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b40c020bd8d74b88af26dcf7b6f86fa5 : SUCCESS in 47m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dc1139a2f214cd5b3fe0bbce4914d72 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"56d04648cb64f9e46bca7af02911f75aca7f7038","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 22:04:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c642974d949446e6a2910f9229816c44940f0a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 22:06:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d338c5ed4ea4a24a0963637eef9f834\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae30f3521bf48dd8253f33235b876b8 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/75b4e1419e23427489f9f611aecc6a06 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa0429ce0038a40213e5b75ac0c7abfcc1c22007","revisions":{"aa0429ce0038a40213e5b75ac0c7abfcc1c22007":{"kind":"REWORK","_number":1,"created":"2025-01-14 09:35:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/09/939209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/939209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/939209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/939209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/939209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/939209/1"}}},"commit":{"parents":[{"commit":"f95315b7111746fc54e4ab17c96ee1c1909ee256","subject":"CI: Remove B320 and B410 bandit skips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f95315b7111746fc54e4ab17c96ee1c1909ee256"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-14 09:31:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-14 09:34:46.000000000","tz":0},"subject":"CI: document bandit tests by reference","message":"CI: document bandit tests by reference\n\nThe available bandit tests change with time (e.g. the\nRelated-Change). We shouldn\u0027t try to maintain the list.\n\nRelated-Change: Ie668d49a56c0a6542d28128656cfd44f7c089ec4\nChange-Id: I6eb106abbac28ffbb9a3f64e8aa60218cbe75682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa0429ce0038a40213e5b75ac0c7abfcc1c22007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa0429ce0038a40213e5b75ac0c7abfcc1c22007"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
