)]}'
{"id":"openstack%2Fswift~939408","triplet_id":"openstack%2Fswift~master~I924c0a1572b009b0a8f62fd2ab69af104dfae90a","project":"openstack/swift","branch":"master","topic":"per_policy_request_node_count","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-16 01:59:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I924c0a1572b009b0a8f62fd2ab69af104dfae90a","subject":"sq? moar tests for per-policy request_node_count","status":"NEW","created":"2025-01-16 00:51:02.000000000","updated":"2025-01-16 02:13:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86023e1ca2402c42f64e33faef834e0f8edc070f","_number":939408,"virtual_id_number":939408,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-16 01:59:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-16 01:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d669852290104ce3e070537e8543a0a869ea50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 00:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb118624e7d7a9c9bdb6ebbe4391b9f780a6651a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 01:59:12.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6607e550b2bd472e830f4724ae80b05a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49c7d94269941ca9bcb2accd9287e8d : SUCCESS in 8m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0eca33f347a41b6b10edd24465981f0 : SUCCESS in 1h 01m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50018a318ac242a3b79951a6f4063571 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc4f0443bccc4ff893a61c7009640d21 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a56fd961c0f4d56a9afcaf259ff7da5 : SUCCESS in 12m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f1e8c8ab3af4e5cb84c3f6c9059b1bc : SUCCESS in 4m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db517e0acfb640159c055e20232d26c5 : SUCCESS in 9m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98a33ecaa1d64ed58420bb03ebe76e37 : FAILURE in 10m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6ef57e1a6ed243d2964e66c186130f63 : SUCCESS in 18m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9c6528014b124b38af8a257ae6c80516 : SUCCESS in 8m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/faf7c238dcf5410e9be90d8a1f70cb8d : SUCCESS in 13m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c207cf3e5c74823aff9b83f1efd837a : SUCCESS in 12m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4527e15a045f4ab28fca6133c96f9c0b : SUCCESS in 26m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d7961e9a21a40889d1fd522f62aa751 : SUCCESS in 29m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cea404703ceb4debbf0dd59e2c0eb12f : FAILURE in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0d1dc38374346fd80bd9d21c8a48366 : FAILURE in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e34c8f6f0c71410a83bb2a27c19e0dbd : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"86023e1ca2402c42f64e33faef834e0f8edc070f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 02:13:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11bef7e7c3044e4c87847357fb65ef5d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1129fda63a96453dbc308bf02445a6ee : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebc263895fc0214762174cd9b649ada64e5297d5","revisions":{"ebc263895fc0214762174cd9b649ada64e5297d5":{"kind":"REWORK","_number":1,"created":"2025-01-16 00:51:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/939408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/939408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/939408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/939408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/939408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/939408/1"}}},"commit":{"parents":[{"commit":"6d94c098f9f50812f00c686ee9820c96c67c58c8","subject":"Allow request_node_count to be overridden per policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d94c098f9f50812f00c686ee9820c96c67c58c8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 00:49:46.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 00:49:46.000000000","tz":-360},"subject":"sq? moar tests for per-policy request_node_count","message":"sq? moar tests for per-policy request_node_count\n\nChange-Id: I924c0a1572b009b0a8f62fd2ab69af104dfae90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebc263895fc0214762174cd9b649ada64e5297d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebc263895fc0214762174cd9b649ada64e5297d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
