)]}'
{"id":"openstack%2Fswift~939434","triplet_id":"openstack%2Fswift~feature%2Fmpu~I2d72a75e6c3aa840ab43f84cf90464d1118e72da","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-timestamp-http-date","hashtags":[],"change_id":"I2d72a75e6c3aa840ab43f84cf90464d1118e72da","subject":"mpu: fix Timestamp.from_http_date","status":"NEW","created":"2025-01-16 11:34:39.000000000","updated":"2025-01-16 13:47:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1207ee625f94615c6cf62749b1a3c8887b6f83c2","_number":939434,"virtual_id_number":939434,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-16 12:34:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-16 12:34:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c9cf661e6e9b8a122470948649b7b4ba3e3e4b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 11:34:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed2bcefa7efd473b3292f40766329f9b2b0bc85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 12:34:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb0c4dd882a4bf5b2754048fb5ad039\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbd275580fb49019a9bd3f73e27659e : SUCCESS in 7m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a79c0a23271f42d78f307152c94cf14f : SUCCESS in 23m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0f8a6eb322742548e29cd914cc31b89 : SUCCESS in 25m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faa615fd92be49d7a9dd1a1ea66cc2d2 : SUCCESS in 12m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68205c7c80f2404282f820157dbf5d24 : SUCCESS in 4m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0ac36d6cfb4451abb9d901c80d8bac : SUCCESS in 15m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36283953071245e5a240a3877ee7de79 : SUCCESS in 10m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c2a4c53c1f54cd5b9fef403a7f1e006 : SUCCESS in 26m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/77650a12a74e495ca8e074280351b7b7 : SUCCESS in 10m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/accf2d60047c4cf0a06aa4d3e028014a : SUCCESS in 16m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b189ee0021c4e1883365a26df9fe1c6 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc2477ff7e534ad4ad8802e4a6013868 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6919f7ac9b4240d08cb05e24b2c0b407 : FAILURE in 16m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a882836f7d3e468d9133d38e4b96562d : SUCCESS in 8m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/933b15ee4aa64b10b458ced605955151 : SUCCESS in 54m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0d1839a4e364d7baac7170d698900c7 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210043ef9a0d4e09abb4ee94816575c4 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e535f3c6739f4318a16fe259dcc53a6c : SUCCESS in 25m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1207ee625f94615c6cf62749b1a3c8887b6f83c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:47:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09c4c4b9c6e4a978b9f124c47650352\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9893267407249f986cf4cc5ff1875eb : SUCCESS in 2h 10m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96e9eab84e3d611b810a855f94fb13e30c311cb9","revisions":{"96e9eab84e3d611b810a855f94fb13e30c311cb9":{"kind":"REWORK","_number":1,"created":"2025-01-16 11:34:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/939434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/939434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/939434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/939434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/939434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/939434/1"}}},"commit":{"parents":[{"commit":"2fd857a8e1ef297ed52efca5ad4b987f80495229","subject":"merge master to feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fd857a8e1ef297ed52efca5ad4b987f80495229"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 11:18:52.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 11:30:19.000000000","tz":0},"subject":"mpu: fix Timestamp.from_http_date","message":"mpu: fix Timestamp.from_http_date\n\nFix handling of missing timezone so that it defaults to UTC as per\nRFC2616.\n\nChange-Id: I2d72a75e6c3aa840ab43f84cf90464d1118e72da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e9eab84e3d611b810a855f94fb13e30c311cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e9eab84e3d611b810a855f94fb13e30c311cb9"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
