)]}'
{"id":"openstack%2Fswift~939471","triplet_id":"openstack%2Fswift~master~Ibf55eab0f6f7f5ec67e18c586779770f78680dbf","project":"openstack/swift","branch":"master","topic":"fix-rebase-patch4","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-12-11 21:16:16.000000000","reason":"Change was marked work in progress"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-12-11 21:16:16.000000000","reason":"Change was marked work in progress"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-03 05:23:00.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-03 05:23:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf55eab0f6f7f5ec67e18c586779770f78680dbf","subject":"Limit node_iter on EC HEAD","status":"MERGED","created":"2025-01-16 17:06:36.000000000","updated":"2026-09-04 17:39:11.000000000","submitted":"2026-09-03 05:23:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"939471-fix-rebase-patch4","meta_rev_id":"40933680b982310f0bd01cdcc81203e22c911f64","_number":939471,"virtual_id_number":939471,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 05:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-09-03 04:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2026-09-03 00:09:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-09-03 04:14:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-16 18:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-17 00:20:33.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-02-03 05:40:37.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2025-12-12 15:41:10.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-12-12 15:41:24.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-09-03 04:14:01.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"e992617606af4cd182bf5cce6799b3030ff7ece1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 17:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c909220b2c6312728b78453487f936972239aeb3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 17:07:09.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bfce691ae0e6434ea62cf96157f3a6aec4cc9c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 18:39:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c8368a9232b426ca44ed13d42940a03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7561d8f36b54fd897c58f1ab865765c : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/b3b472e11ddb48ed88af6ce78b4b0ce3 : SUCCESS in 1h 10m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b12e0d1c62b04ee99c5ead9d5985debf : SUCCESS in 56m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea85983ccaaf4f3daa389a78d77064e7 : SUCCESS in 45m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e44cf5e213941d3a08a26611d535373 : SUCCESS in 1h 16m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/391c32dfc1b547eb8bad4e62e64cdb4d : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a633c64b1444a13b5e6ce2a04ee3a2a : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d1c9178b0ca44a3b689a1daa88fc2f2 : SUCCESS in 27m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b62b4ff838724c2f862bca65711fb91b : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8352052e1a74926adc37740423f15f2 : SUCCESS in 23m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8cdef32e83d45ee806448b0b4ad9866 : SUCCESS in 20m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/07c53ac2310f47ccb7389a469694ce7b : SUCCESS in 29m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7c84d0c535584598878c2e9ab1b6b982 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a49455f1297d45b39c134c074ae857cc : SUCCESS in 13m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0523e1dcf13b4a6891ff77233dc31c37 : SUCCESS in 22m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/21a15f12f48243a2aa04a0978730a7ee : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d026cfd527a4411b0922d01dc1b3cb8 : FAILURE in 13m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e39dc74a14f4f419b23f0867212fab0 : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f3ae18d5b8f46378b47ea54c9d0563b : SUCCESS in 1h 08m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55591338c4a54d63a74d7d005ba42b7e : SUCCESS in 26m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1d50be574de4f3ea322743b15bb0d87 : SUCCESS in 50m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dc61bc0513e4a6aa2eadf0c77dc048b : SUCCESS in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53dad283b7a543a39e08c7bff9dc663f : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d642a8eafe4420ab977c0a55269ecbf : SUCCESS in 11m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f11f53ef57b54fbbb9918bc5b9cb4334 : SUCCESS in 50m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ff7f6e87e5b4901a39a48b9437d900b : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b504cb8471cbcdd005d3e41eee7e68b1c0eaee15","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-16 20:00:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b8998193b6cf91cb1bd86baa1777de698b986a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 22:06:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4607ccb4671c4cf9bdc2a5b4bf58e449\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a51ce64872ea4a04b9a61a868b041c58 : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1b7589f2d80e1b4ef56e4bf035625ec67ecf4dc9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-17 00:20:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b76ba94eac9dd659b4a753a595d32c10edf62c2a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-17 00:29:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f55b357b5f19cab137b25fba548d6c14f0b2bf19","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-17 05:31:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fbb2db86ae56339fd1d3629395e977aad881f108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-30 05:29:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"38097aea0c9a659d304fb7d1052cfa7025ff68c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 06:46:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fb818541884a8abc030c6773a74579\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91928e870474cf5a8447572554a0d18 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/073dc259826446f19bf4ddc737a8c292 : SUCCESS in 59m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0b479e59a954964a27af1e87d6d6516 : SUCCESS in 1h 02m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d464f04f8c314bd7909f6653a6b5b675 : SUCCESS in 57m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0344d99ac53f4106a95a58825c98fdb6 : SUCCESS in 1h 04m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02877a90a8fd46148022851b80d60a54 : SUCCESS in 11m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ae93bea36354cbaa966ace4ac098e3e : SUCCESS in 24m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2be9b00f92d4104937fae81990ad558 : SUCCESS in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f83a31755f54f7e8dad6aaf514fd826 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0dc07e3f736499da2195eb2a3e97868 : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bd29b5caf85141b1b87bdf1a11c6760e : SUCCESS in 20m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/20a01dfe434042d194f58af8f2dec5fc : SUCCESS in 27m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7146d5c1e80547ce97550ed47e15f695 : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/974481685791461ead87e92ab9903089 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01a545b526404cd78f63890a31d0c062 : SUCCESS in 21m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f4ad00961ab4092a49057138cf41052 : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec03c1fe98f84c4fa70493a48fbbb3bd : FAILURE in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/79c0047d69b148dfa86bb8b1c9834318 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/142fd7cb11ea431a9ac29c6bb2099851 : SUCCESS in 1h 01m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00916c2329534c308a19278ebb2a66f8 : SUCCESS in 31m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a46b48191f974700b143f2785afc5fb2 : SUCCESS in 41m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/858afb84bc8943c9be196361f5e7965a : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9c6b2b4e3544829c5eb91a10c289cf : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e25118efb264a1e908277c6deaaabd4 : SUCCESS in 24m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f12ffebdafe5486db82ce989e7c612cc : SUCCESS in 51m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16161c0a437e48848c90ab84920011d7 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2505b4567622c631e35ad1d6f584304a2f4045e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 07:31:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae825aad431b46d3b80be4b5fd4f85e2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1703765419bb4c139bef0a80c9db24e4 : SUCCESS in 2h 00m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a3d6f2808e833f7c9f908b482f15c85d477ff83b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-30 17:48:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e2485f18d8e30aa199afbfde830a05c468201e95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-31 21:51:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"62a62ba8b8ce5a2c9afa5202b46bfc0d3fab89bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 23:06:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5221e215f1cd468db06fd924737266bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8816c05ab79d40d1860018333d71b785 : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9141b53955651bc202636b6fee0bfadb4a8165b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 23:07:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70019bb167304aef8e9e260815678ffd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1aae501a184a9b9012283ac5a56749 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/309eb07593514fd09b27a959007f11ba : SUCCESS in 58m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5643b1c5573b4835861b3f9fd4014ec1 : SUCCESS in 1h 01m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20bcb82048af45868823623629f95dd4 : SUCCESS in 54m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa371044ec514e928405c9b202fc42cc : SUCCESS in 46m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba0d284e229d4c91b5050f294aaa910d : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/befb72458c244b21bcb09ad82527db5d : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c3d7f0f496b4edaa7ee6f930d7bd531 : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6038eb71d534d1aa68354fa835fda4b : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/626272763f2343cb9fd4c78f70540fbc : SUCCESS in 10m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62548805f69942fe8a37372030c5036d : SUCCESS in 20m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d692f0bbf5e49769750101be3559096 : SUCCESS in 13m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4be95862ff0549dfb8ec1bf4dc1cc9e5 : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/34d098e0c56c4d15a44d1ea97953de47 : SUCCESS in 11m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0e11e29a51d54654b05ea99a9c3b7ce2 : SUCCESS in 13m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/25e5366b78294e12a6adcfbf74a826da : SUCCESS in 9m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57a31e66d5804051be9be984b9534188 : FAILURE in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f750baf426714c7081de6835cc137589 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e78f6aeabed641948689db94f07b2ccd : SUCCESS in 49m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a68f7df7d3544bd3aeb99255e10b3d68 : SUCCESS in 50m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d5f0af37985437aae9f3d97baf401a5 : SUCCESS in 44m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0785b4ec3f460e9ea335b994f6fd52 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2121927a984ae98a7e8404867e2dcf : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c856ddf577648298877d994550fcc8a : SUCCESS in 17m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b021158131fd4f158283fa91e2da6bad : SUCCESS in 50m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a154befb08914be3b3fc0651845435e9 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":3},{"id":"1eb7167b0a6f2336ac8527c927745659b95322a1","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-03 05:40:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ca7cad0ba5ee24c466614009b7ec596cfc9185af","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-03 08:40:42.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e216b7a01974ed407b2fb726d6b15bcafd7f35cd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-03 08:41:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"135318e91c82a9128bbdcb840ab7d4cab1895383","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-03 19:03:56.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fc08bac9b0aff8ba818607ce9697cea7233cffdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 21:44:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"eb711666559e0f2c249a09f4e41aeac234dcfd61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 22:46:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/125a92e84dad473bb58fc3b947d42050\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dcb18e2bed9d4575a603b6449e1a0119 : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":4},{"id":"346c1f971730b5da212734816138a0d108a54925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 22:58:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35f54312fc2746b284ce36f5c02e5d87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3daffaa58d4e471d9f7b5b7aa61bc95f : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ef3496388fb04fae93d71f28997fac9b : SUCCESS in 1h 07m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16d8379ca3564a1db00a334d410f486a : SUCCESS in 48m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1616dfea2d44cbc945c86974d3f7bc3 : SUCCESS in 37m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3694f81538cf4dee9da17681c13fc70a : SUCCESS in 57m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aaf1a45a689b47ceb6fabf7cce7fd77b : SUCCESS in 9m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21c4a781f6a440a1ae5225dbb6b5ace6 : SUCCESS in 16m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ae9b987baa94753ad0f61304510e6ae : SUCCESS in 21m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f529035053946199a94470d60e3520c : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa2c6f8b24b24b14af0202f583525281 : SUCCESS in 18m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72a9954f2c9b465f840b46250384e266 : SUCCESS in 19m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ba3cf27f2184a28b76e4dcff9bab288 : SUCCESS in 16m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aae1af579a354012a0963cb7cf5bf4d7 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bf9e117950bd4073ba1b659187938cf6 : SUCCESS in 13m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76aed64034d24edc9411123a04cfeb2b : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/585e1db97e964adab3bc00327ef12d71 : SUCCESS in 3m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/acb9c24c11ee4fe6aabd47383355cf4f : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f22bcda22f0e4e5990d736a521005c35 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6ae65743aa84db2a26a4978c0397d94 : SUCCESS in 52m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26d4d79b6c1a494a82e498eccf033969 : SUCCESS in 34m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c666845d9a2b46bd8c61eb6f52979e56 : SUCCESS in 32m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02430322e03d49f1aa5b086ac0c970a8 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15c6d90ece74b3babbd661f87e7e4e4 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f68dc0e52aa472a92ac7d1957e9e49a : SUCCESS in 20m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/776beafa5ee94bae984b5e138896515a : SUCCESS in 28m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9020b30d174c45a1bbdb14db965920bb : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":4},{"id":"5df765e3c76a390665a518516454eb4a95e507f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 15:56:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1a0e6599fb1cd92943da8f78aeba75772b95a6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-20 17:00:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"3ad822059fa1fe8826f07feb60949354ebd46031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:19:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91664df8f3414cc5bed5fc5b68511d2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e12c1258966476aae0c9c48f3944e11 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/be8302e7c394462b80754bb2d7393ec1 : SUCCESS in 57m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8355d24a26d42d0b7755a84754a75f8 : SUCCESS in 51m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d912a684344d4a06bc9e2081b5035fb5 : SUCCESS in 50m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3fbcf7b4f2c45f197c2fe8a3edd718d : SUCCESS in 58m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22c66789c3b84a73b26cdc70ac34ba8e : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ecf02a4a4f04904976841a2589a99af : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21d3cc14eb0b467d83255da8efea3a3f : SUCCESS in 18m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f773b9ec857a42558b52caf68c00e37d : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f32b2122aaa44e1a2200a8d9dd8fbed : SUCCESS in 17m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d585ffbd78914632b5458b91f68acc47 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d1a3f46cba34463695feb5a26238753e : SUCCESS in 27m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e8a3a03a35564402859811965ad7b871 : SUCCESS in 22m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7d04470c61824f1ba5c4b835a6ac0e11 : SUCCESS in 7m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e55a554247ec44cc9bbf9a30b3b60038 : SUCCESS in 20m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/54073ff4104944c69e63a79bb2d98fcd : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f91aa6a30ffe405ea79452516787fdab : SUCCESS in 6m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d468106c391d4ca68fe882be2ac83331 : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/951df82bcece4566b4f05bf7a547e974 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14d8d5ed03e04a4ebad37a2ba0a7fc66 : SUCCESS in 54m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06aa192e6bca4af6b49b2e063e61fe2b : SUCCESS in 44m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/003a4a48efa0433ebe6e3a4e25da8438 : SUCCESS in 43m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd0b5248158b4838a3f78ae251fd069c : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac203bdcace453a8ff3faae5ac97e0f : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/376f0227929142e1bb71f656a97453e1 : SUCCESS in 19m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/866c2355f15b48d2a3feabecc1700d2f : SUCCESS in 52m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b099eeafb2cc47e59e4bf2e86898a39a : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":6},{"id":"086b285e6c9f0478ed724c0db68eb2a5725b32da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:25:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/336cceda52e94c698e1fa6b415515c4e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0101445ae18646b1be925a791216d5e2 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":6},{"id":"ada4e7bc7ea18e0e71c6e0b21766886d5c0eca58","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-12 17:17:57.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0ca7081409a13c97946b0551ec5002a61ab6075a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-19 16:26:52.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"6ecc92917c16c0f5f64748c0f058b56fe9e388e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:21:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b430e4848e54c7ba7a0d104244f0cc8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04a4788a7c9a4b5a833fa642284f7140 : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":7},{"id":"6f0ef66eb319ff08fe7196c0ca53a5db5dfc3634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:28:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44747aa6382f4da983f11d66f5e1e4c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20b110a8147451f803a60d7881d05da : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ce25e8f7dac6450fa00b3ccdbab191b9 : SUCCESS in 51m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8732225c14874214b08623d8b638f1c8 : SUCCESS in 58m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/334a2956482d4b8ea654ba12e1ed6130 : SUCCESS in 50m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8357e8b8f7cc4190853cfe15f9a63a6c : SUCCESS in 48m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ea421252135497eb7e432348ac5d911 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63a3c142e4d449c1b1311a6a45bb8276 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94177e522bce44f4abe521ee095db18f : SUCCESS in 10m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdb29859809143e7bef5c181c27a3d21 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a9af56954b3d4c82af183f6a0a8d1724 : RETRY_LIMIT in 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3f2e8eed32b24746b6a4f34cf5095b52 : SUCCESS in 8m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9daf10c8e2d746b9af72fdbf6de60e7c : SUCCESS in 10m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8e9ebf31ba0647c6b4dd0ab892e453e0 : SUCCESS in 23m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/279568a3093a43e3bc237e7d0fc761ab : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/78c9892e9d354c51a60088174e7ba5c4 : SUCCESS in 12m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/87b5cd7a0bd14462af1c4a087e91a746 : SUCCESS in 9m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f9d64312c5e464a8ea043cbd56fe71c : SUCCESS in 6m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9bcd1e953bae414e822123f09fc15d7b : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cfa47e65cc8c4a3ab42175fe752ef337 : SUCCESS in 11m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f94fabc8c624b4bb96f495a9a704af4 : SUCCESS in 49m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a01ec4ec3e9240f9b8a845884f0f37f8 : SUCCESS in 56m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e88633053fb4348adc98daf9677b389 : SUCCESS in 58m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/078dd7fd4c144a9eac9c6ecf7aed01f8 : RETRY_LIMIT in 9s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28738b2b3d7e4196952c50f3291804a4 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0cc38b60b6d4c8bac6f26287e715604 : SUCCESS in 17m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92207448fadf4d0e98837c2620eef364 : SUCCESS in 52m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aec29cf0a01b457097f39cf49ef7b34c : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":7},{"id":"be45b39f6f84d0ba43af754b7c71709ab7ded40c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-20 17:06:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"45fb07a95f2012efd2586fb7f02e25d89829efa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 18:03:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52fe58fda1f54c1e9db9c020f3662e7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08625bae089d403894063b136723bfd5 : SUCCESS in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a4c26722f60344c3b4e9f9f42fcea265 : SUCCESS in 29m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28f16b30b1614be29b5099550f31a225 : SUCCESS in 46m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82843e34cb364c85be15dc708bdff02b : SUCCESS in 49m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48e1f86877ab4aecb19d6fa9c71080d4 : SUCCESS in 32m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bafc9cbebee403bb3665708f8fbe5e5 : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46d09d8da6f24972b9914a38ef16e251 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7446791c67e14d9eb8f4406311335e8b : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4ab3cbd7656479b815416fd6a9e918d : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8a84f440015d4c92bb42e77fca7662b8 : SUCCESS in 33m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bfda580eaca477d83bab820391720af : SUCCESS in 20m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1fd74070d6af437d80cff8ac6e43cff7 : SUCCESS in 24m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/70bcb015321c4b8583634e124e56614c : SUCCESS in 29m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdc139acf477471b989ebcd70b563c4a : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff89b97a87ca4312b916a32b43c57c96 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c77cfaef7a94f818ed19f478955e4cc : SUCCESS in 9m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24676c4a92214331a8580e378f8f0850 : SUCCESS in 8m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0517ca533374170b376c5862cc5e153 : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e7cd622ca9646338656aec588e12fcd : SUCCESS in 11m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/98461fb3d30b407ebc7919c13002e139 : SUCCESS in 33m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/396d3202d9e344468b8dff82a427895f : SUCCESS in 29m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84fe5364c3e54c3796b2c93b9486b574 : SUCCESS in 29m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b36e55f260494147a8377e08017a7793 : SUCCESS in 26m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbeb75d5df74d51bc54377ed7b38317 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d295aa303204b2d98b58fe4226e4e07 : SUCCESS in 20m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a54e4c1619441e9934bb2368f29afd2 : SUCCESS in 37m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/419689f5833b4feb85cbdf9a6ab787ab : SUCCESS in 29m 12s","accounts_in_message":[],"_revision_number":7},{"id":"6f5477b06c31f22c60e4b10a327c6d7d92bc778a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 18:46:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4729abae9bfb4d26aad552f6fed00767\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62b50f3371c4475f9ec465051aa6929b : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":7},{"id":"7b61028bf2013063dceaeb25c53154def02b9a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-29 19:06:54.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2520c86af77d26b594c635daddadae104f7f4256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-29 19:53:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"518a8c73967b6583d2fd8f85818484f597d085bd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-29 19:53:48.000000000","message":"Patch Set 9: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1db1ecae290ee5033edd4c5c4b2e03bc7cffdf1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 21:07:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b617672e38c94289a71df2071a084a26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0f9e93066b4fd88f10c19c4516da9e : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/5a0189fdda6842179e473dd869d60add : SUCCESS in 1h 06m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6aee49ab878b413a8fc9e327ba43b458 : SUCCESS in 1h 05m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b03ced8a01648198b1dbae3b7c82f2c : SUCCESS in 51m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d4149b2155f41e789954f1ddd54f00e : SUCCESS in 1h 08m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/867b210e60794e76bc2337da5aa69e9f : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d72f43c8ad8648b88401f910c6b11f71 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e60882f521b4968b9689d1295b7c6d2 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f884524d60840ce91814e1c895705a2 : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e3668a24cd3a4a24840bf7fb7cd9a08b : SUCCESS in 24m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e778fce69eb4ba88c8e665d171d8bc8 : SUCCESS in 20m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ce5be2e302dd47a6a12eb3c3b1690b51 : SUCCESS in 26m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1d127ae527504c41a0e2755955b76481 : SUCCESS in 27m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/733c88b4c9c14981a54ff23d702cdd0b : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d935a33c2d364f12a83a2e5bb4fd2e35 : SUCCESS in 13m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7268777a47514e11ab2200105abad4e1 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b27232e207b14eaa8e46ed2430e6b92b : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/957cc37e03ab49c1b6ec5a98b98f3045 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfb016c941bd4c24a5f74e15cdb4c965 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d7ebda7518347fb8b0e1d7009040e39 : SUCCESS in 52m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a25d0d689edc43619f1cd5b1c2311e56 : SUCCESS in 35m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23db248a7076477481bbe13440c5ba18 : SUCCESS in 44m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e3ec05a52674018a784a729f01b07ec : SUCCESS in 29m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7f9af48039a4f798eeb111c5d37b652 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e0d86b4cba045a28e130c9097623192 : SUCCESS in 18m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d37765349de9406382b686bd629d9d02 : SUCCESS in 28m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bb3f04891694198a10467e1b0c658a2 : SUCCESS in 27m 51s","accounts_in_message":[],"_revision_number":9},{"id":"e7a347d9ed5b03d10c653fd188f16739a2c6f19a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 05:35:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902cffe6453e4abfb375fa4aba184cc0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8d18b7d73c246b7baa39b1497a01756 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":9},{"id":"cb6f1fb5e4accbbf8b527a088004fb25c9e6920f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-12-11 21:16:16.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nThe following files contain Git conflicts:\n* test/unit/proxy/controllers/test_obj.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"2075dbde1241bb6507b987a1f80b34909187e8ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-12-11 21:19:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"808551da91055bae12850573f09cd2b9afa270b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-12-11 21:40:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f654335f6d275938c588e580209b0b4601633343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-12-11 21:59:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6878a0beb3001cff2bfde62ad627fa01463cf236","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-12-11 22:07:03.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"dfc7b83f7ff6b83979a09e7d2afb436638cda179","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 23:13:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/506cc8731fa749518412468408ef3054\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8e6137e0037407087465222261a2160 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":14},{"id":"8895a3775610a5798e0dfb6eec007e8893851992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 23:16:57.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46c8ea014021451d85122815946f8e99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4253f7064c5a43239f07045aac6747e4 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/716d1fa2dfd645a2a3dcf5bc0b9b486c : SUCCESS in 46m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10a1df87ba484c9b8e00fcb0ff774fcf : SUCCESS in 56m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1efc317bc3534a4c8d5e1b1d9f53fed7 : SUCCESS in 53m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e3584ebc59f442f81dab5da51023979 : SUCCESS in 1h 04m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/388921482e624550b0f1659e9e70232a : SUCCESS in 11m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1e5a9c9e3f84f8fa1291f8a19b08d06 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3fc6162585b427797c0187bea11ddb9 : SUCCESS in 23m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7fa3d7ac93a24d06ae543f5326bdd621 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2665985146f74d0f989df3e8fc585b00 : SUCCESS in 20m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d0c417be7ed44d89d0eb7af4fe953d9 : SUCCESS in 13m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ceb34f28099412e81e7d954b07cc739 : SUCCESS in 10m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f4a3cd8d8c5d44fe8f8591714a26bb30 : SUCCESS in 29m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/89646adce5e14b65a9d77435cc24a63b : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d03a48e667624bb58fc8e7ed820c978f : SUCCESS in 16m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c0b4bf42e3174f65aa4c82062df4b3fa : SUCCESS in 13m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b551f1b96134076b027de6e9a32a3ca : SUCCESS in 5m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2cac4434bfa428db60ead67d3de8c71 : SUCCESS in 13m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9dfb3a81624b40d187943dd8977329d0 : SUCCESS in 5m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/696f84cd9f3a46c6a01a77553180b0dc : SUCCESS in 54m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc19827ba05148f2adfe77e830bb38d4 : SUCCESS in 47m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44eee97445784702865e13c03301f744 : SUCCESS in 31m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43159e3b19fd46118e95e03c924af871 : SUCCESS in 32m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f5ff8e3af26440f97fd5226a183c9bb : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69232a3b219e42a184fea969dab97798 : SUCCESS in 14m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c885490d2404a2a8c4cbfddce5bc803 : SUCCESS in 59m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d91d0d8c81c746b28d28553cd42f5ac1 : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":14},{"id":"82ff64dcf586977c88832aa23f3e121ace1eb22a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-12 15:37:50.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"3b8e50c992403b95f6b3b5e368f4c19006c684e4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-12 15:41:10.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":15},{"id":"20bac4c9e20c5cc27c96e32b19af9289435375f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-12 15:41:24.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"8f066297b0add5dc7f021a38a6100213e780985f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:40:03.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360f70830dfe4b03b4a77523975f09cd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a6809cd79d9c48b38fd64d875767765a : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":15},{"id":"ef7c5008684d4a26efa7813b9dc3cbcd0d37acad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:43:40.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5069eadce3c041129a4738b154cccddc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a89d6e391f4e2ba8a6683f74fa2465 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/35416a8fc718484199be84790921c691 : SUCCESS in 58m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7ecff5bea6e425a8beaae6870b09b18 : SUCCESS in 51m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae7fb8bfb3874c5aaf6812744afa33e3 : SUCCESS in 30m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5856004b84c649e8ada128e17262e779 : SUCCESS in 54m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0f81f159d974c3ba3bcc5fccd4ff66a : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7f9205537c44c98ad1a8fade3b55c7f : SUCCESS in 11m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88a1544e13d64989b7687181e6cba5ce : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37fe22d48ded44f597d95c89b379f787 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a399f529ea842b08d5652c3256511d3 : SUCCESS in 25m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/85b545c6feef4313bfff094a572de794 : SUCCESS in 20m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b9d339c9e5274b25acd3ee2fa81bc953 : SUCCESS in 12m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/da00265ca7f44906b18c93d4c3fe081f : SUCCESS in 29m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95eec2c024e6439b8667f87f52dc8c2e : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61d57f4494f443f199e847b1cb32fd1e : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3f5f5cb839d142748c0799fc743e7b20 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f22ca9c13b424b8b85370f3fb0384b3c : SUCCESS in 8m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c43f027b3cbb473fbd873fa32f480a1b : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eae682b7fe8c4adf96d950681660768b : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57dd188d182449ed9fbc7dec2c961eeb : SUCCESS in 35m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f8da3bdda014a96af53a4c03a68c981 : SUCCESS in 22m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/462701edbce84dfea0d05b83b0568feb : SUCCESS in 50m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf763ec99fa946a3aa9a89a53580dafc : SUCCESS in 26m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be142b9103ce4d56a6025de66ab3ac14 : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e07fdc7b546442f79aeba3d41991385b : SUCCESS in 19m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a55b35ebaea841549fcfce06e18acc94 : SUCCESS in 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/684ce67097274619a5ece9475442f048 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":15},{"id":"44351efe1af200f88a17d3643a02c7cc6d65114c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-11 06:06:33.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"b7d942c507296cedc1e5b715348073b24861fdc6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-11 15:34:40.000000000","message":"Patch Set 15:\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"b1c743f682689a5fcbeb62a0345c84b5264ee891","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-01 22:51:39.000000000","message":"Patch Set 15: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cb0592b6a8159e1d0512ea2d5de834405e8b76ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-02 23:58:44.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"8e95ed882fc2d6cf4e8847153c3c0169ee6cd12a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-03 00:09:57.000000000","message":"Patch Set 16: Code-Review+2\n\n(10 comments)","accounts_in_message":[],"_revision_number":16},{"id":"649db990aad0c26eb02ac0ff307c4b45ef1bf93c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:55:08.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20faef2a31604bcbacf62acc3afdd3a2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70181973483d464fadf5c8bcf178cd87 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":16},{"id":"6e24cdd74ce3087274616248ff8186e6fa96f979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:59:13.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dcaf806a873418091c0a7ffa682d5ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0337f21b84934eaeac0a59f52faaf57f : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f97e7cb21b740c6aa6d15835aeaff06 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/1c763a92bcac498484ecc18992699fbc : SUCCESS in 1h 01m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16b83c49fae04a409404022882680640 : SUCCESS in 1h 03m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/99d74935de624d199bfb5aec0b6358ee : SUCCESS in 28m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5faba7ede23542d0b5653f55f5cbd785 : SUCCESS in 46m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33171925bef34978a32ca566ca947c57 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/690bdef56a8d4c2fb3562280f453cd6c : SUCCESS in 12m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a5f04be77eb40b19dd499b66cbe0222 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/10085ce7b39640e5ab15a8c7396de3de : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/23925e8c06544efa978a447b3afa7b8a : SUCCESS in 26m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/28ec83b4b0764b8fa932d89867c1a7f2 : SUCCESS in 9m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3fc96954e46e4245bd5318a5dba8c171 : SUCCESS in 11m 48s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ca495b8471a0492dac8e14209b90f8c5 : SUCCESS in 19m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b551cb1bdcf247799e469e60c4bf395e : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/500f505891394525a8ba8ee61626ad15 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/23a6fe0075254c89a9fd73f48b17fb5f : SUCCESS in 14m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdce81d5457f494cada3464e0eb6432b : SUCCESS in 20m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ab3a55c3f5846a88c317878b0062471 : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/74ebfa3ec5c048599b6e56751f5cf2b5 : FAILURE in 26m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/00a2902f2b30437490b305cfe6a66994 : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7b517f907c44616b63a1066214cfbcd : SUCCESS in 59m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a64232ac1a44b92a459795a11c1e539 : SUCCESS in 1h 00m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/017d3dda60f045319c6592ebe0479f03 : SUCCESS in 25m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f619389da5c04b559155de09c91bac1c : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8f6f9fc95824b8cac5cf5155be70bd9 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6868dd96b3f4158a7ca86ea85d80120 : SUCCESS in 19m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cdac41375664425bb9d42a795e228e2 : SUCCESS in 54m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"0ebd138777c88ae54f7843da588f391eedad5bd7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-09-03 04:14:01.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"9fb2a26cbd0db43642cac4c4e5ea3fa7ce121b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 04:15:30.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"82ff816c39387918901b806ba4f365761382959f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 05:23:00.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f84d6510e843db866cf893648d205a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fc07ba85f7407b8a0c58d1d4e14b17 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d01d6d8d1fe476a8e52658f626cc556 : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/76c0793e2ede40ccb5f9962e1ec9d156 : SUCCESS in 34m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88406576ef344e42bbdddc6d0cb50244 : SUCCESS in 35m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39e80af882064f1f990459e40daf0de2 : SUCCESS in 29m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/107de507544849cdb9c0f7087d171164 : SUCCESS in 33m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6d4d2810459d487eb2e8112d366e26bd : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5102b170d63f431bad0bb58980f099d3 : SUCCESS in 13m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00410f49c5a94cfcb82d4662cdc23b25 : SUCCESS in 11m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0ffe2c2d26734c14b98eb76f3d4a3858 : SUCCESS in 31m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b046e89260a047f089f1e6def1ff3369 : SUCCESS in 16m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6beec1c6fe79455693a7845a7b155add : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0ba2c34bdaae4b4e826fbc92494f3b4d : SUCCESS in 10m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a0f0a83adc2747bfb13909fda738ecc8 : SUCCESS in 11m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a330fe5c09ca4060b1e4a56f22d510b9 : SUCCESS in 4m 36s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/24ca58b4530343d08206ff4fcb98b104 : SUCCESS in 5m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9412b871138f47879d69758a75be4a79 : SUCCESS in 1h 02m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/054893692fdb4a41b2a8e474b0b7f0ec : SUCCESS in 27m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/225d4c4bb9da4688a1af6f16f51d0460 : SUCCESS in 56m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aaf10cdc23a4fb884589c4eede3b174 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4784c924d205482dbabbb0c5f26f7074 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0280482274d84ffca9065a63349bdd05 : SUCCESS in 26m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7908d418837c4a98989f74f92c1d4ca8 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":16},{"id":"6d636b70ff015d770aebc08f6155c328313a1088","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 05:23:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"77115897d58911a85de50d54df9f2f3f3ae72a26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 05:25:02.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e699aefe22247da8dd455e4fd34cc1f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f2511c65f04f6aa427a70905b58d1a : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4e939e2a5b2242e794a0f16a73b5a7cb : SUCCESS in 1m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"40933680b982310f0bd01cdcc81203e22c911f64","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-04 17:39:11.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"f666535af6a6afd4dcc167ba511edf4e1c57d50d","revisions":{"947d00a914cd671e28662669efe2751a71980bdd":{"kind":"REWORK","_number":1,"created":"2025-01-16 17:06:36.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/939471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/1"}}},"commit":{"parents":[{"commit":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947d00a914cd671e28662669efe2751a71980bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947d00a914cd671e28662669efe2751a71980bdd"}]},"branch":"refs/heads/master"},"dab922a01583d1e756c7b98c85945a92281dbe66":{"kind":"REWORK","_number":2,"created":"2025-01-30 05:29:13.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/939471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/2"}}},"commit":{"parents":[{"commit":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-30 05:22:29.000000000","tz":660},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEC HEADs shouldn\u0027t check 2*replica count to find metadata. Every\nfragment in EC stores a copy of the object metadata, so we only ever\nneed to visit any fragment to answer a HEAD response. Replica_count\nwould be plenty.\n\nWhen HEADing an object that doesn\u0027t exist, currently we need to check 2\n* replicas nodes before we can return a 404.. this is expensive.\n\nThis patch changes the number of nodes a EC HEAD makes by generating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, which is much more sensible. For further investigation\nthis patch also adds a option (that is policy overideable) called\n`experimental_ec_head_limit`.\n\nWe\u0027ll use this to investigate and tune further. Note, this experimental\noption may disapear or name may change.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dab922a01583d1e756c7b98c85945a92281dbe66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dab922a01583d1e756c7b98c85945a92281dbe66"}]},"branch":"refs/heads/master"},"fa56156814c4f8fa34800fd8a77a28bea12c5c0f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-01-31 21:51:25.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/939471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/3"}}},"commit":{"parents":[{"commit":"287fa42ab8ff101a8ab59aebd44f25028d096c1b","subject":"Deprecate expirer options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/287fa42ab8ff101a8ab59aebd44f25028d096c1b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-31 21:48:20.000000000","tz":-360},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEC HEADs shouldn\u0027t check 2*replica count to find metadata. Every\nfragment in EC stores a copy of the object metadata, so we only ever\nneed to visit any fragment to answer a HEAD response. Replica_count\nwould be plenty.\n\nWhen HEADing an object that doesn\u0027t exist, currently we need to check 2\n* replicas nodes before we can return a 404.. this is expensive.\n\nThis patch changes the number of nodes a EC HEAD makes by generating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, which is much more sensible. For further investigation\nthis patch also adds a option (that is policy overideable) called\n`experimental_ec_head_limit`.\n\nWe\u0027ll use this to investigate and tune further. Note, this experimental\noption may disapear or name may change.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa56156814c4f8fa34800fd8a77a28bea12c5c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa56156814c4f8fa34800fd8a77a28bea12c5c0f"}]},"branch":"refs/heads/master"},"2b57ad9c9b3bede3ed206058d3228090728ecde7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-04-10 21:44:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/4"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2025-04-10 21:44:26.000000000","tz":0},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEC HEADs shouldn\u0027t check 2*replica count to find metadata. Every\nfragment in EC stores a copy of the object metadata, so we only ever\nneed to visit any fragment to answer a HEAD response. Replica_count\nwould be plenty.\n\nWhen HEADing an object that doesn\u0027t exist, currently we need to check 2\n* replicas nodes before we can return a 404.. this is expensive.\n\nThis patch changes the number of nodes a EC HEAD makes by generating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, which is much more sensible. For further investigation\nthis patch also adds a option (that is policy overideable) called\n`experimental_ec_head_limit`.\n\nWe\u0027ll use this to investigate and tune further. Note, this experimental\noption may disapear or name may change.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b57ad9c9b3bede3ed206058d3228090728ecde7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b57ad9c9b3bede3ed206058d3228090728ecde7"}]},"branch":"refs/heads/master","description":"Rebase"},"07d0458fe85f3ac7adee826281eba1bb66682806":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-05-20 15:56:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/5"}}},"commit":{"parents":[{"commit":"3b3bb1376f56fa65080e996f072a945a4138d92b","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b3bb1376f56fa65080e996f072a945a4138d92b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-20 15:06:42.000000000","tz":-420},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEC HEADs shouldn\u0027t check 2*replica count to find metadata. Every\nfragment in EC stores a copy of the object metadata, so we only ever\nneed to visit any fragment to answer a HEAD response. Replica_count\nwould be plenty.\n\nWhen HEADing an object that doesn\u0027t exist, currently we need to check 2\n* replicas nodes before we can return a 404.. this is expensive.\n\nThis patch changes the number of nodes a EC HEAD makes by generating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, which is much more sensible. For further investigation\nthis patch also adds a option (that is policy overideable) called\n`experimental_ec_head_limit`.\n\nWe\u0027ll use this to investigate and tune further. Note, this experimental\noption may disapear or name may change.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07d0458fe85f3ac7adee826281eba1bb66682806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07d0458fe85f3ac7adee826281eba1bb66682806"}]},"branch":"refs/heads/master"},"ed1abb0a077a421e1d7d3f7411d3a792e29241cf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-05-20 17:00:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/71/939471/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/6"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 17:00:37.000000000","tz":60},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEC HEADs shouldn\u0027t check 2*replica count to find metadata. Every\nfragment in EC stores a copy of the object metadata, so we only ever\nneed to visit any fragment to answer a HEAD response. Replica_count\nwould be plenty.\n\nWhen HEADing an object that doesn\u0027t exist, currently we need to check 2\n* replicas nodes before we can return a 404.. this is expensive.\n\nThis patch changes the number of nodes a EC HEAD makes by generating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, which is much more sensible. For further investigation\nthis patch also adds a option (that is policy overideable) called\n`experimental_ec_head_limit`.\n\nWe\u0027ll use this to investigate and tune further. Note, this experimental\noption may disapear or name may change.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed1abb0a077a421e1d7d3f7411d3a792e29241cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed1abb0a077a421e1d7d3f7411d3a792e29241cf"}]},"branch":"refs/heads/master"},"ddbe7b4fbae76dee7ded79dc49fde05e0cd56e87":{"kind":"REWORK","_number":7,"created":"2025-08-19 16:26:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/939471/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/7"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-19 16:26:44.000000000","tz":-300},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEC HEADs shouldn\u0027t check 2*replicas count to find metadata.\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nImportantly, when HEADing an object that doesn\u0027t exist, currently we\ncheck 2*replicas nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddbe7b4fbae76dee7ded79dc49fde05e0cd56e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddbe7b4fbae76dee7ded79dc49fde05e0cd56e87"}]},"branch":"refs/heads/master"},"2a357621c6132793846f4fe467d0df93295a9e45":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-08-29 19:06:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/939471/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/8"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-29 19:06:54.000000000","tz":0},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a357621c6132793846f4fe467d0df93295a9e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a357621c6132793846f4fe467d0df93295a9e45"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a72c112402a8405f0575e71dc8390cb813e11329":{"kind":"REWORK","_number":9,"created":"2025-08-29 19:53:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/939471/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/9"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-08-29 19:52:44.000000000","tz":-300},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a72c112402a8405f0575e71dc8390cb813e11329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a72c112402a8405f0575e71dc8390cb813e11329"}]},"branch":"refs/heads/master"},"38a433612e06dd05de56ee9ab98ca82cf1f3e03e":{"kind":"REWORK","_number":10,"created":"2025-12-11 21:16:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/10"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-12-11 21:16:16.000000000","tz":0},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a433612e06dd05de56ee9ab98ca82cf1f3e03e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38a433612e06dd05de56ee9ab98ca82cf1f3e03e"}]},"branch":"refs/heads/master","description":"Rebase"},"00cece362a928933906162a80f8a961bdf9cd486":{"kind":"REWORK","_number":11,"created":"2025-12-11 21:19:57.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/11"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-12-11 21:19:47.000000000","tz":-480},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00cece362a928933906162a80f8a961bdf9cd486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00cece362a928933906162a80f8a961bdf9cd486"}]},"branch":"refs/heads/master"},"8b989484f1fc571262711a301a104f6939fb04f9":{"kind":"REWORK","_number":12,"created":"2025-12-11 21:40:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/12"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-12-11 21:40:46.000000000","tz":-480},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b989484f1fc571262711a301a104f6939fb04f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b989484f1fc571262711a301a104f6939fb04f9"}]},"branch":"refs/heads/master"},"0aab07ac3e0221f158e733cd112aa74e85bd254d":{"kind":"REWORK","_number":13,"created":"2025-12-11 21:59:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/13"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-12-11 21:58:55.000000000","tz":-480},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0aab07ac3e0221f158e733cd112aa74e85bd254d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0aab07ac3e0221f158e733cd112aa74e85bd254d"}]},"branch":"refs/heads/master"},"11bf6e25d31564e8e7cb2888f59e33fd8e557028":{"kind":"REWORK","_number":14,"created":"2025-12-11 22:07:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/71/939471/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/14"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-12-11 22:05:55.000000000","tz":-480},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11bf6e25d31564e8e7cb2888f59e33fd8e557028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11bf6e25d31564e8e7cb2888f59e33fd8e557028"}]},"branch":"refs/heads/master"},"d280f5cff87e4ea7857c9fc9d0be8cd0224989cb":{"kind":"REWORK","_number":15,"created":"2025-12-12 15:37:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/71/939471/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/15"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-12 15:37:25.000000000","tz":-360},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d280f5cff87e4ea7857c9fc9d0be8cd0224989cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d280f5cff87e4ea7857c9fc9d0be8cd0224989cb"}]},"branch":"refs/heads/master"},"f666535af6a6afd4dcc167ba511edf4e1c57d50d":{"kind":"REWORK","_number":16,"created":"2026-09-02 23:58:44.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/939471/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/939471/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/939471/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/939471/16"}}},"commit":{"parents":[{"commit":"3201b5393f3f9b7d6bbd292024257c0474e224dd","subject":"Merge \"wsgi: Add url length limit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3201b5393f3f9b7d6bbd292024257c0474e224dd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-16 17:06:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-09-02 23:49:31.000000000","tz":-300},"subject":"Limit node_iter on EC HEAD","message":"Limit node_iter on EC HEAD\n\nEvery fragment in EC stores a complete copy of the object metadata, so\nwe only ever need to visit any one fragment to answer a HEAD response.\nHowever, when HEADing an object that doesn\u0027t exist, currently we\ncheck `2 * replicas` nodes before we will return a 404 - this is expensive -\nreplica_count 404 is plenty.\n\nThis patch changes the number of backend node requests by creating a\nnode_iter with limited nodes when doing an EC HEAD. The default is\nreplica_count, just the primaries, which is quite sensible and worked\nwell in production workloads. For further investigation this patch also\nadds an option (that is policy overideable) called `ec_head_node_count`.\n\nChange-Id: Ibf55eab0f6f7f5ec67e18c586779770f78680dbf\nAssisted-By: Codex:gpt-5\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f666535af6a6afd4dcc167ba511edf4e1c57d50d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f666535af6a6afd4dcc167ba511edf4e1c57d50d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
