)]}'
{"id":"openstack%2Fswift~939512","triplet_id":"openstack%2Fswift~master~I11a64a371566cc2d2b8d4533dc32dbdee7416c47","project":"openstack/swift","branch":"master","topic":"ec-head-node-iter","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-01-27 23:32:13.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"I11a64a371566cc2d2b8d4533dc32dbdee7416c47","subject":"sq: add some initial tests","status":"ABANDONED","created":"2025-01-17 05:28:47.000000000","updated":"2025-01-30 05:29:34.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac9e82cf00d8f0ffb6d495aea4a2377d38db7903","_number":939512,"virtual_id_number":939512,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-28 00:33:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-17 06:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8887db9aa54f91d972f6979148b943e1b77cb1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-17 05:28:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"646b1783db1c05ee22e34da4555bc767bbfed0f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 06:28:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c34cc7a371a40e7bc67e2e6a84fef78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8974d620c33a4437bd9c74bafeddc8aa : SUCCESS in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/7903d02a605f412b929d1401f0932173 : SUCCESS in 45m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ce01fbaaaac46779357513d16e2b45d : SUCCESS in 48m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c610158e608e473d93de5df63b8a45c2 : SUCCESS in 41m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2364103c949a4604bf3559464fe787da : SUCCESS in 47m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4f5f3f24cb04a8b8c7e2e2a74e73974 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58e2c3d48c774677ac1911d1535d5d7f : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79f94cab76664889a480de8334ee8bec : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59d3b416bbaf4db983b68a01f0960b91 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7d46b41f171a489199192ba3bf5b77f6 : FAILURE in 9m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0457ba3664344e38bb9e3fedb660b5e : FAILURE in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c757c21d1ad14a96a0b71cfe223b1349 : FAILURE in 28m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2288b2d6df2f4608ab3d7276d6fdcaa0 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6e0424b8317943ddad900bc35649184a : SUCCESS in 11m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2bcde8e2c0ad427488bb524c39be57f9 : SUCCESS in 13m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6e29204a56740a182f74d90d46e2ab4 : SUCCESS in 8m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/076a61a5c76049cfbe56c00306c2f37f : FAILURE in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/78d8df8b79c8427997f6e523f175164a : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b88c8a74bfb454899d2301769aac60f : SUCCESS in 54m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6db79bc3dad9427b842a4400dddc3341 : SUCCESS in 25m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2734a015a7a442dd8b5b6e7120762fc3 : SUCCESS in 51m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05d76b503ac54e5da2146d0c91557287 : FAILURE in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf86ecf3f874c64beb4da9492b42c49 : SUCCESS in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/716174477008461d992c47912c9f98f1 : SUCCESS in 26m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf53aec16e1b4f4e8d34a3245b7a3bb8 : SUCCESS in 53m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96a5e47680c646f0aa4544c169a32e51 : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"bab66d5fedd6ebed3e136e9b40f1ba4fbf1ca09f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 08:04:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de2f15d2b5df40188d45662562288258\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4e349b8cb0d41f89b0a9120ae6760b3 : FAILURE in 2h 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d7358d1af22518df368da1369e998d3805910b2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-19 23:01:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a6cf279c98b8bb661a63232c62f5a2aceddc50cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 00:08:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23147d36b8e845358e555e513e016400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a13e031071a43529733b2477829adc0 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b7d6bf58dbfc4b6788d1f3ee75d0210e : SUCCESS in 41m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d62d146dd304991836661a6da33519c : SUCCESS in 39m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eee90c69e0a7415fb1eba15722c7a0ef : SUCCESS in 38m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bac2e43b9a349a18272a4d182ce4ab2 : FAILURE in 44m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b426af140174235a2f75e2c999c236b : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da1683c9e0114e1e9fea196bdeff2aa5 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3db8724a35e448f946ce9c7cc7ae0a1 : SUCCESS in 17m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43307acce7d6438c87d91dd9b97c6c14 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/151338e57ab4445bb83fd0fcf7b5aaf9 : SUCCESS in 10m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31ef2213798d45008f2e3e240164527e : SUCCESS in 11m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26047aa17b614d8bae6a0cb59b64ceac : SUCCESS in 32m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c53c1aa601cc4aa691502187b1fddca7 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/186c47ec21e743fbb8742310be711f80 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eaeee352d9a44635836e63a98381a103 : SUCCESS in 10m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ab8605a2bfb4cfe9967a5d131528d32 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9abf3f2d0c664c2682819b5b06922b51 : FAILURE in 14m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c212f0d66e64441a91d331675baecdc : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/109e1f6ec63649c8930de751b92a29e5 : SUCCESS in 55m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad7ed52c543d498b9dacaf45c4b906ab : SUCCESS in 50m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3cb1de203e6f43e38b750f8012d30187 : SUCCESS in 27m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20ecede66631471ca37e7864b62dc586 : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257c01f295ca41bc86f482d7f87f4106 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c5575144ba54b4bb38054c97f65483a : SUCCESS in 13m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c842954e6f4440a1ab27ffe71004b6e4 : SUCCESS in 1h 00m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cef9d120d1447f19822d70928cf48ff : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2580ef7ccbe212f50b810d5cbc8843e3b0a58bb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 00:08:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f157fb8230b046eea66c1ac433660ad1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7b0ac158b8d42c98c0ef297b24078da : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d88e599ae5478e08f8f541a413559569ba903ca5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-27 23:32:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1541a0143c7618a63c0ca6c607e3d3b0b0d3fd6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 00:33:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af19ebe4fbdc48b4bbebf288ef8f4a34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35def26185045cd8a22c0d8eedd6dbd : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5f97f4eb638a4664b0575e78db0e047d : SUCCESS in 42m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86a49104d88646afbcdf834b34d7c186 : SUCCESS in 38m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fad8e41189b447a689d193d6642ff0b6 : SUCCESS in 35m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1add63493a1d46d787c7f25e274eb42d : SUCCESS in 47m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c16ed4f4f6714c96a259125c8b24c670 : SUCCESS in 11m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d54ad7cb30024d3a9586e0dcb72ebda0 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd64f98c9eda4dcd8187531d20856917 : SUCCESS in 16m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/591479022056479c8820ccc38b1c4997 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e12ae1fdb10e4562b5b786fa890db35f : SUCCESS in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6547518de8aa4819b346c03280baddbe : SUCCESS in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/64a9e8f013f84376a0aebc9eefbaf809 : SUCCESS in 12m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4902f60c84924c01be66753668d2f859 : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/edc17f8ea5404bfca3f7e10bdf8a9b8e : SUCCESS in 10m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/145cd0497d9c4113ba7a0d00197c7c73 : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cceab583d3b24ed1a18de1f77358e8e4 : SUCCESS in 5m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec4cb5f0470f461e8f3c9e0ad79495f8 : FAILURE in 9m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/059ddd848c114ca2bc540edfd9a9ad6b : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/88bde40292c54c68844e508874d41e1c : SUCCESS in 46m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7217a94c7d84f7d8bbcc20664c8f43a : SUCCESS in 25m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75b09e5fcc794f1fa00332132d16a977 : SUCCESS in 54m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f06c2f2b6df4fb580e828b9806ffa2a : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b6c6deb15746aea42f7c88cbf95a41 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98f42dc3a6a34aae8df028a8b2323094 : SUCCESS in 15m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8f89ef4f850421ab3efa89a51925849 : SUCCESS in 38m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e8511922359456499e97998202a70fd : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1216592333052febabff1aae44b237d7a369de36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 00:35:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf6407809ce46c6853658023a7a1755\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7176cf0f4ed547d2940403bfc863cfa2 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ac9e82cf00d8f0ffb6d495aea4a2377d38db7903","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-30 05:29:34.000000000","message":"Abandoned\n\nmerged into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82f0a49adc0587d1f1c34c9f139934a555aaf473","revisions":{"459e5ea501b71063f12099a47935af0756cc6b45":{"kind":"REWORK","_number":1,"created":"2025-01-17 05:28:47.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/12/939512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/939512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/939512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/939512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/939512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/939512/1"}}},"commit":{"parents":[{"commit":"947d00a914cd671e28662669efe2751a71980bdd","subject":"Limit node_iter on EC HEAD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947d00a914cd671e28662669efe2751a71980bdd"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-17 05:28:12.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-17 05:28:12.000000000","tz":660},"subject":"sq: add some initial tests","message":"sq: add some initial tests\n\nChange-Id: I11a64a371566cc2d2b8d4533dc32dbdee7416c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/459e5ea501b71063f12099a47935af0756cc6b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/459e5ea501b71063f12099a47935af0756cc6b45"}]},"branch":"refs/heads/master"},"82f0a49adc0587d1f1c34c9f139934a555aaf473":{"kind":"REWORK","_number":2,"created":"2025-01-19 23:01:17.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/12/939512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/939512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/939512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/939512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/939512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/939512/2"}}},"commit":{"parents":[{"commit":"947d00a914cd671e28662669efe2751a71980bdd","subject":"Limit node_iter on EC HEAD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947d00a914cd671e28662669efe2751a71980bdd"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-17 05:28:12.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-19 23:01:00.000000000","tz":660},"subject":"sq: add some initial tests","message":"sq: add some initial tests\n\nChange-Id: I11a64a371566cc2d2b8d4533dc32dbdee7416c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82f0a49adc0587d1f1c34c9f139934a555aaf473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82f0a49adc0587d1f1c34c9f139934a555aaf473"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
