)]}'
{"id":"openstack%2Fswift~939611","triplet_id":"openstack%2Fswift~master~I5b6615b4646d32408307e47a956602bb94768b54","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-01-20 22:06:05.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"I5b6615b4646d32408307e47a956602bb94768b54","subject":"ring: add cli/ringbuilder version command","status":"ABANDONED","created":"2025-01-20 04:25:46.000000000","updated":"2025-02-12 00:14:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af9bd6d28e8f06b4ffe3d0045f44fbb712153a19","_number":939611,"virtual_id_number":939611,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-11 12:43:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-20 05:22:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"947a0019585512f4a230d4d5a0d1f2d2e05fd163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-20 04:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc4ea7f7abdd77ca83b5cd2ca8dc7044b63d23dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 05:22:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c90d467719a146f08aa0832ce8d65371\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72771482944d4ce1ab719f945bed506b : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/9a44c25682ae455ca8d20eef58bfc142 : SUCCESS in 45m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bda69ce0022c461cb04c51b9fa3ea857 : SUCCESS in 40m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68b087ad11a64f9981044be8eca55f7e : SUCCESS in 38m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc3a19fc8ff6496ca05e318284776773 : FAILURE in 47m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f4f8af79b1e45d7a8f020d1d323898b : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/876de65d60ef4838aa2ece492407cb59 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f23374adab34b73ae450621192f1740 : SUCCESS in 16m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc6c43f230064d8eabc21032a61e1517 : SUCCESS in 3m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0981215e97a94e02aa3da0438f735476 : SUCCESS in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b5f6982be2ef44d38d78e133e325b259 : SUCCESS in 19m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f5695a486faf4b34be734a064c3f003c : SUCCESS in 33m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f675e9c299042178731c74d046e0930 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f4add9c2091f484f828d8950dbe6d21d : SUCCESS in 9m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5a9bb3260f8447cc9685db5967b13a79 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3a02cb407d0483383ceee0c18322a7c : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11aa247fe61c4ce1847273920a054f5c : FAILURE in 14m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46b0cbef01184ba6a3e8964020279547 : SUCCESS in 6m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6f9db66d05e44b9a5e7fa52caa70bb2 : SUCCESS in 55m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4aa627b2910a42959998950e8aac4a89 : SUCCESS in 24m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cbcc28ecb254ae99c2c6a34812067d2 : SUCCESS in 38m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee6533bf1e1c46ef959cc31aed50c0bc : SUCCESS in 22m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd06fa3c7f54ac9964ad5ca44c8a0c9 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6338a31eb0b4af1b079cc496070b524 : SUCCESS in 20m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a48e95c052cc4f7ab9c638546c840d75 : SUCCESS in 43m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf357a8da8c2440cb15c49bb13de3050 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"047d010ce31e1b8505da472ea09a000356f22a7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 05:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1c79095b5c4e06b279a84a0196f237\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e14d4547a0354eaaa3fd78d4eb5a5dbf : SUCCESS in 1h 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9eeea2786253efc9394bd50ff91b39fda9450be1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-20 22:06:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ba87785b7d2e1360c5e7344c2323c13ec61ce91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 23:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be42a3cf82664da0b40e1afb413a07cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4a089c1edf408292d400e569e43c5c : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/18e15b024974469d84c8a6c66ddd1c23 : SUCCESS in 1h 10m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d0d023e421749e5b746fa1e48bdcb55 : SUCCESS in 44m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfa325899261471c854449cc45c0ce12 : SUCCESS in 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42d30cff714e45d0ac5b4d4edfad098a : SUCCESS in 50m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c96cfb336ef4c8f916ddf8964be6c94 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ccfca289b044703bd187eeba3e372f8 : SUCCESS in 23m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/600b69458c0b411ea2a4a37794dce2f5 : SUCCESS in 16m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f22cc419269443598e3cad5108fe2466 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cafe7739c8f2435a99556e8625d88bce : SUCCESS in 20m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/56f29b287f6e41e2814aafbeef588594 : SUCCESS in 22m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e28e8416d9584cb99e4ec7e068739833 : SUCCESS in 22m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c2c2af15f8744dcb996369f7158e4ed3 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b3e39494a19a4a96b743af2f9a8a4df1 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/65a6eaaf4bf841b2a1340d615fee5211 : SUCCESS in 11m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3fbd7cf29fa945d9b800cabff22f15ba : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d31414a5e65f43f4b8ede75b57a5e76c : FAILURE in 19m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46df3f2afdf840ef92ad6f5034a48df8 : SUCCESS in 9m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f91e4e492794d569fe595e47a8d764c : SUCCESS in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d85dce27f5224ddc8c457ff908cfec6e : SUCCESS in 45m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9e6bcedc5614cdaadc9232f3eb3bdf1 : SUCCESS in 52m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/563dbce1698a4c828234c008237335e9 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533944197c794adf8efede34d47fbb5d : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63d3286caebc49c2b09a2136840b724f : SUCCESS in 23m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a27d15f45143cc9ac38f240673b65d : SUCCESS in 53m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/240dfa5c0792417c9aff9d2febbd7fa1 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c3f1e7fb5a827732cac6b8a41721d8eb5dbee2dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 23:41:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19616be7aff244f4ad88265fb77e10ff\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e459434925ff4dc9bfcabc64597f3fe4 : SUCCESS in 1h 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c602082a5495af9fdd20dbecca1ec0fc96103cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-11 10:00:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9ac294b6a362f884c2744fb30f76a9956a8f9624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:43:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/374d7bb799ac4f80885b7983650f9fe9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0f7828f51e462cbd9d828b8a628efd : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/267737a0342c45b8b9d86c2af480e6d1 : SUCCESS in 55m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/feabe406fc4d4066b620aec1daf7d6e7 : SUCCESS in 53m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8206b0ff13a4dc9a9dfe26560020482 : SUCCESS in 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/029af29f321b49f3b434756f32315242 : SUCCESS in 1h 53m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf47e03b34e84442a9660c50a0de8b45 : SUCCESS in 39m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fa8246aadfb49408f149eee68d3655f : SUCCESS in 40m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26c48e538432468ba3b721c0851692f6 : SUCCESS in 25m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/936cb5b0de5c437a8ade707128c01395 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47382c2fed2d4d9f91d5baa23c9e1e40 : SUCCESS in 17m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/847b0f7fe3524102b76879cf660c2fd7 : SUCCESS in 20m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5828a7dd428d4f229fdd0e27f0d275eb : SUCCESS in 33m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/46ac17cda9114371ab05fb63eaa98b3d : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/497810f870264702bbc38ef3c87f0c49 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4347347c0eff455c950844fc75b5300d : SUCCESS in 9m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5304f9d3d49474bad3e30b6717633ba : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26ed1378032843b2996e8385bdfa42a5 : FAILURE in 10m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7bd3ce11720a4219894b292390c93633 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d1c987c32be40858c2215cb15ccbf54 : SUCCESS in 1h 00m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3356a93bf3b4d4fbff13b86b4ea62fa : SUCCESS in 49m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c508ca966764d66a7c46c344fba4413 : SUCCESS in 36m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20885e1b922f47a68c7f4d37d4f7e2dd : SUCCESS in 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ba4c613c2940a49581604647fbd235 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a963328acb740908222bb24792c68e8 : SUCCESS in 21m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1ff773596ff42549452441e6ced953e : SUCCESS in 44m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf363500bc534062a14f0f31c8999c85 : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":2},{"id":"26ae203480b573c372e69b564c36dc40db921983","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:51:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0314ce96c2074e3ba02465e93d863f23\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f5e8db7cbe7438dbf50f395a0e8216c : SUCCESS in 1h 54m 16s","accounts_in_message":[],"_revision_number":2},{"id":"af9bd6d28e8f06b4ffe3d0045f44fbb712153a19","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-12 00:14:51.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d80f616bc703687381322d310af36386fb17328","revisions":{"0e578dce16a19071b594ea6d8125d98cb4efe56e":{"kind":"REWORK","_number":1,"created":"2025-01-20 04:25:46.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/939611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/939611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/939611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/939611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/939611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/939611/1"}}},"commit":{"parents":[{"commit":"338e28fee15c574a5b45f83b81c2f68fce164614","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/338e28fee15c574a5b45f83b81c2f68fce164614"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-20 04:20:28.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-20 04:24:40.000000000","tz":660},"subject":"ring: add cli/ringbuilder version command","message":"ring: add cli/ringbuilder version command\n\nNow we have serialization version 1 and 2 rings, it would be nice to\nhave a simple way of knowing the difference on disk. Seeing as the name\ndoesn\u0027t change.\n\nFurther, the idea of version if kinda conflated with 2 different\nversions, the serialization version and the ring build version.\n\nThis patch adds a new command to swift-ring-builder which only takes in\na ring:\n\n  $ swift-ring-builder object.ring.gz version\n  object.ring.gz: Serialization version: 2, build version: 54\n\nChange-Id: I5b6615b4646d32408307e47a956602bb94768b54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e578dce16a19071b594ea6d8125d98cb4efe56e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e578dce16a19071b594ea6d8125d98cb4efe56e"}]},"branch":"refs/heads/master"},"1d80f616bc703687381322d310af36386fb17328":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-11 10:00:29.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/939611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/939611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/939611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/939611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/939611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/939611/2"}}},"commit":{"parents":[{"commit":"3f072b3bf2c3239ae96ee7388a7d5cd9e7aed9a2","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f072b3bf2c3239ae96ee7388a7d5cd9e7aed9a2"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-20 04:20:28.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-02-11 09:58:33.000000000","tz":660},"subject":"ring: add cli/ringbuilder version command","message":"ring: add cli/ringbuilder version command\n\nNow we have serialization version 1 and 2 rings, it would be nice to\nhave a simple way of knowing the difference on disk. Seeing as the name\ndoesn\u0027t change.\n\nFurther, the idea of version if kinda conflated with 2 different\nversions, the serialization version and the ring build version.\n\nThis patch adds a new command to swift-ring-builder which only takes in\na ring:\n\n  $ swift-ring-builder object.ring.gz version\n  object.ring.gz: Serialization version: 2, build version: 54\n\nChange-Id: I5b6615b4646d32408307e47a956602bb94768b54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d80f616bc703687381322d310af36386fb17328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d80f616bc703687381322d310af36386fb17328"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
