)]}'
{"id":"openstack%2Fswift~939828","triplet_id":"openstack%2Fswift~master~Ib8ca59850c72aa8445a838a7656d4b3bc818da7a","project":"openstack/swift","branch":"master","topic":"expirer-review","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-02-04 17:50:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib8ca59850c72aa8445a838a7656d4b3bc818da7a","subject":"sq: add test for parse_error","status":"ABANDONED","created":"2025-01-22 16:59:51.000000000","updated":"2025-02-04 17:50:21.000000000","total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"cf62b05b6bbd589466d69b2da802036baae95b04","_number":939828,"virtual_id_number":939828,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-22 18:37:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-02-04 16:57:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-22 18:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-04 16:57:43.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"c8949bf01dfee7493e821cd9360f2096f2a9bad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-22 16:59:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b1d7f783224d5e1379379eec7989d5ecd173036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:37:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab41c1d5e314969aca40cc85151f2a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb9dcadf5e14306a42972378207529b : SUCCESS in 10m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a67ec57ddaf549c3ba6529035afb96e2 : SUCCESS in 1h 25m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/688190a66860486a89c09c7b48cd7759 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aaf9cf5d3c2245cb98963d672ac8fb25 : SUCCESS in 27m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5633a456666c43dbb49e7b865202e114 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b42d36bfd6944f0c96cf48ccdb9e4253 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a62e35319a4693a8bb46c658717ec3 : SUCCESS in 25m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5941e2e710154d5ab76d1448af046658 : SUCCESS in 15m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fe23a397540c4ae19bb584d422be4f27 : SUCCESS in 21m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/be07bf563e4a4097b25699c4aa07fa2c : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/57aec1b12ddd4e7b9504e11bec78bf3a : SUCCESS in 14m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9a9e70330f7c40dfa9ee1f428a8c9b5e : SUCCESS in 16m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d4edd87afb54d0aa8e86e08975030e5 : SUCCESS in 52m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96df63f39ada4fec9ab8e3fa21db87bf : SUCCESS in 53m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b36a2032b3d44771905741d7b8af67f5 : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2584d39f696f4b7580a9126f76de7374 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f6b18656958f4120a66baba85ce1b1e0 : SUCCESS in 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"80a3c96d31460c1807c1b79836e0c0ba6eb2c2d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 21:54:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e363ce891b4aec8ec03c8aba0fada4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56673a2344e74b33a5dbeda4bb1244e0 : SUCCESS in 2h 57m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ccf15235e47ff5a5fd8fc3e8121dbf56e696f446","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-31 20:18:05.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"05c64ea57d9be3fba6be1453a2ec4152013b6128","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 16:57:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"619ef6cfdbf92f30c19a3dee0e513353698b97e0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 17:39:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf62b05b6bbd589466d69b2da802036baae95b04","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-04 17:50:21.000000000","message":"Abandoned\n\nhave been squashed into https://review.opendev.org/c/openstack/swift/+/932447, thanks Clay!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"faa74b0c38114452cf1b0c8935d87f4023a6317c","revisions":{"faa74b0c38114452cf1b0c8935d87f4023a6317c":{"kind":"REWORK","_number":1,"created":"2025-01-22 16:59:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/28/939828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/939828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/939828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/939828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/939828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/939828/1"}}},"commit":{"parents":[{"commit":"782e610b5c045f4f880b15c7abdead11e39543e2","subject":"Object-server: return 409 if x-delete-at is newer than queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/782e610b5c045f4f880b15c7abdead11e39543e2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-21 23:00:04.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-01-22 16:58:43.000000000","tz":-360},"subject":"sq: add test for parse_error","message":"sq: add test for parse_error\n\nChange-Id: Ib8ca59850c72aa8445a838a7656d4b3bc818da7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/faa74b0c38114452cf1b0c8935d87f4023a6317c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/faa74b0c38114452cf1b0c8935d87f4023a6317c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
