)]}'
{"id":"openstack%2Fswift~940059","triplet_id":"openstack%2Fswift~master~I20741ab58b0933390dc4679c3e6b2d888857d577","project":"openstack/swift","branch":"master","topic":"p-generic-input-proxy","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-03-02 22:39:47.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-03 00:49:17.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-03 00:49:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I20741ab58b0933390dc4679c3e6b2d888857d577","subject":"Refactor some file-like iters as utils.InputProxy subclasses","status":"MERGED","created":"2025-01-23 17:01:49.000000000","updated":"2025-03-03 00:50:36.000000000","submitted":"2025-03-03 00:49:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":47,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"940059-p-generic-input-proxy","meta_rev_id":"fced2b5b214b37f487dc9ec37f22b0bdee34b6fc","_number":940059,"virtual_id_number":940059,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 00:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-26 18:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-28 15:19:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-23 19:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-23 20:14:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-02-06 21:00:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-02-21 04:00:56.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"3c130c1e098e239b40fb922db577434faddfc20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-23 17:01:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe8dfe18c74c05dcec8e87f4904740e54c19542f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-23 17:34:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1358092d48c2a1247771201bceb7c63efcf21f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 19:11:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4c5581cc0cc441b8aa3752025ab1dd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e429650d0d44a8b2a33990fb68a59d : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/dec1efce882247409e7b1bc525ca7ab0 : SUCCESS in 51m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/915e4943f31a4384b672f840470fcf33 : SUCCESS in 53m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93e24ce185264e33b021400dd1ebb604 : SUCCESS in 1h 01m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d618e50e4734237a115712086524f17 : SUCCESS in 55m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a19d7c00e724e03ba02e14908341d4c : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7db44155dbeb4242bef46aa4418a5e97 : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dcbcc8c071a44aa0815b7830391433a0 : SUCCESS in 22m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27b0054635a34cceb743b56550261697 : SUCCESS in 9m 12s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99592292f8384789a8f3f6d118eb0bb4 : SUCCESS in 19m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7e67b10fc7347a8b6cc663ee85b5a8c : SUCCESS in 21m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81df414c8e8941dcb7b0c4399f2d08e1 : SUCCESS in 13m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c3806a7c9c2b42948bef03e18512719a : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c5bbe6f224794e15af81cb0ef867f8f7 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4f39d1a98a9488295b89f328d3bd9a1 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/36f2f8c145b64466bc9102d895b9dad3 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fbf6e29a68bc46afa9880bd4ccdb1b8e : FAILURE in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b72b5bfb4cb4815b13a7691630a2fab : SUCCESS in 9m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8ff8ec3f72e4de4a62b58fb26eac096 : SUCCESS in 56m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a532c4170b934327b4cc1a52362940af : SUCCESS in 26m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d54d255f1414824a89332095e9c5093 : SUCCESS in 24m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ef357bd709641f8ad7b768f754ae765 : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7197c5f821344da1a5536b1d37471f8a : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0bcccdae36b349e981bbb5b90db995a0 : SUCCESS in 16m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c788c4a399643869112d5e2cd7ecd1f : SUCCESS in 51m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2063d793b240430d9beec26ec3db5285 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ee24f7ecab900505d42b99c3953c84b33a50dad8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-23 20:14:11.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"49d5379b10c46c1e8e808ca58b0cd078e1d668f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 02:06:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f775c01ce40c491988c29b022e3c3ec7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11e475c6a6054a949326ca5d0e0b0be2 : SUCCESS in 2h 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f3ff62b289b1e1bbc6517e4ef7f097737b54f033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-24 13:02:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b8c853d388c08c686b8fcab2edf79930a7d5222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-24 13:08:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa0454d34e4fc38dbdbeb8be7b5efa2381e21a0a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-24 13:08:40.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"10a17d1ee301fa1780bd3aca13de7f97ba8dcfa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:13:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a2fd231f974977ac73280c61b5fe6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2212cea4cc4f98bd7fb4a85fbdb4ee : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/62803b6d6e0d41139ff2f101e18223a9 : SUCCESS in 1h 02m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0fb1b4d5840944279db8e0be48a4f121 : SUCCESS in 1h 02m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a26d40f28e77460ca2f862bbb01697e8 : SUCCESS in 1h 58m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cd2941b3ea7490caf01ec5fbd1a3159 : SUCCESS in 1h 02m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25b3e3a4293749f0a715233d29b7eb79 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0660c9edf6a4d74ada2560963be61a4 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4603083245f04b119815035c1b2cb64a : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65d92ca6fd404c779f8ce63c77a1577c : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6527d661c6f542e1af37ba354a695d28 : SUCCESS in 14m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f11e986170b48a6922c9b6044d64424 : SUCCESS in 30m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b4c65ee4e5e42ccbee82d5e0ae0afc2 : SUCCESS in 21m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/397f0225630f40cfba0070b35603042b : SUCCESS in 11m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/07e36c1d53ff40878173b2f4eee90658 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aca9fcf74d5d428a84c933dc9a2e3464 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39389481807d40ad9da097c56b6b3bf1 : SUCCESS in 9m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9df8d2fa84b94b60a689580ad094d4a8 : FAILURE in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d5b27548c184064ba8c5229450fe20b : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/018ec6f3b8194ed196e81d682b32d476 : SUCCESS in 1h 12m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/195221df11a04820822052da453151c0 : SUCCESS in 42m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db36f79a141b471e9c9970f1e2e255d8 : SUCCESS in 37m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8b0a7c09d7b49b7916dd5aee5fd0a9c : SUCCESS in 19m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0264424e53e44edbbae0dce9a7a1e87 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfe8b11474e74fb98f53340da178e821 : SUCCESS in 16m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c2f94eb6fce4c3a95ce148541b7913b : SUCCESS in 59m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/499bf7f0034a4a9a8c97c75aa445de90 : SUCCESS in 1h 55m 38s","accounts_in_message":[],"_revision_number":3},{"id":"d8085610283e575d0ef69d6cdbc64260688203ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 16:54:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d76e00da27d4fbabf263381b9c6f575\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3684a13ec0fd420a9d4bea802d7866c2 : SUCCESS in 2h 09m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c525c56491b205b2db641add538c05853d6fb45b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-24 17:22:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fbe4677c11c2b3596d48a6c662336137df4a614b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-24 18:49:20.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1604c9ea7e8fb624445aed2e03a0b8c5a9372c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-28 12:22:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ed1c540f7372982bd8e48aaf3a5d348cb5d45917","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-28 12:24:08.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"531036bf6526012ebb645e6f0221df426c065cd3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-28 12:25:06.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dacfdb08c149001872a3820c6f7f36feaebc86ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 13:46:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79a745a9b67046fba8b8e0e9385e1282\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49bdf8f23b574146aa3ec6dfb3959bfc : FAILURE in 3m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0b293e88baf6426184b938632940882a : SUCCESS in 52m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca868b4c61e04905925c71c75f3fb2c3 : SUCCESS in 1h 16m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3e56f1218c34a81a75aae52a11d2f9c : SUCCESS in 58m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c6fbe1e627a4798851582523524bbd4 : SUCCESS in 1h 16m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a60d9e0a74f4df1b5c9c2a98cb34003 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4365cb628ceb4f659243213245cb644e : SUCCESS in 12m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69b78adc154c442e81ba3f10b6d55278 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7323b40e48604d75b18158a1219f9953 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd02512f3fac436a96a5156fb6ff4b14 : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c7f33a39bc64940ab4c5ae48d2b8422 : SUCCESS in 9m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8cc117c1041e4ec3822d2e633f2c2f6e : SUCCESS in 26m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/51b1ca96a3b44c12a079af65703d043b : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/51c6bab0a9194a0bbecdcab16e6deaf4 : SUCCESS in 9m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd111e62b08f4f10b4bb300c4099607b : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a3546aa994b4f43af60acba0cff4d02 : SUCCESS in 8m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ba77398a19a48f1bd3d34d43f7f5f53 : FAILURE in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/679751e8f56b411e9fd71ce3910d938f : SUCCESS in 15m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94d5987a743549a1acaa2d3774aa422a : SUCCESS in 59m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8428985166d642a9abbe80a7f6c04191 : SUCCESS in 50m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c8b62cb75d54e39bfa7298117514a71 : SUCCESS in 42m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c17ec485fcd45f1834839e094f4d8c4 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2578f15f4fd24949898befdcff1ea2d0 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e74cf1dddde9411989eb5279a760f037 : SUCCESS in 21m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/781336e8bcd1429ca0d5a85a76d1d629 : SUCCESS in 1h 10m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/905a702940524d06b43230a08c5a8c62 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":4},{"id":"3bc7330c67134e35d76bfdf8386c6b5ae4192fb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 13:50:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a002b6a50f458785cc5f53d62e6ad6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b35abda2eaef4937af824c2da66d6926 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":4},{"id":"6b7bc92093f7c3a54d9b27fbf4c3a1cd9427fafe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-29 02:33:57.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"63ef512809df59ba4abcf7f8984f2a235575874f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-29 10:17:48.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2b9df1876f7a3b563eda9db1c52358ef480eb7c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-29 10:20:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ecf869b4fc607105ab540a107d066653ce2a8eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 11:54:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ef992678df4ae6b30ff6c654c08753\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe807f835564f898543304447a0eb12 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/7719f940974b47e9b02323b81bd60be5 : SUCCESS in 1h 01m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50484d8a3f9940ee927c5fc4d7799535 : SUCCESS in 1h 04m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d75797bd3ec42e0ad6d50153fde6df9 : SUCCESS in 52m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73cb59d790ea4cabad91df9d845e503f : SUCCESS in 49m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e333feda87a4783ab924c6cff73f167 : SUCCESS in 24m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ae9f86c76744fb8acbd78d5cace25a0 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6851d536b98b4efc9c581eafbf9065f3 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/178a5b11396a40b88dd4bc05c65fe9bf : FAILURE in 5m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a9a568e56634181be8f29646d9593da : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/802d4c4483294d47b966ddccc2922053 : SUCCESS in 9m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e512d6b6f4a7491cb4b4a271b87edfaf : SUCCESS in 26m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ef4c7d5283984158810c7ff25abea16c : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/236895652c5942c58584d6fd236fd0dd : SUCCESS in 13m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/17f4adc171ac42ff8ffd6d175260920b : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1a8f2954d1d4b1d8070609375534f48 : SUCCESS in 16m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a7bf7b7689134185b5e97fe2849afc07 : FAILURE in 13m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/55ff7de1e1354d12bc43fcec6e4ee41c : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d8199b382a54887a710e5cb1b9a3074 : SUCCESS in 36m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52cd655a3a5a4899bf3298924d050a3f : SUCCESS in 54m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3dce128022049119c15bffba0b16d26 : SUCCESS in 54m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/785b46eb9fe74e1c91afe652d362ed36 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422f02fb2404455ba6b6c9d3e9d4b3cb : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32d7f90f4bf543c3b6bbf8f7b588f3e4 : SUCCESS in 18m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05d38c35a73045d0a8b33d64fa0a6b33 : SUCCESS in 52m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a8df25992a2440885dbe15585328d82 : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":5},{"id":"c5eab00c02505b20576fdf82effb1904bf9549ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 13:23:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea2200f664341068b6f2279bbe69d9e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0426da2bbf9a499ead66893953bedaed : SUCCESS in 1h 45m 28s","accounts_in_message":[],"_revision_number":5},{"id":"33559105db634d8775d141a489fffcf2cfb72921","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-06 21:00:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c04054b6bde127238a156e8d9b3f7e3d7b5a4101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 19:57:54.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"82f895881855746d7e1386919181bacf07196286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:13:42.000000000","message":"Patch Set 6: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89a29b2bf29f4457913c17fcfa961e8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b4200a68064da5a8d559e00ca78710 : SUCCESS in 5m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/80c8922c15504dd8be0d96414c5b4546 : SUCCESS in 36m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34d37281e9394f06a817724abe637194 : SUCCESS in 33m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2dc7ce9abc884f8b84b6a7516915f70d : SUCCESS in 43m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/567dd272c9a648f2ac73b083c2147a87 : SUCCESS in 1h 08m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50a5fb191a724e618db52450e1dc6721 : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50c4810fea2140089f3aeaec3a11a176 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5a5c11e112a4460a4b0a764cd0315e3 : SUCCESS in 22m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/043bee88f8da4101bf87c46879e17db0 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd382726a4014760921524c75c6c70d2 : FAILURE in 18m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d3e78fdbd43f48b7b9be1bd537763c8a : FAILURE in 19m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2c7a5dd9c17c4ce8a936afa885e5261d : FAILURE in 33m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/afdfb433af5649abacd76f8b4aa3e627 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7833137cd909405ab3f8a3203a65c641 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c81409af791f4ad6addc9f9c9ddc18b1 : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99ebce0489f44c16a45f0bc7ecb7ae99 : SUCCESS in 8m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0d7e795ba7b4c3e9384a9e55b6660c4 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e913f5b6bad4a50936dd02d4d9f4187 : SUCCESS in 9m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2b794ab7af84e2e8027eafe59f0ec22 : SUCCESS in 58m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a9e3e3f274a4029aaf141705960287d : SUCCESS in 49m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc96876c724d40e89a37b7dacdbadcd2 : SUCCESS in 48m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00d7729948b64aaf9351ce8e6aa0e0e9 : FAILURE in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03143202171a4820a48a652875874e6e : FAILURE in 2m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf80f8d8464b4d74997fa797354efed4 : SUCCESS in 19m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cde6b9a05b14465e88a622e7fee5c0a4 : SUCCESS in 39m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d6507aada974ec2b7385d71dbf23263 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":6},{"id":"f23d0175420c1e1e3bf49f01e2ba334faf1ae6a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 22:50:15.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8606c0cd4d504b9ab1cdc7cd4f7ab15a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aeec1536ae7e4e49a16f064fe8ae9ccd : FAILURE in 2h 48m 07s","accounts_in_message":[],"_revision_number":6},{"id":"05025e252525db050b43d83240a89138a250aeef","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-21 04:00:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7dd8fa34de54c6cdca1480463c13db953106c944","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-21 04:53:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5adc750eb75651925f11e9de73adde0603cb9b3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 06:19:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ef82fd791a4681824d14dc3e60e9bf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1643c67e3fe46fb8d80c1de77708cb3 : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":7},{"id":"dcff68771c52a1ec6b98ab35b0712e2f28041e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 06:23:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a11dc673d33243639bc0d87c827fcb9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b777058466c4c1ab092c4b92bb60400 : SUCCESS in 4m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/500cd8b183f147a7a47ff65adf7f0678 : SUCCESS in 57m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9c7e852f4944c1b8a51c261717af9ec : SUCCESS in 59m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b570c8b7c4224c0485bff3ab629352bb : SUCCESS in 55m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85ba05706e594c57b63dcc3d8304994e : SUCCESS in 59m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40522ee3841d40ee800b51f825f4c48d : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f32568558dd49a3b39eb84aa99cf9cb : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41c1d622af1d4806b4233ed60fbe7dd3 : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70cc27414795490c9a0944ec558a8148 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b3fef000ff084dfbb4b811fdb99828d5 : SUCCESS in 20m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35bceda3439c49f0b4e05e5187135a49 : SUCCESS in 21m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/70a4ef36af45406fa03da1b266bcfac3 : SUCCESS in 20m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8143dff4ebd84e038254d66721f7d235 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b7544cc732d54a1c9d1d0404fd26af4a : SUCCESS in 16m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/20342328922741339812539d167f4ca2 : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b44937f6cf34befbdaf7a5419b5b361 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d99e34de23614db8b63821253938e595 : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37f1baa29e154ba29b1cee502f3f74fc : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68865e9c13ed4760b51a5225ae3260c9 : SUCCESS in 46m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c2bddfae7044d38a1b48996de47b0ae : SUCCESS in 38m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02902e275f754ca28ebfe748775d1278 : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01f0fee8577d497aaa7c40425ca441ef : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659f9583cab94a2eb75d364416fe293a : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/936d26c4a2ee47f79bf8d5bf4153ce52 : SUCCESS in 19m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44d513c35874466697cd4613bca6c50d : SUCCESS in 36m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a48162fb90784a61b4ce5476cfabe7d9 : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":7},{"id":"30b1c9dcb1ce8c156ad67993cf9b3ab9100b2338","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-21 09:00:10.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ebfab6874a0baa877a965248c9651d3f97d83c1b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-24 17:03:38.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6975c6b579aea796c264c54b2813f7056c4e57fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 17:04:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8991ee41cfb21b292f0d49c7a65078c4a705680f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 19:50:27.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb1230ec6d854c68862c07150ba7c962\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f817cab6ce234df39784e564f2071936 : SUCCESS in 19m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2049aabd95e4404ea8b666e9af52d56e : SUCCESS in 1h 21m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb7b988740f34db29c8108b68c4e08b5 : SUCCESS in 51m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c72386c38f384e05adebe6c882287e40 : SUCCESS in 44m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41800c100a164e14a2f4473bad59c85e : SUCCESS in 50m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0ebfe7cf76304d259e1940dc9a4b7ed3 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddf937c4851447b099127005872d0226 : SUCCESS in 18m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/db47d350d0c344f29e2d809badaca49f : SUCCESS in 19m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/382aaa50a5874c88bf531f64ebd73dab : SUCCESS in 13m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9c70ce57e7aa447e95d28ff7f6994d1e : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f6ab0eb681cc49d48069e35b93400778 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a2efba7c1e1c47ba857824aa9722f8f6 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/177d92e03bd94254bd7bfc01f6509301 : SUCCESS in 9m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba7d5186404e44ed8ed1cbc4418d0dc6 : SUCCESS in 11m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5ead55107b54fc29abc782d2a278faa : SUCCESS in 49m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce6734d164914970ae75150c4ff201bc : RETRY_LIMIT in 10m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b9bc17ce8b04eb7b01855780afd65a8 : RETRY_LIMIT in 19m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8162496212ca4ceca436d14840403260 : SUCCESS in 22m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d653a54bf6463b9ed381e24c0a2aaa : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b452ed0edb9f4eaf89e41b9f0291bb7d : SUCCESS in 22m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d21fecd05bc4fe3a306921752371d61 : SUCCESS in 1h 06m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a225fe3dcba473294c394dbbcd226a9 : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":7},{"id":"5d9a544749b0786a147ab03fb9ee4c3bd3e5b2fa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-25 10:18:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bfe593aef5a323ec175fc2a3874e496ea4fa2308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 11:27:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40ecb7ffbed24af8b0d80e2aa3300780\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd656c436e249a283addb018eb4a249 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d5430fded6c64243a8d89f0b31180fc3 : SUCCESS in 50m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37a83dbb3e1c4bb89a38b78d8ddb813f : SUCCESS in 52m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e8404d8bd38442bbf92eb061030a4ff : SUCCESS in 41m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e52202d10d9c437b9fa71233b4a40d1a : SUCCESS in 54m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5db695e4a0764e1fabe84a89660d48f5 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ee9e54bbd304a028f92d682433d3506 : SUCCESS in 26m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/714533dace95453e94cdd9e0b1b0541a : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22668404a0724cc991fcca15a836a77f : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b66941dc7e9f40239adaf32da16ca1bb : SUCCESS in 20m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46df827ef448451ba52eecf2b531d082 : SUCCESS in 31m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b31887659cae469ea0c382e982d20f31 : SUCCESS in 13m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f92ac755d0364b06a6847a5810b2e9e5 : SUCCESS in 9m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2b1629f3b35406a807d886080645cc2 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a677a7d0e85b456f937493b6f83ad982 : SUCCESS in 15m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbd829ef35be409a8ac5c2f5e15f646e : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c190f18a003449d4b3e14b3ca51a5764 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b41277e2cbe47959340d8db506b571e : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e866647661c4d4e90bf2aa69c0e1be0 : SUCCESS in 53m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f00b7d17977f4beba554d5d2c33e02c4 : RETRY_LIMIT in 9m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee5984e1690d4beaa3385ff7cef39dee : RETRY_LIMIT in 9m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f54b7e1addc342e283260777458d28d6 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab0ac22c3c74c96aafa044ccba66145 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8a64fbcf7ce4559af7b79ae238c41b8 : SUCCESS in 23m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c29ff1d127d48d983d45b4b1410636a : SUCCESS in 54m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a8d428c1ab48c7a3f0522372d68cff : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":7},{"id":"d52039c92a410e724579d0b9796f1be768d92138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 11:52:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e3664d203ba47fdaea1e621c21eb5f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/009c2d8d0e094b6990d4b2bca0c18e60 : SUCCESS in 1h 30m 41s","accounts_in_message":[],"_revision_number":7},{"id":"01ecd1bd92ecc6bff6e23e2bc391deeeb43af328","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-25 18:32:03.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"0704b97f94f5df311a61d5cce8a19a02bac431a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-25 18:33:12.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7132dfd4760770e898460c1a67f6b9102be7cc73","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-25 19:38:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bc4852c472bf19cfca8f789fe36bf5812ec5d731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 19:54:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b085e38177a244a8b5018585b2c4f510\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19ae8a46b6048b5b71750d4044cdbcb : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3335f5552a384b7ea939fe4858ec43a2 : SUCCESS in 47m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b5c6b0ffbb9c4d9ca5b25f746a7e0aad : SUCCESS in 51m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9390f65568274c4eac55be567c778ed5 : SUCCESS in 52m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27d3ba3b05e64097ae9a5b91337c08a5 : SUCCESS in 47m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0f636963411485bb6e279cf8a086673 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd59780bd7a44a629a3f3549eed72bc5 : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e11dab568f8a4b3ea4f06255ae3d2785 : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1df50e3a09c94f379d5e6c03d3c91e19 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97abbb42f3de4a659e587cd46c120ce5 : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2da97e6272ca431fa35713224e4a205f : SUCCESS in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e405e021f2544fa5a99364a3ff468e3e : SUCCESS in 20m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e569da7788f44e40906f5ba810fd93f5 : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/827977e00c734ce58b258f8438d92163 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/052ab53b80204145b613f846b28cc139 : SUCCESS in 9m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6d2373917f284ad6afc6a266dc1d5bc7 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2653da8ea1b438b9f2c6ec27c890fb5 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae7bd90838104badb0cdb514e1d33c84 : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3649cd55ee61457da628368cee275945 : SUCCESS in 57m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcac3e3b8e7d4ef2be1bfb512a28ea95 : RETRY_LIMIT in 14m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d95756b27644b1e9fd6cb5dc1ffbe28 : RETRY_LIMIT in 10m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c898cdebf93045cdaa82b6104304a16c : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26dad19f3adb4e6aac4b60ac96e37194 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b06588370810472a8d711992fe4cd0c3 : SUCCESS in 19m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/820b36275036446baeb83e705c5e18f0 : SUCCESS in 40m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43ccf282310e4e1fa439a782c135527f : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":8},{"id":"f9679dc81c822f65b90d32bb0fee8268c9915523","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 20:25:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9cf8bd6c5b94b5f82b7c22469e537ce\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/419cece37aea46d6b505971f46ace5d2 : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":8},{"id":"af6c7591734c413f35cc62cdf33b3af60890e177","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 16:36:33.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6d4021b5386c42690d584e599aa7ceef985bf032","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 16:39:06.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b5fd2e4b608eb0728503bcab426306d62e91ad4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 17:40:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a450179ac0d7408bb1ecab73e73bac08\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e5f9e9bb88d42ae9b66122f57ad36ac : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":9},{"id":"44d3349456a722557cd2682e1ec500f81fb1c796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 17:58:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/387d0bcacc664bb0b2c159f4c92db490\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a9091c070e848f88027702b973c5135 : SUCCESS in 21m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6adcb8b4085f41ce96065fc10d684573 : SUCCESS in 1h 01m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bbaf573aa3814e048390cc8809b2aa3f : SUCCESS in 40m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24d11b52d1bc405ca9762924f71683c7 : SUCCESS in 47m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22e80aba17ea4187bb9cefba323bf9c2 : SUCCESS in 57m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e2b5073eb2f4a668e1da112c5c17b7b : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/affaf561e8bd4c4eaacff65daae0222c : SUCCESS in 27m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3b46bcad51e41b7ada42e38a20e28ac : SUCCESS in 27m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6641f0a60d6f4beb96019bec114dbd8d : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bfa1a9251a64b609dc1c0301ef2cf69 : SUCCESS in 21m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e865ace6a2a94f909b974d67f0b1b67c : SUCCESS in 20m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3b561008b684d609842b74466213d75 : SUCCESS in 32m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4b35dc1ca2754d1cb92ac7cd79a5b75a : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ebd553c3b54e490a96705a5600c02c50 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a4802330cbd94345b7456e9bb4dfb278 : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b353482966d2416dbfc69040e6923b9b : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/972904cabf3148b697c7227bc0a58d2d : FAILURE in 19m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ed14e6421983460b99cec205ed6331f6 : SUCCESS in 8m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/60a0c0169d0f4380a799043d31d859b6 : SUCCESS in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1f084aea3a84ea3aeb4c0d06601ad77 : SUCCESS in 51m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78e6c9378cc84ac8b648debccfa5b62e : SUCCESS in 35m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17bd3302cd324a5482b27a3a71458e59 : SUCCESS in 23m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e384154fb1464084d7ad2815967c83 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92e80ab68ce24230a7d3de893bb3c873 : SUCCESS in 22m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f46b4552b0046318c732d321afdbdf6 : SUCCESS in 1h 14m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8392f4f052b4543b1406c6801577dd6 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":9},{"id":"8c96f71664beef91b4d93bce5288a19c6f29fef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 18:40:51.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":10},{"id":"43e49bcabd21c08c6983323fa8e562a5d242d9b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 18:41:16.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a0d088fa304c4b5c61b95c6ccd18d9310adb8199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:33:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/072c92650a4746e2ac6ac3aae71b1a37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c9666ed8e74bf89582970ddd83eb58 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/4da635deb8e6459287198e578e785170 : SUCCESS in 1h 47m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a5dc38e807548c8af8e65468739fa95 : SUCCESS in 1h 43m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5d87c352c5b455993c52101d768d73f : SUCCESS in 1h 43m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60008033ec984c2ea4c7cd15b4e4a8d9 : SUCCESS in 1h 01m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fe0725e7e8247809a4499687962ffd9 : SUCCESS in 25m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8db99e62854450297a64fe101efb3b5 : SUCCESS in 22m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eca957d28c914973bcf460628fd14920 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/09db9e606aa146bcb99cbf95af644ba8 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc26484cfd524c5cb0552c824f04d2ac : SUCCESS in 21m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ac704246d6b466ea994fca221b8a276 : SUCCESS in 22m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1c5325139b6b433e94fce97dcce8a733 : SUCCESS in 33m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/48b0d7b05c634c4cb5028fbf211dcf8d : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5474484bc80c4c6fb50f02f2ff7713b1 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cbfafa40423a4503a24224a9c12cf88e : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c394ea103d9245a1bf34a924e4d240eb : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70579afd19bf434bb2434f7277d7b949 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ed2b8fa8f1204bf79c03daacb2e10580 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4e43fef54bd40ef92043243d7b1e968 : SUCCESS in 58m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/858a4a0f719e426f913e6f588fd788d7 : SUCCESS in 49m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/074020ad456a40319ed32375bb427ad0 : SUCCESS in 1h 06m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cff9870997234593b8a7ae5773c00b6e : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6466d05468a344cb9fc384dcabfd81f5 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7477413ba59041469a2ab43009cd9887 : SUCCESS in 21m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/955fab40da204690a6b0121cf9237019 : SUCCESS in 1h 04m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2705c9a35d9d42a8a39015efde40e707 : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":10},{"id":"782a2c9ab7d2b6bb35b6716b822fd23c7ed4e8b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:35:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f8cd56fb81e41a0bf95e9b31700fd96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31b0819ad8e84f0eb774ddb41deea8a6 : SUCCESS in 1h 51m 00s","accounts_in_message":[],"_revision_number":10},{"id":"059c4f0a5b95db2b5b4fb80449594aa6ad763180","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-28 15:19:44.000000000","message":"Patch Set 10: -Workflow","accounts_in_message":[],"_revision_number":10},{"id":"d589e24c75bbbe3071e6cb84d3ba57d9327f5652","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-28 15:19:49.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a3008404360ad911154519f7a946c90841744a13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:21:12.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"c6ccdfe03c3538fe6bc0bebbfec6a4245b4bf623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 21:11:47.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9784060d73f14f36b79063e827b6a542\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb022fb84714095a0bc7b14af0a769a : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/9a9bce067ba34c5c916bc8d28aec9311 : SUCCESS in 1h 05m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e58b6fcf52804f529936d1e4085c8e45 : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d902b4a681b74600915e4a9fb9309744 : FAILURE in 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e6c2e97169f477db9415b6555f86c8d : SUCCESS in 1h 09m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3e65e7cb29d94924bd751a22b74af692 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/31e62440e86e4221821b14527491504b : SUCCESS in 18m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/817e16b6f0ba4f63b406cd6300847a1c : SUCCESS in 19m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/21296c33d2984d7298ebb4dbde33a8ee : SUCCESS in 31m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/29d73440dc664d9cb5154bbc6b1d7b14 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6c664deda00445a82c2f9ab3556bac3 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/53b2c6108c26419a9f1b58e31cd28557 : SUCCESS in 19m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17c69d6e895c4e07ad9f26b2bb57a9e5 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d7255d41457d414d93c2ab55129bd13e : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40a950799ab640d6ae8128752197ecb5 : SUCCESS in 38m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de68c031efbb45c290516cc3f40d77c0 : SUCCESS in 26m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54aeb70fb9574e7785132579f70e34e7 : SUCCESS in 26m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56010f29af094777bb778241ac00a7f0 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ad520aa6b34a87b6a04aa51a3f83d5 : SUCCESS in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3747a200a6e34f98affe78414167e299 : SUCCESS in 11m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ec2785f38204f08b6e81ec43e9a8410 : SUCCESS in 56m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0992ffa6f8b54a3a8ffcae01fedc3eca : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":10},{"id":"b5c0ed051c38b24bd28d9d4d29d281ce2bb7de9e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-02 22:39:47.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"216abaa5cf9dfc34400d3ae94267b5ddc0c55c1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 23:37:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a0ed8c857b54cbe83fca8440a4abc38\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c145310828d48b5b654254118b68afe : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":10},{"id":"71770647903c7ba40187f6f84eea791749b871cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 23:48:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88f04dc35bd145ab8a734bc6c7ad00b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21cb798aa02455482db9d16bcc15f6e : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/55c07d4ff0ad4de099a72ae48e17d294 : SUCCESS in 57m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8940f663aa9414b94b9b15877eddaf4 : SUCCESS in 51m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aff0bf42e8a9473caec35ce5a5f8ebe7 : SUCCESS in 52m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44f52a4fa1b74d9f9876d2cac4332aca : SUCCESS in 47m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b2b0937032040b9bd9d79281ba7b8d4 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5377d27ef1804efb99a43d12b930f869 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4c5f1054fcf436ba5acc7b5c5afda3a : SUCCESS in 18m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1401eb431f9449f485c35afee4781f6e : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48dfac1b80da4d5cbdbd9b593b484e6f : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c9957a5d3334e28947d8832aed658ff : SUCCESS in 19m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b0e3989ecc474a16a8964d5e1e704ddd : SUCCESS in 28m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b4b956c5808e43eba4150b0794344f64 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9d0ec1b6c8bd4b3bb8db7043a6fe86bd : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1ba1add648ef4fd0974d5f8c94dcbec4 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4498af8f415d4b5888f55fa239ad8b81 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5eee2c5897341bc984bba749c510223 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/324477f851934b52a2420bfc95419797 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec53300424db44dfb67f4b0fde72e842 : SUCCESS in 58m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c5af9180bfb4bd48d402259bb39bf55 : SUCCESS in 51m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb843e8fed114e94b0472bd47bf32754 : SUCCESS in 51m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03a6625154b04e299e96fdc955e73be4 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762c3103e0aa494c9a3366a1bb2c1801 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9955c9ec3254a6eb0cf37aefc9c810e : SUCCESS in 15m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac9e9f61cc6d4a43baf10afc1ee601d5 : SUCCESS in 49m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5ec47209e4b46419aff59ac0f97c47c : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":10},{"id":"5afaba1d303d70e65a4642ccea2358734a26ad7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 23:49:27.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"6e5191714b07750817235ef7a8428f5e5095fbf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 00:49:16.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27071ad3716945cca165529e08358567\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f6fb032a274d66bbceeb5465efe137 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/d5dfeb88c6164611b3f85ac233249512 : SUCCESS in 54m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b821adfe7a24f6fac84ec9fd00d9e00 : SUCCESS in 49m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd8e902ad3d441c3853af6e3da3c283b : SUCCESS in 59m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2ccf188e81d4fe69ae1974537d28853 : SUCCESS in 54m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4863d8bc08dd44418bb521ead844dfe9 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/feaaf2cca4f14197beafc94f47bca9be : SUCCESS in 23m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b84cdec34de4666b4e661b152a9d79b : SUCCESS in 25m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dbdb37c7767b439b9c6015d1e48ae735 : SUCCESS in 13m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ef3e2d1a1da542069b130cb19ff71c31 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/57ea93ed83f5452baa6d8f8b5861ca59 : SUCCESS in 9m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/71e35a61744f4a5abd66beec0827b9c2 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/26b9ff00db8d4287ba78f78c03d078d1 : SUCCESS in 7m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b499881fe7e461d8e1a53731cc3ad19 : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7349cf8d183428893e4d71806cd5ebd : SUCCESS in 57m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44f2281b170f4cf2992fd314a81b9491 : SUCCESS in 49m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a487eec6cf9e462f952afef921ec236e : SUCCESS in 36m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/177e473b1f254d17915d7eae909ad778 : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3492cb373c174214b4af2da2fcd9ee7d : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7f05bd1301144faaa59d219227322d6 : SUCCESS in 18m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13c69745b633441aafe1817ec38e0c3c : SUCCESS in 54m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b86447caf07e45dcbc9e2819e0a38872 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":10},{"id":"c3cac206ad1802aa29f8c73a4723ced6eb5ad6d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 00:49:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"fced2b5b214b37f487dc9ec37f22b0bdee34b6fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 00:50:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f3ee81ac284383a67a8bee5ac381d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f1949bfe5e40efa36af15dd71bee9d : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2a05c813ea9b471c864061e8b7b7a6f6 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e4cc228ed09eb30a4e186515f801d9cde0140c4d","revisions":{"755bbfee2fea360a32da572fdafbbe87c49795fa":{"kind":"REWORK","_number":1,"created":"2025-01-23 17:01:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/940059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/1"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 14:43:47.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/755bbfee2fea360a32da572fdafbbe87c49795fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/755bbfee2fea360a32da572fdafbbe87c49795fa"}]},"branch":"refs/heads/master"},"7b5d37474d49aea77d75485b09fed1246e74daf8":{"kind":"REWORK","_number":2,"created":"2025-01-24 13:02:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/940059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/2"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-24 13:02:18.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b5d37474d49aea77d75485b09fed1246e74daf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b5d37474d49aea77d75485b09fed1246e74daf8"}]},"branch":"refs/heads/master"},"ff8ec22d9591fc5d9a8ce22b9c15e471ecce2462":{"kind":"REWORK","_number":3,"created":"2025-01-24 13:08:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/940059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/3"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-24 13:05:42.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff8ec22d9591fc5d9a8ce22b9c15e471ecce2462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff8ec22d9591fc5d9a8ce22b9c15e471ecce2462"}]},"branch":"refs/heads/master"},"7c9b1b8f6d1318d5dc394bca8a1a2130df0a14e4":{"kind":"REWORK","_number":4,"created":"2025-01-28 12:22:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/940059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/4"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-28 12:22:27.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c9b1b8f6d1318d5dc394bca8a1a2130df0a14e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c9b1b8f6d1318d5dc394bca8a1a2130df0a14e4"}]},"branch":"refs/heads/master"},"f46aba1cc5d462d6d40654a8c8354aea9d99e676":{"kind":"REWORK","_number":5,"created":"2025-01-29 10:20:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/940059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/5"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-29 10:20:16.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f46aba1cc5d462d6d40654a8c8354aea9d99e676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f46aba1cc5d462d6d40654a8c8354aea9d99e676"}]},"branch":"refs/heads/master"},"e64f2ead8e2fee4bb6bfd3ff013642466966d38b":{"kind":"REWORK","_number":6,"created":"2025-02-20 19:57:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/940059/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/6"}}},"commit":{"parents":[{"commit":"97cc335d545a745b3b1df0ededf6179937eb2a46","subject":"Merge \"tests: Remove a couple more py2 code paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97cc335d545a745b3b1df0ededf6179937eb2a46"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 19:57:04.000000000","tz":-480},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e64f2ead8e2fee4bb6bfd3ff013642466966d38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e64f2ead8e2fee4bb6bfd3ff013642466966d38b"}]},"branch":"refs/heads/master"},"a75329b4205e37938c15e40c7ab8b4508487762a":{"kind":"REWORK","_number":7,"created":"2025-02-21 04:53:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/940059/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/7"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-21 04:52:34.000000000","tz":-480},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a75329b4205e37938c15e40c7ab8b4508487762a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a75329b4205e37938c15e40c7ab8b4508487762a"}]},"branch":"refs/heads/master"},"a0d101113c414e841abc4fb041a5d06a3f7f3790":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-02-25 18:32:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/940059/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/8"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-25 18:32:03.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nDepends-On: https://review.opendev.org/c/openstack/python-openstackclient/+/942721\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0d101113c414e841abc4fb041a5d06a3f7f3790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0d101113c414e841abc4fb041a5d06a3f7f3790"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fa22e0c08429ecc9630a22e6709303ed331ac8a4":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-02-26 16:36:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/940059/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/9"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 16:36:33.000000000","tz":0},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/942831\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa22e0c08429ecc9630a22e6709303ed331ac8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa22e0c08429ecc9630a22e6709303ed331ac8a4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e4cc228ed09eb30a4e186515f801d9cde0140c4d":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-02-26 18:40:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/940059/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/940059/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/940059/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/940059/10"}}},"commit":{"parents":[{"commit":"7140633925dd6f48eaf35009fc5f2c2689dc909e","subject":"Merge \"tests: Allow more configuration for S3 cross-compat tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7140633925dd6f48eaf35009fc5f2c2689dc909e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-23 12:17:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 18:40:26.000000000","tz":-480},"subject":"Refactor some file-like iters as utils.InputProxy subclasses","message":"Refactor some file-like iters as utils.InputProxy subclasses\n\nThere\u0027s a few places where bespoke file-like wrapper classes have been\nimplemented. The common methods are now inherited from\nutils.InputProxy.\n\nMake utils.FileLikeIter tolerate size\u003dNone to mean the same as size\u003d-1\nso that it is consistent with the behavior of other input streams.\n\nFix docstrings in FileLikeIter.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/942845\nChange-Id: I20741ab58b0933390dc4679c3e6b2d888857d577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4cc228ed09eb30a4e186515f801d9cde0140c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4cc228ed09eb30a4e186515f801d9cde0140c4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
