)]}'
{"id":"openstack%2Fswift~940088","triplet_id":"openstack%2Fswift~master~I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55","project":"openstack/swift","branch":"master","topic":"no_dev_id_1","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-06-11 19:20:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55","subject":"Add magic rules for swift rings","status":"ABANDONED","created":"2025-01-24 02:23:03.000000000","updated":"2025-06-11 19:20:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cba0b4e01d2f97763ab036bfd39e2d9bf0ba1be0","_number":940088,"virtual_id_number":940088,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-03 07:00:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 04:00:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a403ee9f612385368f68e346831f60b989091c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-24 02:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"210bf731b5ae4dbd93a001cc3af3f36fb894bce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 04:00:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb0168c2eb942ff8115939c60600fa5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a80e922b1c40f0a2bcf44ec711f515 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e4cd1d720ea747499827d52d5d73c19c : SUCCESS in 57m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e69744ac46554b25b87b191d56aed332 : SUCCESS in 1h 06m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d62d3b05342c4f3685a809721bf36865 : SUCCESS in 57m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85fa553d62c34fe39465c3779423ca85 : SUCCESS in 54m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a87136f8a6dd4452943d8332711f0604 : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/471eed390cda4a059e716746cca2dbba : SUCCESS in 29m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85b6bffb113b443fb9bc656a7d8b2835 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a700fd32d6f942f8b14014fb4fd5ac21 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fece460c942d4f958adb92d8813d1984 : SUCCESS in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3f0849527f0549fab66baa3edc2de540 : SUCCESS in 19m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45f64a0450044b308fd13598f769e1a5 : SUCCESS in 31m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8dfdaa45e6f242d28cd6f1159e19857b : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/de5e687d6df34bba938e68717983d85e : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/443e6a6d71bb4b908374a34fc5762edf : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b56c12fd3294140ad0c063356db93c8 : SUCCESS in 11m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89f421d035a0487f81cb3bd135d1bb50 : FAILURE in 15m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b6aa5577a0c8453a9407616df0d35469 : SUCCESS in 5m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7704178ce0b4dc29383f04e1e915cfb : SUCCESS in 47m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b6b3985214540ce8907074dea5105ff : SUCCESS in 37m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed9fff7ed3ce436a85dc81e485360ea5 : SUCCESS in 41m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da22205caec4438e8700457438e28851 : SUCCESS in 21m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cded24926eba4204814349a2f2e4cc93 : SUCCESS in 20m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/837c2fefcda64fd1b3caafe1c568da22 : SUCCESS in 56m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9be21e38397b4b6ab6a6fba4f017a44f : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d46eb9158466fb853fa3d54a1f86bab0fbf91e9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 05:04:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce46aed414d341ef918aeff733f21b85\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f1e85b74a004a1b8d32d68c29f23b5d : SUCCESS in 2h 07m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9cc9e43006a5978e6c8c9972b2f28ab5177d0200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-11 10:00:29.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0c25ea18ff8d86a05e28d1b1de74da1432e7a0e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 12:58:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb4808bbea44c76acd752f5c3158283\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84071e3446b74e17bc1abcc1a86d1d93 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8856aec9369c46bca42728543adefa11 : SUCCESS in 47m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ded2e2fb8d62458187f3be192c5a9d12 : SUCCESS in 51m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/621c385faff444f3b26cc3e93d63775d : SUCCESS in 59m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0446f486ac914ec585d1d4f194ef92fa : SUCCESS in 1h 05m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/149fb6820f0641c9a12fd8c35c6ed873 : SUCCESS in 24m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92baf079ecfa49498b29af747b75d2f4 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adc61e0f38b54225a43fba191f1c3f30 : SUCCESS in 21m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77a68b1ba8cc4275b6ac7b50b33dfd49 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80baa65704864922896f9456d0494281 : SUCCESS in 23m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f0be12c7bf642f080df560a8aabf0d3 : SUCCESS in 20m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f76a4e02b72241af9a5a92b7a7d3fa1a : SUCCESS in 24m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5076046e6189495986918e0d1dc93f1b : SUCCESS in 21m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aaaa30ce102b48cca58817c24509325f : SUCCESS in 21m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9c5cab3202e0478e885716772e96aae4 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78dc2ffa1adf41a58af95a85b0bee41d : SUCCESS in 9m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f4efa66997784fc398bac2389bde2960 : FAILURE in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/30ca88724eb14c7f8a7298aa381de278 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cc6ab40161e4c0f86cf5af3d90e3591 : SUCCESS in 1h 00m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/018096ad7bc94d8893c15266045b09d1 : SUCCESS in 1h 02m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7a4a7fb205148219def4d873e1d75e3 : SUCCESS in 34m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aefd5b568d454c49aca1b86f648de9e3 : SUCCESS in 22m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9966f31fea17400e9c2beb2a0e02f347 : SUCCESS in 29m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1beb5d1241354fb1a505bda0de4490b3 : SUCCESS in 37m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/140b395996dd4a5fa6a98390991514e0 : SUCCESS in 1h 39m 32s","accounts_in_message":[],"_revision_number":2},{"id":"8848123b2acee65db88c11b5b519ade790c8cfb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 13:08:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc988bb40cac4968a83e04a647df0168\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5246a2536c946eab97ddc483deefb9d : SUCCESS in 2h 08m 36s","accounts_in_message":[],"_revision_number":2},{"id":"727c3b06990090acb76fbc156b48f118a0ff07aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-12 00:12:12.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2cf9b9554464db593d4cef6bdecdfa5cf8fab633","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:54:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ea8bb9689e49c9b8bef8f7eb0d1481\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6202cb9c45ac4b29998d04cdddea50fb : SUCCESS in 1h 39m 37s","accounts_in_message":[],"_revision_number":3},{"id":"5fec5a744688ec5ffd0317c5f9b9968c8a8d8dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 02:00:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ab3a6934884f20bc2f7c3ac1d2b4a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95e3e1efa7d45288c77642d1a3cc693 : SUCCESS in 5m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f3dfa78b17a64fed9e072048bdfb5116 : SUCCESS in 59m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e58b1025c427405c81d39456e34e59f9 : SUCCESS in 33m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3fe85cb8101a4e1aaa839bc62fe8340e : SUCCESS in 47m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baa8bd2ea9d54ceb92337c8ad87b85e9 : SUCCESS in 53m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a5b502f222844e6900b91be2ea2b281 : SUCCESS in 12m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57c1de292f974fb7aed4ba2f84a89c4f : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a2221da0c804d61aa5e7f3d7371fd88 : SUCCESS in 21m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83ffaa966bbc4546aae60bdb98079741 : SUCCESS in 4m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35cc99639e004819a0465156bef693b5 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f6d50ac48c54cc2b4380d5bf7f39281 : SUCCESS in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/648384a90e834658bb1f32586438845f : SUCCESS in 23m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/72acad95879e488a95895cb7eb4458c5 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0615e850f68048cba369208ced2fa1c8 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/513ebdce62694b1299835cd8ce7f3181 : SUCCESS in 13m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ebf7db728182445b9bb51d8d8e6ca4a6 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15b9748f1d47432b847d79d6c768dcbf : FAILURE in 13m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ca7614250b5d45a586d4c54a8d6b9047 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3323c4a80ec4048a870f5160d78dbd3 : SUCCESS in 56m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21ee45b1f9d046b18363f89f57b4ff4b : SUCCESS in 50m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ece9a92ea60c4ad3aaf518e4c74a855f : SUCCESS in 54m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0fbd04f427e4d5d8e5b196446190b4d : SUCCESS in 21m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d0a761e8e2f4f82b4b12038c897efa5 : SUCCESS in 19m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b958dd26fba4e30b5929eaa2959be29 : SUCCESS in 59m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b122ab8dfbc54e1a8befd1fa02ae9d03 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":3},{"id":"732c4f09f91fa4d3df3b758e2d01e431f39e53fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-28 03:15:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"39f8e47de7d005fba07df004412be5703c8ae582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 04:50:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fdabc04751545da87c8818851b583e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ffd9a755e24976be0d9803ef829801 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a9b923759da9401e8bb20079669ed8bc : SUCCESS in 1h 06m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43197886e7c0478fbfaaa2a415d33044 : FAILURE in 50m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0ea7b2d5b6a4f5baa426b82fa466388 : SUCCESS in 54m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f9f3ca6e19a4a83a65edfc63bd009f0 : SUCCESS in 55m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92b95fd4e2d344c4ba7e09ec5d74a866 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/080b053431ab4026b495b9b15a975de0 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bd0187b29544733aed3e164644d2711 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b274edad6a984f34841b216b4cc5e473 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c916fc1907d6468cb1847b831a2d22a4 : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41d06b79dfd04eaf842164495766169e : SUCCESS in 20m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49e8a72a684d41e385e28831f712f127 : SUCCESS in 35m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aedad1b57d8a44c890356e34f72e58c9 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2a0b0c35fe164a619fc538dd678d1467 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a565c676c0e842a69938ec95f15eb2bf : SUCCESS in 18m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0065abc1770d499d9361ec48a3b4ddca : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a24b5846d794737bf597e8e778bcb27 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57a65574db804fcb8f59018894c6d693 : SUCCESS in 10m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16cd0d59d057432e99d8ffc551305dd5 : SUCCESS in 54m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/346fb84dc83d4efd83ed35a09cf7117e : RETRY_LIMIT in 23m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/397669716c634588a0d00389bbd23ef1 : RETRY_LIMIT in 12m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf88a0b84ad648519a633efb82734ddd : SUCCESS in 16m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4d6681732c6490bbde897a965317948 : SUCCESS in 18m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62eaae821ba149588d6d4fd3b7ea5e72 : SUCCESS in 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f130c2e90814880acba5605a1f1b657 : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d516e340feca1bd2bbb7ab6dc12f777c3dd4c9bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 06:53:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451c51ea93a743cb9b8027d20af28a48\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/58ef03c21437490e9f4c58d208a4e225 : SUCCESS in 2h 54m 49s","accounts_in_message":[],"_revision_number":4},{"id":"7a5b7b395fc1eb3be481395ec64172f063a8b81f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-03 05:18:04.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"93fd1a740d7a3c0b6c80f72bf9a260d3f31e4f2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 06:34:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37056a8cd17e4fcd9d6b2251b2ed10d1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a383a227dbb0421192075c7554180720 : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":5},{"id":"6c8c9af6ee5e6526d0def625d52390ecd7fce844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 07:00:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cf585d939d84940ac920f307be522ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62dde396bccd4f55a0011b97eaf86214 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/58a917c8590e44939b903931bf9a4cb6 : SUCCESS in 54m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6fe435841f2b4c0aab22d07b15a06542 : SUCCESS in 46m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74b3cd2a123a4ef098c8b6cc1ef15855 : SUCCESS in 1h 03m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfe571b280cc4badb50844aa8acf24b7 : SUCCESS in 1h 05m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/363404d797b14d9c9ff5fbdaffd2798c : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/404bab57b5ab4b489424ec1e54924066 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ff48dbe00f745d3b3e53c9f4b5c3d39 : SUCCESS in 21m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee0e2269d95f4951be0d3ba8828458cd : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf166219024646a1972912d8947fa005 : SUCCESS in 19m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0fd4b6bdd1bb4217ad8cb24ea343e067 : SUCCESS in 19m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d7ec8126698f4232b286279cf8c50fa7 : SUCCESS in 21m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9c91b47b491e4e15b9360eb134a9331f : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab0a21e2be544a64b39d2c128f338a13 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b8190f2f231f4b36b5756096e4ac5049 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fc82c8b53994575b54b8c0b9735a66a : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd1310e6cf6f4d2e92c009a3738aae49 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d16e10a3ef7c44d198ecf72bf521b4f1 : SUCCESS in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a437ba77ad5e484581c32a52aafa4afa : SUCCESS in 57m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfb2678536364df9aee18c6600f284c6 : SUCCESS in 40m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14f161c363424d81b70f72b083537b92 : SUCCESS in 45m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a64bb9587e7458e9e1a900ec899139d : SUCCESS in 21m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5fbd7cf0d7c4ca6bd0e568ccba4f6e5 : SUCCESS in 17m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/442d05b7482d4bf1b2f74420b49f086a : SUCCESS in 58m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49d27fca246744f5b0a6b8ef68af3bd9 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":5},{"id":"cba0b4e01d2f97763ab036bfd39e2d9bf0ba1be0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-11 19:20:16.000000000","message":"Abandoned\n\nWe\u0027ve got some decent magic rules in the main ring v2 patch now; I think this can be abandoned.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e7bb3d0161413574d0e604f857e1d274a6ba4a85","revisions":{"ca0b2838ee6bbf8811d1e1ea1386b2a5c649a41e":{"kind":"REWORK","_number":1,"created":"2025-01-24 02:23:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/940088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/940088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/940088/1"}}},"commit":{"parents":[{"commit":"e8befaa12c44d6669217e336cca6aed9757b7e77","subject":"ringbuilder: Always save with a 2 byte minimum dev_id size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8befaa12c44d6669217e336cca6aed9757b7e77"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-24 02:12:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-24 02:12:51.000000000","tz":660},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nThis patch adds a magic file snippet for identifying swift ring versions\n1 and 2. All you have to do is append this to either /etc/magic or\n~/.magic. Once done you can use file to determine your ring version:\n\n  $ file -z object.ring.gz\n  object.ring.gz: Swift Ring version 2 (gzip compressed data,...\n\nThe `-z` is required to tell file to read into the compressed file to\nfind the magic.\n\nChange-Id: I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca0b2838ee6bbf8811d1e1ea1386b2a5c649a41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca0b2838ee6bbf8811d1e1ea1386b2a5c649a41e"}]},"branch":"refs/heads/master"},"bb4c229a89a5943b58b86315b299a506a806daa8":{"kind":"NO_CHANGE","_number":2,"created":"2025-02-11 10:00:29.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/940088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/940088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/940088/2"}}},"commit":{"parents":[{"commit":"1d80f616bc703687381322d310af36386fb17328","subject":"ring: add cli/ringbuilder version command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d80f616bc703687381322d310af36386fb17328"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-24 02:12:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-02-11 09:58:33.000000000","tz":660},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nThis patch adds a magic file snippet for identifying swift ring versions\n1 and 2. All you have to do is append this to either /etc/magic or\n~/.magic. Once done you can use file to determine your ring version:\n\n  $ file -z object.ring.gz\n  object.ring.gz: Swift Ring version 2 (gzip compressed data,...\n\nThe `-z` is required to tell file to read into the compressed file to\nfind the magic.\n\nChange-Id: I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb4c229a89a5943b58b86315b299a506a806daa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb4c229a89a5943b58b86315b299a506a806daa8"}]},"branch":"refs/heads/master"},"815c6abaa599751704c7032020810886de84ba1e":{"kind":"NO_CHANGE","_number":3,"created":"2025-02-12 00:12:12.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/940088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/940088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/940088/3"}}},"commit":{"parents":[{"commit":"1e0fb4dde53d336040f0bd07c683a521b2056197","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e0fb4dde53d336040f0bd07c683a521b2056197"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-24 02:12:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-02-12 00:11:36.000000000","tz":660},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nThis patch adds a magic file snippet for identifying swift ring versions\n1 and 2. All you have to do is append this to either /etc/magic or\n~/.magic. Once done you can use file to determine your ring version:\n\n  $ file -z object.ring.gz\n  object.ring.gz: Swift Ring version 2 (gzip compressed data,...\n\nThe `-z` is required to tell file to read into the compressed file to\nfind the magic.\n\nChange-Id: I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/815c6abaa599751704c7032020810886de84ba1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/815c6abaa599751704c7032020810886de84ba1e"}]},"branch":"refs/heads/master"},"1daa4535488ff0dc0696c9989a87cac16122d3c0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-28 03:15:53.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/940088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/940088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/940088/4"}}},"commit":{"parents":[{"commit":"c5da94af3f66b3d247e7dd4201aaeec020b3557f","subject":"Ringv2: Make sure the minimum dev_id_bytes is 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5da94af3f66b3d247e7dd4201aaeec020b3557f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-24 02:12:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-02-28 03:15:03.000000000","tz":660},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nThis patch adds a magic file snippet for identifying swift ring versions\n1 and 2. All you have to do is append this to either /etc/magic or\n~/.magic. Once done you can use file to determine your ring version:\n\n  $ file -z object.ring.gz\n  object.ring.gz: Swift Ring version 2 (gzip compressed data,...\n\nThe `-z` is required to tell file to read into the compressed file to\nfind the magic.\n\nChange-Id: I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1daa4535488ff0dc0696c9989a87cac16122d3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1daa4535488ff0dc0696c9989a87cac16122d3c0"}]},"branch":"refs/heads/master"},"e7bb3d0161413574d0e604f857e1d274a6ba4a85":{"kind":"NO_CHANGE","_number":5,"created":"2025-03-03 05:18:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/88/940088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/940088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/940088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/940088/5"}}},"commit":{"parents":[{"commit":"4456f59aa0465b143280cf64bce1e701795b060b","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4456f59aa0465b143280cf64bce1e701795b060b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-24 02:12:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-03-03 05:03:54.000000000","tz":660},"subject":"Add magic rules for swift rings","message":"Add magic rules for swift rings\n\nSwift rings start with a magic string (R1NG). The linux file command\ntakes magic rules from a magic file. It would be great to be able to use\nfile on ring files so we can find out which ring version it is.\n\nThis patch adds a magic file snippet for identifying swift ring versions\n1 and 2. All you have to do is append this to either /etc/magic or\n~/.magic. Once done you can use file to determine your ring version:\n\n  $ file -z object.ring.gz\n  object.ring.gz: Swift Ring version 2 (gzip compressed data,...\n\nThe `-z` is required to tell file to read into the compressed file to\nfind the magic.\n\nChange-Id: I2bf2b5698e4f82d0a88ca70421b98ed9d805bc55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7bb3d0161413574d0e604f857e1d274a6ba4a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7bb3d0161413574d0e604f857e1d274a6ba4a85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
