)]}'
{"id":"openstack%2Fswift~940535","triplet_id":"openstack%2Fswift~master~Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90","project":"openstack/swift","branch":"master","topic":"labels","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-03 18:57:30.000000000","reason":"A robot voted negatively on a label"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-07-11 02:07:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90","subject":"WIP s3api: add labeled version of error response metric","status":"NEW","created":"2025-01-31 15:59:08.000000000","updated":"2025-10-09 18:04:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7643fbb6902c22ec6cb72d6af7417d56f5f0716","_number":940535,"virtual_id_number":940535,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-09 17:21:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 17:21:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3cf657292b693d65804d61cbaadbb43ece58e354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-31 15:59:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3004f2dd04dde6cc9218fb5e4d10ee0d59780132","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-31 16:37:28.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1192809865aa7797d46fbcd58de50eae6eda399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 17:21:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/352401b8d40e4a7ba836845a84fb2170\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882dd46f51ef4438ae3a25a727a1521e : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4c63fd0e28b14e1da01dc1054889814d : SUCCESS in 42m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3106542cf714c1ab989ed6ec01a555a : SUCCESS in 43m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5dc693bae7c4921b3c60ea3edc90968 : SUCCESS in 1h 01m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98806aebcd9742a78745129f54018b7a : SUCCESS in 1h 15m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40473042e52642c09ccd1b7e378149bd : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e87e633de6b64b869bdafbeb72699efe : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/908504c65eb649f7aed8b12558c7b374 : SUCCESS in 25m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8e036331731449bb3176d626987d333 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0cee15f3cb004394baf91d4ffaaf254e : FAILURE in 15m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2b8d09ab66c4277b096c0545b348a90 : FAILURE in 21m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/573df2dd746f438892c4215f10c0a01f : FAILURE in 28m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/69e39b8c44ac41c8bc74560a2c526dde : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/584fa878025d4367bdb7581801cb292a : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e714cf1257bb4fdcbc9efd155a4ef916 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55b73a3e0b984e7d88060b00276f28f3 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0e60ada2d504c24a86e9f712a2127c4 : FAILURE in 19m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/de2f7056229b403187c005b8d5d1396a : SUCCESS in 9m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4b6233e93084219b0cbc116872e6ff9 : SUCCESS in 1h 04m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8e01a25692e450dafda2b26fb2fb9c3 : SUCCESS in 45m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b16c8626655b44d194ce580253038395 : SUCCESS in 44m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cb24d2e899e490c852cd8f498a207b9 : FAILURE in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ece26ca62d6490a9b04e718e6c6d46c : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7b87fb7105249eb8eaf0d08daa48149 : SUCCESS in 18m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7394bb5346754163aad883bf74311dfa : SUCCESS in 52m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1449a07005b44af483491085185dbb90 : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"37d102555029fdd45b0943fd90457f28e0eb3063","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 18:16:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/135b966c2f104589a5f244601caab6a2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6894f36406c2431e948b84407e49bb6e : FAILURE in 2h 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f7d507bb750ae99d29c56d205667b57df829bbd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-02-03 17:44:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":2},{"id":"09f220f93e0ee9679848ad4c6691059f40971fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 18:57:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76cb90ad0bb44632aa9fcf8b3dde0cc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cc78c418d04c608a26ac0c4be218a5 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/eee4dcf2e13e43ecb749253e2d1b61ed : SUCCESS in 45m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df082c59921e48bb8c693e1c2660819a : SUCCESS in 45m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a85eb2b200d746659a971e86d1af8bc8 : SUCCESS in 37m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f563b73939b4921bdd42de4af31fee1 : SUCCESS in 1h 06m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ee5bf0dbc0a43409381388c8a56d26e : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b21a8148b1904711984d1cd05a8a397e : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/179ce3c9a5f14c76a670ac47de44baab : SUCCESS in 20m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e23aee88f7d45269157620a89ac6294 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95e85a519f034710be0d6e9d1b229462 : FAILURE in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/171027b50673485d8efd51d66279ca8c : FAILURE in 15m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f8f92fe990ce473e8db416607ac4d77c : FAILURE in 23m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/21a22561277e4331ba815a48a431639c : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c7a2c6aa65a74b4e88dc2138fa24eeb1 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/34743e17855a4a599a572584ab9b2061 : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9771336b872446198fbbc3b53fb5bb39 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89577f37525243458673db8e1996f111 : FAILURE in 16m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2f9fc67f7b34b869291a35807caded1 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48bfe7bea0274a5e9d4a2a9b2ceee2ad : SUCCESS in 57m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eff961977b8343d79efcfc182bdb5646 : SUCCESS in 48m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9672112e42e84a9b997e7a19e2be790a : SUCCESS in 45m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8173dacef69d4af98aaea39ba1558cc5 : FAILURE in 23m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4920b13f8d14e1399b07be1b34c4fdd : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1e0b018c1994183a96c4a33a645053f : SUCCESS in 18m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb4626c83ca147f68050b30702d89e62 : SUCCESS in 38m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76552ab977a34106a23a0aafd087ffac : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4a1cd86bf7780dd30aa1a25761444ba15b9ed9ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 19:41:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd695bb819a4674b9f69815e4a907a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22246dc6b2c54f33aaeb6781b30a21ce : FAILURE in 1h 32m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1f875dae2271e6d6795bcf43d4ff60b5011f4c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-10 23:04:35.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ab0affab0097c76041a38390edb93c16d0641637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-10 23:04:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"56d5835fd636d35b2727ead78352ff15b628eaa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:34:21.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47d25b30e6ce422f84b5ac374a990ee1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e21f1f51a56a4cdcb262414277a7f234 : FAILURE in 1h 08m 26s","accounts_in_message":[],"_revision_number":4},{"id":"28d966b0bc7f47062467ada8298baed9258845a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 02:07:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/708f888d92c14edeb691526ec782229c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8692744982d24564b07446d956018c25 : SUCCESS in 5m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/06fdfd70672e49e8ac44adc0cf0fe013 : SUCCESS in 29m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/03bdf429b929469d8a4991ee7b6b11c0 : TIMED_OUT in 3h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf63f87a077d4ef48a4147c20292df8d : SUCCESS in 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf99847a9ed64d4994573b6e13de5d2c : SUCCESS in 58m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/822ef6be197e47929df77acf08e97996 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69c5a4452d7e4306b123aad042aac5e5 : SUCCESS in 23m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d46a53e9faa54f5dadd324bb5ef50392 : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f938e5d9f36404ebbbbb1745450ff3d : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc2a612b98bc492b840fc9a003dd932a : FAILURE in 14m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a843b99e48c74dc8a914a10748e8fb1c : FAILURE in 14m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1df1ddb0d1eb4589b3c3231c57bc58c0 : FAILURE in 17m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/abebb841668f430dba2c2b67bab43b6e : FAILURE in 19m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d8cadcc3588748c4b1f53544bc46b216 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bbe4073e52634d659311e4b7985595aa : SUCCESS in 12m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4ac4acecb4f402c9e50ec4d003fd4d4 : SUCCESS in 7m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f57ee21354f74f55ac08a43b841d2fc1 : SUCCESS in 7m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98042750989841689e24fd6f9e80a045 : SUCCESS in 11m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff6156c10327425282b5aa3292ae25d2 : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f01e54f2f0814edb94fe9b6581536310 : SUCCESS in 51m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b8099de5119414099351bd091e10ea3 : SUCCESS in 33m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/503ef039e74d477d8173f915b2eb7b54 : SUCCESS in 54m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0073abc6e9c944c78143010774bee0b9 : FAILURE in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77545101bdf44218f5ba444cf6c6170 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78597fdc35a644ddaf12ac666959cd12 : SUCCESS in 10m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/023f385e680b4cbaab47d3cbcd30ce8e : SUCCESS in 38m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b07825f44c3747c89a604a1ec13119f5 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f493e16d5015b96fd19bffa3772548a7ce8a3693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-11 15:56:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"340370f04ef0a99cd16324a6d6beb585db9b033c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 17:01:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ae5a94b2e3c43939bb686007e8d110b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325db74e84e74a40b4eb265e28a441e7 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/5756ce71e19748b0a1a897861f71301c : SUCCESS in 59m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b05c140d2fe4d1a8c9a90b7313481ea : SUCCESS in 58m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfa72ebc5d164363b3ce24d2c8e1d6b5 : SUCCESS in 40m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f86e540004e43e7ae3b868d9d059931 : SUCCESS in 50m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1825f0972ca840e2aea3b67eba45800e : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fa83fa45b9240f8870792ee3ce543fd : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e646f591e849463db79e1a93723702dd : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33625e78bf0e4a8281a95b27d83139a1 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1265f97d205a4c7e95b4c9a4ea761f44 : FAILURE in 19m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8bf50cd616044162bf9406fd893e0d20 : FAILURE in 20m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49b55c2bdace4d14a8cb127f99c30b19 : FAILURE in 29m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bf68cb28b68a44eab45a19d669994d0d : FAILURE in 22m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0fed316dc3994d9f8909707c7e847b62 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f5cfd0e8f0f448187289da131358906 : SUCCESS in 13m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6af81b110ec24500a041a76710daf160 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19f12db1f3d5402c9fd906625425bd84 : SUCCESS in 6m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/162c20911e4643df9d648907575cc547 : SUCCESS in 15m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/91244b1e8e0d402a99f17f5dc11ec82d : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2442ce9b56a94f52b37df60fb72680af : SUCCESS in 52m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/651fd2b2c159446081fa0999983c25f6 : SUCCESS in 38m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ce461baaad74736b6b89e1a56fa3427 : SUCCESS in 44m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2a6b5f28c5e4baeaf81307628424527 : FAILURE in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842c4bc704bb49e6b5af0328b776f5e5 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/280e36ac2dcf4d67946b2001a3fca555 : SUCCESS in 19m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8031fd23c1747f6b964b7b1197429bc : SUCCESS in 48m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44c6d6d4fd5f47779b4faf77647d1520 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":5},{"id":"57f321c0f1a6950d577b8f3d13486ed17795d889","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 22:42:56.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11c438b47d54defb22c8b6c624801c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/657553dbdaec4d9282630c6fdb589eb2 : FAILURE in 1h 19m 37s","accounts_in_message":[],"_revision_number":5},{"id":"4df97b6b57591224136bdd3f3cf70a0e494a3a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-07-17 14:09:32.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f49241101a434d8497cc1075bff873fe85efb608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 15:16:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7202ecab4fd4a559cb536aae0632c5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4dc84ff416432bba3e04da2f89ffb8 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/14b82fc2638e433ab8171e94024b26ed : SUCCESS in 28m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50f0f488e6694de985450b50ce308e00 : SUCCESS in 33m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/908f79cbd41945c7b3520d88063c7447 : SUCCESS in 52m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48c806d1f5684a63a79736f07bb2d17a : SUCCESS in 59m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca4bab716ff64f3d8b11317a96fba8c3 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8053f936b81a474fb3d5116aba81bfc9 : SUCCESS in 23m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef034c018cde464d99e93871bb21cbb2 : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bee32cfedf4a4077b287542ede53f7d1 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/625fcdc32ee5416ba5304abd9e4cc53c : FAILURE in 20m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2eed2f258b04057940240f2458355aa : FAILURE in 20m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2dbecb4ab94b4ae59bc2434458c0c2a7 : FAILURE in 23m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5b5858bcfa5842219bdc6ff9fb57373b : FAILURE in 24m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/675e4be9d7c24305acf0440dcafc3641 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/288896698d0d47f98997a88454ed05c7 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/994d6f43f903483c9bf953ee44de1602 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e3f308197844cf1a456fd8478fbedd3 : SUCCESS in 8m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb4bdcf27892416fb11ddc81fdcca6be : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8771311bf6534eb69b28b2a1cb0bbab3 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8c329d8d416406d94c3417e55bcdf30 : SUCCESS in 58m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a794619054ee43b09df7f749091bc687 : SUCCESS in 43m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dde4226af2944daa8b9b0a9a3dbbab14 : SUCCESS in 56m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70888a6f8f034514952e0b5e90c7453d : FAILURE in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14dd6611ec3d4735a5db85e44847d7fa : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13084731c55749e49a585f0dd80fb809 : SUCCESS in 25m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94c6954091ef42ad8016e80d78cac27f : SUCCESS in 51m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d344db21b5c44196b04f63a1bf47f29c : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":6},{"id":"ce7a7dd0269e9231ad1a4578dedfbb89b5c37264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 15:46:47.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fbcf46c07514e8fb0aa4f8dc0514a76\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11a875ddd6f84f55a7a714ae77ba79cf : FAILURE in 1h 08m 04s","accounts_in_message":[],"_revision_number":6},{"id":"61f4eef0a2aa4a03c8634d3d4f208c1f0ca51998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-09 16:09:14.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ac4a9d66961ce1bdd236c306d74819c4d6707983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:21:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99f40bdc280f4c3c802098f56d1ded2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc08caca288a4a1a9699f6670e528ddd : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/7c79dc8b507c4a7ab5b2e7f9410b7094 : SUCCESS in 1h 05m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c67fc624382f45c29a316af0c4c7c6d2 : SUCCESS in 47m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4085eebe442943539bee56e2ab8f4094 : SUCCESS in 1h 01m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61c0d9622f4d4aeaa3a385b199d40883 : SUCCESS in 1h 04m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13ff18b710b749df9aaa8666afc007ad : SUCCESS in 22m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4890e4d39a648e884b4f3dcba8a894d : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5904191ef4ff4a67bbe0b6731b0cbea5 : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11a1d86f6f3d40398a2631f47b09042f : SUCCESS in 4m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1000f5726570436091f801ab35c76b9b : FAILURE in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e02f16ad97f649468b5562a509048fc1 : FAILURE in 8m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7c5fe7fa131478eb9cd0a4cb90766a2 : FAILURE in 23m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a4b6bfc206c94e98b912212a07523238 : FAILURE in 21m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/451e34fde6ce4d56a7517628e18671b4 : SUCCESS in 13m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/75e5d8bce91c44259966394ef3d05a5c : SUCCESS in 12m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62058f6b22c948a5ba38322a4c8cec30 : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c753ba2263ae4e0aaa40293f611dd450 : SUCCESS in 7m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98c10b7e047b4b9b9f595c885d4b53fe : SUCCESS in 13m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e2b199c49d64462910a7f9a41291c8c : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ece09eff5b84b04bae00e7cc86fdea6 : SUCCESS in 37m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a1667020dd34b6a87f15a3e2af43248 : SUCCESS in 56m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b9078001b9ea455c8b249bc6fba4fa1b : SUCCESS in 21m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebd8483199844893bf7dd6d809f63222 : FAILURE in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f623033f0074c6497691d1c6260aa78 : POST_FAILURE in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4b3bd4631544818a662ace82ab75cff : SUCCESS in 16m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/499f0665ef2b4d01928ded348d96bbe3 : SUCCESS in 53m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66d5941f65604f8abdacebefeb355683 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":7},{"id":"a7643fbb6902c22ec6cb72d6af7417d56f5f0716","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 18:04:49.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552a4d1ddac9422e99cea113f29ea68b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8069cc14e674e55b96d9cc91a641627 : FAILURE in 34m 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3b7b503fdc92796376f90e0dee4fd2ad6544ec0c","revisions":{"542e65272d8da705fc493410f6f27887e87d0c02":{"kind":"REWORK","_number":1,"created":"2025-01-31 15:59:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/940535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/1"}}},"commit":{"parents":[{"commit":"64278526b91ac05a4352f8c5cc57758576ace984","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64278526b91ac05a4352f8c5cc57758576ace984"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 15:58:29.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/542e65272d8da705fc493410f6f27887e87d0c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/542e65272d8da705fc493410f6f27887e87d0c02"}]},"branch":"refs/heads/master"},"d76fff4256391bd882beb1b10664104b96de7721":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-03 17:44:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/35/940535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/2"}}},"commit":{"parents":[{"commit":"c54dac02863c30183e4f8c587aa463b8943be55a","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c54dac02863c30183e4f8c587aa463b8943be55a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-03 17:44:04.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d76fff4256391bd882beb1b10664104b96de7721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d76fff4256391bd882beb1b10664104b96de7721"}]},"branch":"refs/heads/master","description":"Rebase"},"08d5acd954e24e4084dd3bc0eedf5ae0cbcc67df":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-07-10 23:04:35.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/35/940535/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/3"}}},"commit":{"parents":[{"commit":"c54dac02863c30183e4f8c587aa463b8943be55a","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c54dac02863c30183e4f8c587aa463b8943be55a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-10 23:04:35.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08d5acd954e24e4084dd3bc0eedf5ae0cbcc67df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08d5acd954e24e4084dd3bc0eedf5ae0cbcc67df"}]},"branch":"refs/heads/master","description":"Edit commit message"},"edc63d7cbd09315078c208e49ef3850527fd2649":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-10 23:04:40.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/35/940535/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/4"}}},"commit":{"parents":[{"commit":"151fc267a14d5cd5ef057bc3c7d94f7340c1e19a","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/151fc267a14d5cd5ef057bc3c7d94f7340c1e19a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-10 23:04:40.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edc63d7cbd09315078c208e49ef3850527fd2649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edc63d7cbd09315078c208e49ef3850527fd2649"}]},"branch":"refs/heads/master","description":"Rebase"},"d0e91b82140aa6c341b01cd18dc899b2ecc862a4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-11 15:56:20.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/35/940535/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/5"}}},"commit":{"parents":[{"commit":"4a66a16a657b03daef351bc14db25bf10945d22a","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a66a16a657b03daef351bc14db25bf10945d22a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-11 15:56:20.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0e91b82140aa6c341b01cd18dc899b2ecc862a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0e91b82140aa6c341b01cd18dc899b2ecc862a4"}]},"branch":"refs/heads/master","description":"Rebase"},"ba389ca76f171e59a8735d809d94e7756b7ae427":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-17 14:09:32.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/35/940535/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/6"}}},"commit":{"parents":[{"commit":"d3997395844e2d3be0bc6be7521e7412177ed55d","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3997395844e2d3be0bc6be7521e7412177ed55d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-07-17 14:09:32.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba389ca76f171e59a8735d809d94e7756b7ae427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba389ca76f171e59a8735d809d94e7756b7ae427"}]},"branch":"refs/heads/master","description":"Rebase"},"3b7b503fdc92796376f90e0dee4fd2ad6544ec0c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-09 16:09:14.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/35/940535/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/940535/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/940535/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/940535/7"}}},"commit":{"parents":[{"commit":"1cc9a1c552da04779c341e0d56c740f015f423a7","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cc9a1c552da04779c341e0d56c740f015f423a7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-31 09:41:33.000000000","tz":0},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-09 16:09:14.000000000","tz":0},"subject":"WIP s3api: add labeled version of error response metric","message":"WIP s3api: add labeled version of error response metric\n\nTODO: add unit tests, likely add to the ones added by Related-Change.\n\nThe Related-Change added statsd metrics when error responses, such as\n403, are returned from the s3api middleware. Now there is support for\nlabeled metrics, this patch adds an equivalent labeled metric.\n\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nRelated-Change:  I4976b3ee24e4ec498c66359f391813261d42c495\nChange-Id: Ic59764b01e9cd4888c6fd1645eb5d7bf9e079b90\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b7b503fdc92796376f90e0dee4fd2ad6544ec0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b7b503fdc92796376f90e0dee4fd2ad6544ec0c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
