)]}'
{"id":"openstack%2Fswift~940601","triplet_id":"openstack%2Fswift~master~I52b37cfa49cac8675f5087bcbcfe18db0b46d887","project":"openstack/swift","branch":"master","topic":"bug/2097074","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-02-12 00:22:25.000000000","reason":"Change was submitted"},"36393":{"account":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"last_update":"2025-02-12 00:22:25.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-11 22:49:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I52b37cfa49cac8675f5087bcbcfe18db0b46d887","subject":"Add per-container storage policy to account listing","status":"MERGED","created":"2025-02-03 08:30:21.000000000","updated":"2025-02-12 00:23:50.000000000","submitted":"2025-02-12 00:22:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"940601-bug/2097074","meta_rev_id":"918bfe4b8251408c46d5743d95feda9398f3ebd4","_number":940601,"virtual_id_number":940601,"owner":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 00:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-11 22:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-11 22:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 09:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-05 16:32:10.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-02-05 21:21:42.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-02-11 22:49:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"47d7a4c4ddfec0af67b6b229c63e0ba1f813851e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-03 08:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d566573a4503847e4356d6c8bc56cb72ccf983a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 09:28:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4af90d793f243da918a63f53ad7ef67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830759927b3843628b397e9dbf5f4d76 : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/9bda6b51c542445097fa37138798ae8d : SUCCESS in 47m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adb8deb2f5aa49f9b0519a930911fd6b : SUCCESS in 46m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7cd3ffa0f08409fb968537715c70738 : SUCCESS in 41m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/571d3f4e371e4bd0afb899ae37a848b4 : SUCCESS in 48m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47e36f5b7a674774b214b3a915e28a63 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26d9a01ee1f343c9be3e48f82c1e4c6f : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7121c792ea64ee78cd0026514d21289 : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b82118e8caf4bfbae5b01b7e5b59c97 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d52bd501eb574c8ca4d57c7bbd67e6fb : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efc31a425be54a09821fb33228edf95b : SUCCESS in 20m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f7d953da026e41708d8fcd3e82780e4e : SUCCESS in 36m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/97cbfb0f9c5f48958aa04f909447cd86 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b056147cef247bc81fb2792ff683d51 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/535c494af5a24313b7eed982f55973fb : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ecd04728373644128d8089cb276f9613 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd4b61536f5b41aa8d7d1af3c72f30d2 : FAILURE in 9m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e55441b26d8429eb38e663a7c168de1 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d864b3845b84b35a35e03f56ae0d9ab : SUCCESS in 38m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20c24cfeeae04a42bc70a3c1317625bc : SUCCESS in 47m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74fffe9dc2a74ade8dbfec6cad902033 : SUCCESS in 27m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00627bbfd81f470b8526faf5c10dae7c : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e61f2fcb1b34c6c83afbbd77c0190ce : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d72acd5629ac40658984f5164fdb28f9 : SUCCESS in 11m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a8e21e6f94a4904a6686f2559acc3fd : SUCCESS in 42m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43231cffebcc4b52a095c7806493898e : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e8eabb5ccbaf1a273196f3171499dbc9805ef1a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 09:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab2bbad9f744563abe4a242d2a4cd92\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/609cee49cf7844d59e9b92d41ce1bbb3 : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8e31de0eda8e6942f05b232d5b23084febb27395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-03 20:11:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7d196cf1460d1491ce4d223136ec9673b26f2f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-03 20:17:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"128b49cdb86fc0d4b13808e1a9e312e711bd4871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 21:31:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee020967d72c45cabaa70d17f419465b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2018a1c5d048388b8f97e19596016a : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/68d395f5b2d048ed8520ea3ab18b4d7d : SUCCESS in 1h 03m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a01a7db0c6a346b6816a5530e6e4c544 : SUCCESS in 1h 02m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26a549e2f75849a39580f873e7ff9837 : SUCCESS in 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/269ef5c10bb748a884f7479eefb73bcf : SUCCESS in 1h 07m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea23ff4e1c9c4f61ab71283ca0674055 : SUCCESS in 12m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cec869304b3a4292bde069d6a830ecf2 : SUCCESS in 22m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b638aef738e4007829032167b2257bd : SUCCESS in 12m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d80a0ab1b5c4831b47904663f27c832 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06486f2ffd5f4c6383ac892c37e8989f : SUCCESS in 17m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/880b781492d74b1591b2c6afca516ef9 : SUCCESS in 20m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5801966e8015469586439f58380d4737 : SUCCESS in 15m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/65c033a9479147b58dcf2134cdee8252 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b18e9b71861c48ef982fa189212a1433 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dbeef22948ab4226a4b700d7db985ca5 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf6c5b1bcca04b59aa8dd8baa4aa792a : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c5337f9df8e4b4198a3feb950b77c3d : FAILURE in 10m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a5e36620436b412e99730b48ae107cb5 : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/879b85cd148b463aa921a4bb239e54ac : SUCCESS in 56m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbf54bead2f54b84bf638522dc0bd317 : SUCCESS in 36m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6a003cf3fe545faa923aad4e3981c65 : SUCCESS in 37m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1e3b36dd56f41c38ed017f4cf7c8e81 : SUCCESS in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ea41fb31ad473c90146e3d937bcd35 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce5284c3660d4a1cac8cfc7d1943274b : SUCCESS in 11m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c4262d0a0c14ac7af1309d0928d83ce : SUCCESS in 50m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3c31f29451c46b7ad6fd98d989c8059 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a2e6b458970533d415e732c2c6b04a147152cd23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 21:33:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef91824bf174dae85cca517b6644aaf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8321a40f09e04e45b756dd9b9f3fca07 : SUCCESS in 1h 14m 17s","accounts_in_message":[],"_revision_number":3},{"id":"b740554721cc2c8e9b923ed11aa7349cf5fe0e98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 16:32:10.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fb9fe97c7315e85bfd88efcfa22324301584aa81","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 19:04:23.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0ec33a1e2e431ca334689bbf0d6a1fdd55daa770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 19:35:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f2aa6874dc0842a603b5e28a60c165084955efed","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 19:37:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc15970aed6956c85bf29f820331efd5b1c13277","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 20:11:17.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f26f2a78bcbefe6b42b4c2e514f0fdb9560e8210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 20:53:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4de83f6ddd14fe3a159af7b78f753b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30983165c2204a97a8e4bd3e71749ccf : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1cbd9753e98847bcb78d220bf345de3a : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/008c63749ea94bc98d2960a9c53bfe12 : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d7a0e10adff42889c5e9ae813c01fa9 : SUCCESS in 48m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/728848a09ccc443badee8aa6e7d3b4c4 : SUCCESS in 46m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef9a39d925ac40aa95d03c0c820c4276 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca0c25a69b6244f2a34bc2c4746e9c02 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ad5ea2507ac4610af81e109024a706f : SUCCESS in 19m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8f0efe22c694fceb75b2be4572ea15b : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2fb0c426e1034628a85caae676bc9fb4 : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/082489564f4b462fa5daed68fbecaa3b : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8a36dffd022645dd964c618b8faaba95 : SUCCESS in 20m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d60dee2cb7d48b0848bd55f0a71e5e8 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f838592d8f8048e487b7fe3aca1af6d5 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/16b2be4bbf9a426f864eb6cb2319527a : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d462ed1bced544299f10bd438de5b7f4 : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/956c25bd39be45d385668def6eee1789 : FAILURE in 15m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7beb5b5a718a4504b3291756b53433f1 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/309546b6ab8c4751834eae2061c1138e : SUCCESS in 53m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d254ac41447d490b81df56826e85d232 : SUCCESS in 49m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3dc1ac9215af420c979442c8ab0a4413 : SUCCESS in 35m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dc43e2c48ff4bff88a2f1cd2820ed91 : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd16edf10d4461e94c0aaad5c0d7792 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fb1a05270e04b2282a2297cebd855d1 : SUCCESS in 23m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e6b143f019c4ff391e90b52aea85a6e : SUCCESS in 36m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc7f9edd7a76412d8efd8d71c30ed5d1 : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":4},{"id":"175f54114a5b93979d9386e70fafc96785078179","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-05 21:21:42.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b0f788164eef3f64b97efe65e77df3564290026e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 21:26:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8406f1a9481944a28d0bfb39a8757a12\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75c0788aadea4dd2b59c82a37420b8fa : SUCCESS in 1h 49m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c8d07ff352912aae78123513faca513679aa581b","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 21:34:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c00a7a5feab0e9bbf97d2cb136f87a79bf11242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 22:09:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d8f6a01ac288f619fb161169ae30b8e54aa4e8dd","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 22:17:30.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a1a868c0e4f32d127d218619bf0393f7af3fd971","author":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"date":"2025-02-05 22:27:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a64516deb51437b7cefe59c9c541144a3e1971e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 23:18:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76fde911eeb3478eb62fce2bf6887bbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d369152e27a46c9b998fb9cc44ddaab : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/803c76f5d4964a209b0bf07bf6c3ec2b : SUCCESS in 1h 04m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d62123432ce4979b62c464235ac569b : SUCCESS in 49m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdf9c581762343b5bb34ddf5c4009875 : SUCCESS in 1h 02m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/703c6d3e04f541f686e6329644dc3919 : SUCCESS in 53m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adaddc43683e420e8c9ec3a903481ba8 : SUCCESS in 23m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab454bb19f744cb8b9fd60f796c036a1 : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/714f4b4867f44725b2372ce4b69e4830 : SUCCESS in 24m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9cb294e0b53a4139944b04ad74909241 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92a2a72aaa3b4543a1212020f3a1993f : SUCCESS in 8m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52ebb95935784c229b51719a1e53d977 : SUCCESS in 9m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/19b4cda81ed74f22a7c9c9a71a165184 : SUCCESS in 32m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/830acb4b67da42a7a2463d4b0e9004aa : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/688cf357e1024bb19356f81c19a51ddc : SUCCESS in 17m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6a7618dedee543c0abc815a95873454f : SUCCESS in 12m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea21675986b54076bd2a6e9007921bc5 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b069a02c76544c2080a2f17d8ac5825b : FAILURE in 16m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c80c29f11c0429dbda7babc78ec2552 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/49b66a3f401643b3bb2a6e65031f9663 : SUCCESS in 53m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8650e6ab0ba4f918ab1abe52069bdc7 : SUCCESS in 34m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e347c04022284790a7fb21c195d0b60e : SUCCESS in 52m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4ae0cc36d0e4bb9ba01312c478593d7 : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e6e941257ea4f4d892d3edeadd4ee47 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3dbbb26a475044439f0a529a597ee689 : SUCCESS in 19m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c9a3cdab2794c02bbcb1343e1edaecd : SUCCESS in 44m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eebc26d60d534dcf92738c29c8336a39 : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":5},{"id":"ca81d5d2b5e4976fe002385693fad9dd9e10c37d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 23:32:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10dd91a2075c454b84c029e07f3a1302\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/24fcfb9d81b34d33b74ba9d919dbe78b : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":5},{"id":"7b7df2d8a64ba344d5bd17ca251e7e5c78f2a540","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-11 22:49:59.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e3f4f71f790753af7cd6deaebb8ff84c6b2daa82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 22:50:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f624dc417b66d2b238000a3131fb36d72daa4cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 00:22:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0714423022244781b3d3f4a579904897\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc336ae14114ce3a0edce880d06f872 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/75052eea72574db4a6171d880119ba31 : SUCCESS in 55m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc008d2751b54f189532bf22fa71c245 : SUCCESS in 56m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8e7c78a13f140a6b780a9a45bfcc566 : SUCCESS in 46m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8676084c05344598e0e18bbe13c0677 : SUCCESS in 1h 01m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f0b8801e94b149f7af28958e9825782d : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/02a7d1072a284cf885bb2dc1e778a6c6 : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ca461def5cbc4b40bd1ecfa6771f1d92 : SUCCESS in 17m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a0a6648707bf4aec9e42b284f018f3cc : SUCCESS in 13m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/af0e095f4dca43c496f774126d236c4e : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2241d7746b7a47a0a17a222774a95de6 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c77c70b4f2b54ac58776af8a53bba8e4 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d33f2272540a4e24996c320c87bc1c89 : SUCCESS in 4m 21s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3219b4c44d504bc58098852716ce19fb : SUCCESS in 15m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdd18d3b270e4537b55fd16bff79bca8 : SUCCESS in 1h 03m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b81e44313d6449cda8d66fc2a70b6e78 : SUCCESS in 49m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18eee407b99b479d9b7867a7355b4be5 : SUCCESS in 58m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdd37a7af548439bb47b0de3f657677d : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2455c1e389f744ffb38a635eaf34cd93 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26c058e1c71440c192a6a5b816d44838 : SUCCESS in 18m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa5ad7f4ca2243eb9ea38dea0851dff5 : SUCCESS in 52m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebaf8210c2774ac09265e148b4f46b9e : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":5},{"id":"41b487dd080337a2c3978ca2900d253608f1e41e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 00:22:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"918bfe4b8251408c46d5743d95feda9398f3ebd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 00:23:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/557dc802eb664c7f95016b6ea7d6dd23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f7017e582749009fa2f1084a13cea9 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/74f29698552f45b999dcf4b98118e02b : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"965cc2fcbc18c025639c61e945bc211bf9a2783b","revisions":{"74e22e6938172c9fb8abdd32ce3a5a19c6fe9d14":{"kind":"REWORK","_number":1,"created":"2025-02-03 08:30:21.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/01/940601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/940601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/940601/1"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-01-31 06:09:05.000000000","tz":780},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-02-03 08:30:18.000000000","tz":780},"subject":"Add per-container storage policy to account listing","message":"Add per-container storage policy to account listing\n\nAdd the storage_policy attribute to the metadata returned\nwhen listing containers using the GET account API function.\n\nThe storage policy of a container is a very useful attribute\nfor telemetry and billing purposes, as it determines the location\nand method/redundancy of on-disk storage for the objects in the\ncontainer. Ceilometer currently cannot define the storage policy as a\nmetadata attribute in Gnocchi as GET account, the most efficient way\nof discovering all containers in an account, does not return the\nstorage policy for each container.\n\nReturning the storage policy for each container in GET account\nis the ideal way of resolving this issue, as it allows Ceilometer\nto find all containers\u0027 storage policies without performing additional\ncostly API calls.\n\nSpecial care has been taken to ensure the change is backwards\ncompatible when migrating from pre-storage policy versions\nof Swift, even though those versions are quite old now.\nThis special handling can be removed if support for migrating\nfrom older versions is discontinued.\n\nCloses-bug: #2097074\nChange-Id: I52b37cfa49cac8675f5087bcbcfe18db0b46d887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e22e6938172c9fb8abdd32ce3a5a19c6fe9d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e22e6938172c9fb8abdd32ce3a5a19c6fe9d14"}]},"branch":"refs/heads/master"},"af9ff8af6007d27c5bf4d35fd7b50eb79fe663c0":{"kind":"REWORK","_number":2,"created":"2025-02-03 20:11:46.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/01/940601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/940601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/940601/2"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-01-31 06:09:05.000000000","tz":780},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-02-03 20:11:08.000000000","tz":780},"subject":"Add per-container storage policy to account listing","message":"Add per-container storage policy to account listing\n\nAdd the storage_policy attribute to the metadata returned\nwhen listing containers using the GET account API function.\n\nThe storage policy of a container is a very useful attribute\nfor telemetry and billing purposes, as it determines the location\nand method/redundancy of on-disk storage for the objects in the\ncontainer. Ceilometer currently cannot define the storage policy as a\nmetadata attribute in Gnocchi as GET account, the most efficient way\nof discovering all containers in an account, does not return the\nstorage policy for each container.\n\nReturning the storage policy for each container in GET account\nis the ideal way of resolving this issue, as it allows Ceilometer\nto find all containers\u0027 storage policies without performing additional\ncostly API calls.\n\nSpecial care has been taken to ensure the change is backwards\ncompatible when migrating from pre-storage policy versions\nof Swift, even though those versions are quite old now.\nThis special handling can be removed if support for migrating\nfrom older versions is discontinued.\n\nCloses-bug: #2097074\nChange-Id: I52b37cfa49cac8675f5087bcbcfe18db0b46d887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af9ff8af6007d27c5bf4d35fd7b50eb79fe663c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af9ff8af6007d27c5bf4d35fd7b50eb79fe663c0"}]},"branch":"refs/heads/master"},"0556b8977a8d4cd06a02fc25f9846b73c39551ec":{"kind":"REWORK","_number":3,"created":"2025-02-03 20:17:29.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/01/940601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/940601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/940601/3"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-01-31 06:09:05.000000000","tz":780},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-02-03 20:17:14.000000000","tz":780},"subject":"Add per-container storage policy to account listing","message":"Add per-container storage policy to account listing\n\nAdd the storage_policy attribute to the metadata returned\nwhen listing containers using the GET account API function.\n\nThe storage policy of a container is a very useful attribute\nfor telemetry and billing purposes, as it determines the location\nand method/redundancy of on-disk storage for the objects in the\ncontainer. Ceilometer currently cannot define the storage policy as a\nmetadata attribute in Gnocchi as GET account, the most efficient way\nof discovering all containers in an account, does not return the\nstorage policy for each container.\n\nReturning the storage policy for each container in GET account\nis the ideal way of resolving this issue, as it allows Ceilometer\nto find all containers\u0027 storage policies without performing additional\ncostly API calls.\n\nSpecial care has been taken to ensure the change is backwards\ncompatible when migrating from pre-storage policy versions\nof Swift, even though those versions are quite old now.\nThis special handling can be removed if support for migrating\nfrom older versions is discontinued.\n\nCloses-bug: #2097074\nChange-Id: I52b37cfa49cac8675f5087bcbcfe18db0b46d887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0556b8977a8d4cd06a02fc25f9846b73c39551ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0556b8977a8d4cd06a02fc25f9846b73c39551ec"}]},"branch":"refs/heads/master"},"098309193a2a9ae07a386522925c3502813c1509":{"kind":"REWORK","_number":4,"created":"2025-02-05 19:35:16.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/01/940601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/940601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/940601/4"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-01-31 06:09:05.000000000","tz":780},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-02-05 19:34:14.000000000","tz":780},"subject":"Add per-container storage policy to account listing","message":"Add per-container storage policy to account listing\n\nAdd the storage_policy attribute to the metadata returned\nwhen listing containers using the GET account API function.\n\nThe storage policy of a container is a very useful attribute\nfor telemetry and billing purposes, as it determines the location\nand method/redundancy of on-disk storage for the objects in the\ncontainer. Ceilometer currently cannot define the storage policy as a\nmetadata attribute in Gnocchi as GET account, the most efficient way\nof discovering all containers in an account, does not return the\nstorage policy for each container.\n\nReturning the storage policy for each container in GET account\nis the ideal way of resolving this issue, as it allows Ceilometer\nto find all containers\u0027 storage policies without performing additional\ncostly API calls.\n\nSpecial care has been taken to ensure the change is backwards\ncompatible when migrating from pre-storage policy versions\nof Swift, even though those versions are quite old now.\nThis special handling can be removed if support for migrating\nfrom older versions is discontinued.\n\nCloses-bug: #2097074\nChange-Id: I52b37cfa49cac8675f5087bcbcfe18db0b46d887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/098309193a2a9ae07a386522925c3502813c1509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/098309193a2a9ae07a386522925c3502813c1509"}]},"branch":"refs/heads/master"},"965cc2fcbc18c025639c61e945bc211bf9a2783b":{"kind":"REWORK","_number":5,"created":"2025-02-05 22:09:50.000000000","uploader":{"_account_id":36393,"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","username":"Callum027","status":"Catalyst Cloud"},"ref":"refs/changes/01/940601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/940601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/940601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/940601/5"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-01-31 06:09:05.000000000","tz":780},"committer":{"name":"Callum Dickinson","email":"callum.dickinson@catalystcloud.nz","date":"2025-02-05 22:05:23.000000000","tz":780},"subject":"Add per-container storage policy to account listing","message":"Add per-container storage policy to account listing\n\nAdd the storage_policy attribute to the metadata returned\nwhen listing containers using the GET account API function.\n\nThe storage policy of a container is a very useful attribute\nfor telemetry and billing purposes, as it determines the location\nand method/redundancy of on-disk storage for the objects in the\ncontainer. Ceilometer currently cannot define the storage policy as a\nmetadata attribute in Gnocchi as GET account, the most efficient way\nof discovering all containers in an account, does not return the\nstorage policy for each container.\n\nReturning the storage policy for each container in GET account\nis the ideal way of resolving this issue, as it allows Ceilometer\nto find all containers\u0027 storage policies without performing additional\ncostly API calls.\n\nSpecial care has been taken to ensure the change is backwards\ncompatible when migrating from pre-storage policy versions\nof Swift, even though those versions are quite old now.\nThis special handling can be removed if support for migrating\nfrom older versions is discontinued.\n\nCloses-bug: #2097074\nChange-Id: I52b37cfa49cac8675f5087bcbcfe18db0b46d887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/965cc2fcbc18c025639c61e945bc211bf9a2783b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/965cc2fcbc18c025639c61e945bc211bf9a2783b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
