)]}'
{"id":"openstack%2Fswift~940609","triplet_id":"openstack%2Fswift~unmaintained%2Fwallaby~I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf","project":"openstack/swift","branch":"unmaintained/wallaby","topic":"zuul-config-error-swift","hashtags":[],"change_id":"I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf","subject":"[CI] Use defined version of probetest job","status":"MERGED","created":"2025-02-03 11:32:05.000000000","updated":"2025-02-03 13:58:27.000000000","submitted":"2025-02-03 13:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940609-zuul-config-error-swift","meta_rev_id":"425ea0822c758bf26809d90a168a0193449f2a37","_number":940609,"virtual_id_number":940609,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-03 13:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-02-03 12:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-02-03 12:46:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 12:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcbe8aafec356fb992c93f155c2dc5936eb8db4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-03 11:32:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c11cc7a19b160dddd427d543e67895c4b75ad98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 12:42:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9833662d149b4fa38378ea630bfd96ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1276ee7a86e046d5a96c76f816473842 : SUCCESS in 7m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5754b42518d24896a71d8cf2e12a563d : FAILURE in 3m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6ca0b4cf3b824a79b51a4aa6da086f0e : FAILURE in 3m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a1f5919632e4bc6a05143b60e201dea : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a98800ceae341c6965567050697b35a : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b198a8ce7d1c47ebb71918947c782e2b : SUCCESS in 15m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8c547c3d09904f1cb82688a8ec117103 : SUCCESS in 15m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6ff32e3fb834458a81b7ca4b107957e2 : SUCCESS in 14m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4909165e8ccf4a399300a515d3c12061 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5350b1f8febb4a15822d6f22c031aea6 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/de5d13bf58d54d23ac6f8a09e28c532d : SUCCESS in 17m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f743ab19a941467eb9d531b690d1ad9e : SUCCESS in 21m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/588323902b104386b1a9adb92313c176 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fcaab366307f4ebe9e10be2a0b659bcd : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7870ae522594140b9c8f4735663bd53 : FAILURE in 5m 37s (non-voting)\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/82632f3c5fe24e1da4d82f0ca6e8be62 : SUCCESS in 45m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97fc761871934b42ba76b00743a964ff : SUCCESS in 44m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6c5d35b0bad40a280f3ecb6abfd1c27 : SUCCESS in 50m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5368815dc814466080b0ac19ef5bffc7 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c21e3026391d4f8ca49bb7bce33dbec6 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a19e0bb983449e28d370ba44fe96f67 : RETRY_LIMIT in 4m 13s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e1c67ef710f4968b759c78305721831 : SUCCESS in 37m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e742d2541675416a89c4b10db1c9432f : SUCCESS in 53m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d0b4aa3177a246f2a394e1cfe600e906 : SUCCESS in 1h 02m 45s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d717fc3c06fc40af9716dc0e88052421 : SUCCESS in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88c254be99e648e3cdb8a54b7779c85197497dcd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-03 12:46:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4d9aba1cba5f3b6430da5753fde43bbc6d74d03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 12:47:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"108e79cafcbf5d5f8a06e3c01200c787d4fbaa69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef9a15b4023d4915b435d0402b884d47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a883eee49634b38b45268625b764a58 : SUCCESS in 7m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0c263bc7df8f4ec8b075c3a4e2143daa : POST_FAILURE in 6m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cfa45b2826ea4c33aaeb44dbf5039618 : POST_FAILURE in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c5cf32116734099bff64ddda0314389 : SUCCESS in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/510eacd12e394f26a4e9dd663327617f : SUCCESS in 16m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/99544b0df2dc4df8998332a462bcf75d : SUCCESS in 8m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d95f5601db9e42fea01260bcb88578c9 : SUCCESS in 14m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a9e13d9980c4369aa41f7ccc8820e8f : SUCCESS in 15m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a28a9fffacc4515b85c292babe502ef : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f523681f79b411594169b117de398d0 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c2d44fc230b4c21a598ead30e31fdd4 : SUCCESS in 18m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f369a8f959d2493988264a804d64a1c0 : SUCCESS in 21m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ba7c3ee328394ea0aa40a10b7f7a42fb : SUCCESS in 17m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7c746c657e384768ac3f8afd10d8bbe8 : SUCCESS in 18m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/762255d31ab64ba196a2ea28acfdcbea : SUCCESS in 40m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d235d424244433bba592b946fa062ee : SUCCESS in 33m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bcc6fad8440d4a6fbb9d3f512e27ba51 : SUCCESS in 20m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34006db07d9f4cc2a53fed8c67eaebe4 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e294285d14042d7bb22c3f575328f55 : SUCCESS in 5m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19225983ea9545b584be43b8f0cac839 : SUCCESS in 43m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5b0caed358d45bc9b9eb83a1ac4d73b : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/088c746180484c708a228c5fb17c68c4 : SUCCESS in 41m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81033b10dcb0ec992615ba23d8359d12e39d7d2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:56:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"425ea0822c758bf26809d90a168a0193449f2a37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:58:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423452b25e4f4a91aab3f416f7f26561\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d113fb8cff946fb8d6cdb84e9a9c5ed : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/36e4c57e1ea4419ab432780aad152c38 : FAILURE in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c51d1fd9f7df42e79982399722b56b10 : FAILURE in 26s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69da1ab55a3be2f4f22a653397e0c8c145e040c5","revisions":{"69da1ab55a3be2f4f22a653397e0c8c145e040c5":{"kind":"REWORK","_number":1,"created":"2025-02-03 11:32:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/09/940609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/940609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/940609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/940609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/940609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/940609/1"}}},"commit":{"parents":[{"commit":"ca53d378ca31ea53116958658b10ff686d1a5c60","subject":"Merge \"Update .gitreview for unmaintained/wallaby\" into unmaintained/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca53d378ca31ea53116958658b10ff686d1a5c60"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-03 11:29:55.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-03 11:29:55.000000000","tz":60},"subject":"[CI] Use defined version of probetest job","message":"[CI] Use defined version of probetest job\n\nswift-probetests-centos-8-stream was replaced with\nswift-probetests-centos-9-stream when centos-8 nodeset was removed from\nzuul, except one place where accidentally the old job was kept.\n\nChange-Id: I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69da1ab55a3be2f4f22a653397e0c8c145e040c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69da1ab55a3be2f4f22a653397e0c8c145e040c5"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
