)]}'
{"id":"openstack%2Fswift~940626","triplet_id":"openstack%2Fswift~unmaintained%2Fzed~I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf","project":"openstack/swift","branch":"unmaintained/zed","topic":"zuul-config-error-swift","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-05-07 18:05:28.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf","subject":"[CI] Use defined version of probetest job","status":"MERGED","created":"2025-02-03 14:05:49.000000000","updated":"2025-05-07 20:51:07.000000000","submitted":"2025-05-07 20:49:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940626-zuul-config-error-swift","meta_rev_id":"c34d04a997b4a14554db73f963f8066b5faab913","_number":940626,"virtual_id_number":940626,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 20:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-05-07 18:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-05-07 18:05:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 15:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-03 16:44:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fcb1b3313eeca928985a2228ecfb7bf8f816c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-03 14:05:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdbf21e4dd0165d53ec6376bdabf47c13317ae46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 15:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e8beff5a1344699aebd63ed47ee6ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c722f7321c04081b324796206064637 : SUCCESS in 1h 19m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6a9c642e81840b19ec36d3d847bf551 : SUCCESS in 1h 14m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8e68fe6513940718bdebd9fa1e651e4 : SUCCESS in 56m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5942716c34214015b9cf0e1ca6d51508 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"888ecbd6c5b971f11725d87272f78d0fc938868b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:44:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed00f311872f45ce9d2072fd1981c9ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f3e70080a1487490a48039cfbb4afc : SUCCESS in 10m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44fa01fe03d7455aad9f1d791e6a5679 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d72460d38684dd18b3f328de13243b3 : SUCCESS in 14m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cca38ab32044be0be4bf6cd63915aba : SUCCESS in 16m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39cc2266b9a04138b294e5cf664b8450 : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/302141bcefb34eaaae259e58d9a7b6c8 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6def9b7d35334f8097525a25cef29241 : SUCCESS in 22m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e3db7bea793451bbe2e0ebccb820f50 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d43d60db93842b79cab63190a913404 : SUCCESS in 15m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea21c660d0e747598cd9ea04ca96ef17 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7143749b02374058bbec3b31a0cf9d50 : SUCCESS in 11m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd79d0464f8d4be0b9a9b4c752e4bf84 : SUCCESS in 17m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5b154523d35949bdb4c99f337f293dbf : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f0a01dc20da84a9d8f3235d88d1fb2fe : SUCCESS in 13m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/db5c05f38bc5450487c5893ea221b771 : SUCCESS in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5ca348e317d4a8a9f34ecfee24ea21d : FAILURE in 4m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84e3376cc92e4d9188084fc416915596 : FAILURE in 15m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d930417a62742d9be4b4eee0e9504c3 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b78ddbc573f54ad49b009a5d4b412eb7 : SUCCESS in 57m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/097d110a9d7e4bfbb82882f93cbde41d : SUCCESS in 54m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fd0d656c12541e99352d30a942af9b1 : SUCCESS in 58m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9352c4cfb8084fba8759816b91fc28bc : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7588d5cf78f84d82b44a93324c6ac19a : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03d9e2f32ba94cfe8582c29ecfffa32e : TIMED_OUT in 1h 30m 51s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0ffd6529e4742c4a2d9f3a842a35b52 : SUCCESS in 46m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1f0543b790e4a00a4da94b03d5adc84 : SUCCESS in 48m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/6aee726884ea4a0c953515c6039ec06e : SUCCESS in 56m 30s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f57e20c54f314146a5e84d3e8d809004 : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5ac187f9d4108ac3e82d99ae17939b79b507492a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-07 18:05:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e69f71b3413c1fbebecc70dc0abdf51db2f94d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:46:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97911fa265044125b988a2934b7f8dea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b745a2442ac42f49ff3f09c2a160438 : SUCCESS in 6m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/797b945e5e70469e8d4a473487ba6b1f : SUCCESS in 11m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b62d8cc1c3a9497a8e11adca5607cf8d : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38cc899905054789889b2ebea58755c0 : SUCCESS in 23m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc4914276d4b488986086e73a993a7d5 : SUCCESS in 6m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8d5b3a2fb748444a8b9905a5856a985c : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/399054ee24ba4e05925073d5cf14eabe : SUCCESS in 6m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6945204791844ce92b4365f7384b3e5 : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d1ddc84252342cead2744bf95567ab6 : SUCCESS in 14m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b60ab60412874dfab8bda85783c34cb8 : SUCCESS in 20m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc6db6fbe1ae4a99a8778061c9c35e77 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b02043717e784d3c9a9f830e56af4a1a : SUCCESS in 16m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b9ed5fd66da64990881138244c3458cc : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e6a511ab54fb46b5b83ff5ffc012a645 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e30923478aa043379bc62120364fb070 : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ddefbd3e9f9e4e4e804ad519fc601914 : SUCCESS in 5m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/151003c734b748fd81f6b6f6d21cada8 : FAILURE in 14m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2de3dd8681004b5ba7224e742e4eb4c0 : SUCCESS in 5m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f7264782c44465aab2e51c05d77e1e0 : SUCCESS in 56m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03462fd44bef4a8daeaac3eb1c16b103 : SUCCESS in 43m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a8eb5e6df4c48c7a1e2a86b5c376d20 : SUCCESS in 47m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/553e0c0b2f834cbf86d9b6d1ba48b987 : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681d17285cde472b8ed0615890f3a050 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e69cb683524f46959c6efc7a6b294480 : TIMED_OUT in 1h 30m 54s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d100d1d0da374941adaf79adc2082f68 : SUCCESS in 35m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4a46468c35d41d0b970d9d04335a65e : SUCCESS in 48m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/250dc4691a5d4d7d8a24574f9f96d22c : SUCCESS in 56m 58s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6244e1c896e04a41bef147e91b2fd64b : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3edc0eeb1ab795c6f6d1f4afa252c4cd5edf2f9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74faebb1694f87115cbfb6c592527bd0b82bbe1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 20:49:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb096f5d1d04f75a6e757f73062099f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81cab4486c9e47e0b0f243e20fd1b4b0 : SUCCESS in 9m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1c522db43f4a4201993784321bfbb6d1 : SUCCESS in 11m 31s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ffda13ea312942dcbf2c04e94f54552e : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68fcb0806a1246beb62e4a8f5db7ceb5 : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1256ec55af6c454b902f773ec804ce56 : SUCCESS in 13m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2153663e0f4427a745b8483bd54a3e : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/190e24c381bb4a3484783d6211aa1246 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a443f795e764da693058ec0b0d093fd : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cdb2797e6b5c47beb22fcf54d7674865 : SUCCESS in 17m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0138c752cf3e4c86af63493d9f9fd744 : SUCCESS in 10m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa17580db16c4c1494a0a330ebd50d1c : SUCCESS in 14m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b9a57333d21416ca1a8d5cba577021d : SUCCESS in 12m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8b013389b6a4f8780a390e5dd558d1a : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4a2aedeeaaa4afc96795a476bc5a689 : SUCCESS in 5m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d23ec4c2eb334bc0aa75cdaf92f86eb7 : SUCCESS in 42m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c11eadc61fc4221a7856e51f9c9433e : SUCCESS in 51m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88722da093c241a3a008af2a3d5a6a73 : SUCCESS in 19m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32e4614a16c7424a9aaeb1063a780615 : SUCCESS in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c053f65da78046f794ed454be31f067d : SUCCESS in 6m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db5ad2456fc04264b3259c05a6a8f508 : SUCCESS in 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aebdd64748e45f39fbd2acb9afd69cf : SUCCESS in 47m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/259c87c52c3f40e68e7b5f836ad61a2c : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7c9a225968518b7f322d2a53e79ac02a23b11eaf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 20:49:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c34d04a997b4a14554db73f963f8066b5faab913","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 20:51:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ad9c5442dd46ddb4aa252283a03182\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4b18d16a9c48868874a885b5b90461 : SUCCESS in 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/36580d8f03e5486388fdbeddebff45ab : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0f203374ce934ae5839938d1a7fde4e7 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90cf3cdc6df161088c0b88b8f791c6a2ab7f67fe","revisions":{"90cf3cdc6df161088c0b88b8f791c6a2ab7f67fe":{"kind":"REWORK","_number":1,"created":"2025-02-03 14:05:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/26/940626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/940626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/940626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/940626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/940626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/940626/1"}}},"commit":{"parents":[{"commit":"6f318202892cf34b2c1c9f834b7567ac2b4b7f64","subject":"Merge \"Update .gitreview for unmaintained/zed\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f318202892cf34b2c1c9f834b7567ac2b4b7f64"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-03 11:29:55.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-03 14:05:38.000000000","tz":60},"subject":"[CI] Use defined version of probetest job","message":"[CI] Use defined version of probetest job\n\nswift-probetests-centos-8-stream was replaced with\nswift-probetests-centos-9-stream when centos-8 nodeset was removed from\nzuul, except one place where accidentally the old job was kept.\n\nChange-Id: I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90cf3cdc6df161088c0b88b8f791c6a2ab7f67fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90cf3cdc6df161088c0b88b8f791c6a2ab7f67fe"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
