)]}'
{"id":"openstack%2Fswift~940628","triplet_id":"openstack%2Fswift~unmaintained%2Fxena~I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf","project":"openstack/swift","branch":"unmaintained/xena","topic":"zuul-config-error-swift","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-12-19 15:00:55.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf","subject":"[CI] Use defined version of probetest job","status":"ABANDONED","created":"2025-02-03 14:06:58.000000000","updated":"2025-12-19 15:00:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6008caba51e2053fa699674ddf10c64b049d177c","_number":940628,"virtual_id_number":940628,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-03 15:20:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 15:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79b4470cd4d6354522471dd2a8ead79be1b44771","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-03 14:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"397f95dce2254dfb1d4a2ceafaf00a1aee689fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 15:20:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe4025ee9714b0aa728861aa0e50aaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58af396062141dab6437149dfa7a90c : SUCCESS in 7m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f66f58c7eeb54f8c98b34be4c2a7510d : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7899b14942a94c1cbdf16ce778eaf258 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d92baa2e9f764a3893a2167e847a7526 : SUCCESS in 15m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cf9fda22ce94d829ec2237f294df566 : FAILURE in 4m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8b292fc3b5a142228fa7839e8728a673 : FAILURE in 3m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7eb0a7d09f44b2286c626c2b9d8b602 : SUCCESS in 12m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5365b98b6d48508d229cd6d1e0ab37 : SUCCESS in 16m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b692d169d6c7493ab12960198c1f80ca : SUCCESS in 18m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f61606008f804dceadf47921a9b3819c : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14f6dfe809dc4df4b5bfb85fb32002f1 : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e585cc7df2a4df4920536f82552f657 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82ed43b51a8742ad852dad458776fc78 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1e446e4168142859ed5f11350186408 : SUCCESS in 17m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5b898808f498433180c284a46f8a0934 : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c52e211ed7164feeab3424077a6800f4 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d19125c5cba4e7b9d2f2bf710190609 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8d379fa020741f2a1805c91601550e7 : FAILURE in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/665bdb4f9db94985a48bd8cbdaa46dba : SUCCESS in 53m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0eb4ffa58aff495e89df2e6b39f10c36 : SUCCESS in 37m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ad57625a4b54184904e76cfe9ffd0b3 : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3c32039a87842bd9b49fa382b7b7f11 : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404a6063e0ce46f8853df19bf6567dbd : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43b1bdc8359840418f884596cd859225 : FAILURE in 7m 16s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7603dbe83f884bd48d99311a721a16b6 : SUCCESS in 40m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b085089bc7fb4ead9832aa1dedb9782f : SUCCESS in 43m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/c2115fccefe6484db73b3508fff02026 : SUCCESS in 55m 52s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bab54960b92468b9c8ace96a35cede2 : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f8ec3c69e7d36dc85c068b8a6a82fc80716c06b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 18:26:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/188a732da8eb40deb7d58b3fe6e9dec9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca4cd3a48e15479db51f8c74f5b4d7d2 : SUCCESS in 59m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f8d25db928d4dd7bd28ee8b4c2c6168 : SUCCESS in 1h 07m 16s\n- swift-probetests-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/4c08f9faaf02439ea33e82a4d28eb005 : TIMED_OUT in 3h 01m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/04fe2d74ed434205a24de99378e2abc1 : SUCCESS in 1h 27m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6aafc4d96294e1a8c0d82034bbdca0e : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6008caba51e2053fa699674ddf10c64b049d177c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-19 15:00:55.000000000","message":"Abandoned\n\nunmaintained/xena branch of openstack/swift transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e94a94f3e04008c1576f570b541f102590bc4957","revisions":{"e94a94f3e04008c1576f570b541f102590bc4957":{"kind":"REWORK","_number":1,"created":"2025-02-03 14:06:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/28/940628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/940628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/940628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/940628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/940628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/940628/1"}}},"commit":{"parents":[{"commit":"4e35060889940f237b2136b7c5291fada9c875e5","subject":"Merge \"Update .gitreview for unmaintained/xena\" into unmaintained/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e35060889940f237b2136b7c5291fada9c875e5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-03 11:29:55.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-03 14:06:44.000000000","tz":60},"subject":"[CI] Use defined version of probetest job","message":"[CI] Use defined version of probetest job\n\nswift-probetests-centos-8-stream was replaced with\nswift-probetests-centos-9-stream when centos-8 nodeset was removed from\nzuul, except one place where accidentally the old job was kept.\n\nChange-Id: I71edd6f3fb3d7141978ec5ed8351cfe24783e3cf\n(cherry picked from commit 90cf3cdc6df161088c0b88b8f791c6a2ab7f67fe)\n(cherry picked from commit d8a309c756a3d9ef96812b7012f0a6a4c60f6e82)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e94a94f3e04008c1576f570b541f102590bc4957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e94a94f3e04008c1576f570b541f102590bc4957"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
