)]}'
{"id":"openstack%2Fswift~940755","triplet_id":"openstack%2Fswift~master~Ide1271dc4ef54b64d2dc99ef658e8340abb0b6ce","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-02-10 19:11:00.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-10 20:21:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide1271dc4ef54b64d2dc99ef658e8340abb0b6ce","subject":"tests: Fix flaky reconciler test","status":"MERGED","created":"2025-02-05 00:12:25.000000000","updated":"2025-02-10 20:23:30.000000000","submitted":"2025-02-10 20:21:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940755","meta_rev_id":"4cf480e421e4724c6fba01c408abf390822f3969","_number":940755,"virtual_id_number":940755,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-10 20:21:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-02-10 19:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-02-10 19:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 01:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-05 19:56:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"c0ddeb2de337aedd9229a92be92a7c240d0de8d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 00:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7809b905a24a1edd1e926eeff5d129e0c0cf0f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 01:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca607ae84acb43818eecd46db2de000b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd16e925a78342ca94e635fbfb8abccc : SUCCESS in 8m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a39658b6ab1a49d2aebfa0eaddfe5b9b : SUCCESS in 1h 01m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f65b07539e47471da3d7afacc244d599 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b18bdc36a25540ed8f3292823e56c647 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94d10c814c4a453ebf1174a395b83ae0 : SUCCESS in 22m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f57c9e0273674e12ac2d4c68b1faf313 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cba1cb25877c4333a1c5f841ab20a485 : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e348efa97e47461289e3e8e5a1968db4 : SUCCESS in 9m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb9e78069a4248a69566178aa4982bf3 : SUCCESS in 13m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c905da1c680840d8b19a1ebd12e9cedd : SUCCESS in 8m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff4bd5779f2f41aeb4ec2267f3143d60 : SUCCESS in 9m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c0b812209a9b4b1b9b6583982ea5eb87 : SUCCESS in 15m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9deb9e4719c34839a68f6e76fbebae59 : SUCCESS in 25m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2e013ab11b746e593c0b84dc2014614 : SUCCESS in 51m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/396d8b935e5d451dbde221033f72d157 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8ea342fe0640f4a78dc46cc3123386 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/895fa70bd22e449b8f62e45d47d01efb : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b8d0b81b25529502d445c2c5cb0235c3e018a091","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 03:06:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8e9386e477b4feb9da0438cc37de1b0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cf18e56eb9e436abfee983e7ff90248 : SUCCESS in 2h 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0569c44ad9fe91206d372d572d26d8c14e0783b0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-05 19:56:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c26b7feed74b681316b58a20d79845c235b6495c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 20:53:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"39cdfe827dbf70972907a49c982f0cd534c6716a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 21:45:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97fe8757d107477e9a5f0d03f896d445\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b2e1549eb54787be846b03107bf642 : SUCCESS in 5m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49f184c2887945958ec7f9a1552a9c9b : SUCCESS in 45m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33a56ef21ce04a4ea9f12fc7d352346e : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95001267d1204f1490d8bc7b0957729d : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f0eebe64af74cf3a1df7f690fde351e : SUCCESS in 16m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a00091bf0f3e45deb0720349e19f1b99 : FAILURE in 3m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/903506a92d6a4ed3a2d94cb70e764ba6 : SUCCESS in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/49b8742774d74bdd8a0489e64c68cf89 : SUCCESS in 17m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a469243ca8784e779cc156310dbf5d29 : SUCCESS in 20m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2ad22e8f5a71459e9b289e829e77d0b2 : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/036bafec9bdb44bc9806581b2174a1a2 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bb17beaaa7dc4043b81f550e536e6a56 : SUCCESS in 17m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/daa347d0ca2d4568812ddd0fc0dd9ae5 : SUCCESS in 49m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89cef103cfb04aab857ad76cce39fed3 : SUCCESS in 34m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cf24f1171a7455e8c52c2eed8d14307 : SUCCESS in 18m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6aabe0a193480eb423176e88d84e75 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b00c35e4e09445e1a9bdea8ce3796353 : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1ada710ec6dbc6c63e85d37c9d00677c06820de2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 23:06:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fadd1fe930144fcb8f952492879d6c7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a226b898cae84b679de18e4588b206b2 : SUCCESS in 2h 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"470df591075f7fcd51a1c2143bbc64bc3e6a2b2b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-02-10 19:11:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc3c87c9c8fc88d8d5075fffa30d5c7d2580b1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 19:11:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d11ece8e1cedcbad21021665d425560d67d871b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 20:21:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7ed483dc6143a2b8a1be0aea2f460b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42568460c3024d3fa69e0ef325ad1f86 : SUCCESS in 8m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/badb7fac8fbe4782a0b87c79aaadae82 : SUCCESS in 1h 00m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e160af6d9be84e2f8bff3d95c40cef28 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af66593f9fe6401aa4242ff06ae5b70c : SUCCESS in 18m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6e285ba671340d5b25c8948ffecea3f : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/02225dd0e3264100b78b69b77a157cea : SUCCESS in 30m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/832b58cbc58c46029080f85b28b3dbaf : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cba93fd7c7be4d7891504e32da56a2dc : SUCCESS in 30m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d220a569854e4cfb9fdbf2096e41e558 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a112f99f48de431e8342c9fa7655705f : SUCCESS in 6m 34s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9dcad645c97547a1a1f6c8b0805c1003 : SUCCESS in 7m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7d2f982842f46788b0906ae959b0bce : SUCCESS in 32m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a36a0454d824000864fbeab33d9d146 : SUCCESS in 36m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b2ef83584643b7be9bb76d9736eef7 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047181369ae041c08100b5b789a6add6 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/589992189fb14858bdbe94f8acc993cd : SUCCESS in 24m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7250867f5d6d86fe3fcad69bac7af58d49660a9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 20:21:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4cf480e421e4724c6fba01c408abf390822f3969","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 20:23:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e247b0ba91406984afcd7a5f58d174\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd8e3b8960d74f12b607d7cf0acf41e5 : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/75f98cbf34904037ba1151737b2d65c2 : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ccf749043555cee77bbec89127546a0e0b04419","revisions":{"a587cd162654e489ce806fcd6697376210958be8":{"kind":"REWORK","_number":1,"created":"2025-02-05 00:12:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/940755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/940755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/940755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/940755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/940755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/940755/1"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-05 00:07:51.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-05 00:07:55.000000000","tz":-480},"subject":"tests: Fix flaky reconciler test","message":"tests: Fix flaky reconciler test\n\nPreviously, test_object_move_no_such_object_no_tombstone_ancient\nwould fail intermittently, with an assertion that two timestamps\nwere almost (but not quite) equal.\n\nThis probably comes down to the fact that it\u0027s passing floats as\ntimestamps down into FakeInternalClient\u0027s parse(); specifically,\nvalues like 1738046018.2900746 and 1738045066.1442454 are known\nto previously fail.\n\nJust fixing the usage doesn\u0027t fix the foot-gun, though -- so fix\nup parse() to be internally consistent, even if passed a float.\n\nChange-Id: Ide1271dc4ef54b64d2dc99ef658e8340abb0b6ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a587cd162654e489ce806fcd6697376210958be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a587cd162654e489ce806fcd6697376210958be8"}]},"branch":"refs/heads/master"},"3ccf749043555cee77bbec89127546a0e0b04419":{"kind":"REWORK","_number":2,"created":"2025-02-05 20:53:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/940755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/940755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/940755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/940755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/940755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/940755/2"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-05 00:07:51.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-05 20:53:02.000000000","tz":-480},"subject":"tests: Fix flaky reconciler test","message":"tests: Fix flaky reconciler test\n\nPreviously, test_object_move_no_such_object_no_tombstone_ancient\nwould fail intermittently, with an assertion that two timestamps\nwere almost (but not quite) equal.\n\nThis probably comes down to the fact that it\u0027s passing floats as\ntimestamps down into FakeInternalClient\u0027s parse(); specifically,\nvalues like 1738046018.2900746 and 1738045066.1442454 are known\nto previously fail.\n\nJust fixing the usage doesn\u0027t fix the foot-gun, though -- so fix\nup parse() to be internally consistent, even if passed a float.\n\nChange-Id: Ide1271dc4ef54b64d2dc99ef658e8340abb0b6ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ccf749043555cee77bbec89127546a0e0b04419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ccf749043555cee77bbec89127546a0e0b04419"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
