)]}'
{"id":"openstack%2Fswift~940762","triplet_id":"openstack%2Fswift~master~Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa","project":"openstack/swift","branch":"master","topic":"heartbeat-resp-on-COPY","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-05-09 17:52:33.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-05-01 01:16:41.000000000","reason":"\u003cGERRIT_ACCOUNT_37337\u003e replied on the change","reason_account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}},"37337":{"account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"last_update":"2025-05-09 17:52:33.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-09 16:45:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa","subject":"Implement heartbeat response for COPY request","status":"MERGED","created":"2025-02-05 04:40:05.000000000","updated":"2025-05-09 17:54:01.000000000","submitted":"2025-05-09 17:52:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":53,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940762-heartbeat-resp-on-COPY","meta_rev_id":"d5f8ba1d854d0ea9f392007e5ade470c5f775924","_number":940762,"virtual_id_number":940762,"owner":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-09 17:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-05-08 18:26:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-09 16:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-09 16:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 06:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-05 14:59:18.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2025-02-05 19:21:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-03-24 02:35:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-03-24 23:16:25.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-24 23:18:36.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2025-03-24 23:25:36.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-01 01:16:41.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REMOVED"},{"updated":"2025-05-08 10:26:03.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"8f98deb00ab3e0f413300b6ce297d21d1323874b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-05 04:40:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e251319266b96939aab4843187926c4a8bc2f8c","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-05 04:40:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d5bca14b8ea331bbfcf158d961d2c301fea7bc37","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-05 04:40:31.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_37337\u003e\n","accounts_in_message":[{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}],"_revision_number":1},{"id":"a03a61e475f5b303ef3b51263c920cec1bafc5f6","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-05 04:54:10.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e06d8686096fd282cd757da2c494632760dd34d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 06:38:18.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c38c63101d143a1b66c78f7ae55d7c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd3c65e7208b4894bab547adc4667bc9 : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9003e4341c8347d9a219ccb6cb488534 : SUCCESS in 59m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3a5990a7da741be9af91007659e5143 : SUCCESS in 1h 04m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/453b86742a4644f581ee4a914d83a0d1 : SUCCESS in 1h 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3da1a0d21664ae7ac5f7bf25024ab4e : SUCCESS in 40m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dce3c65d614b41eba7f8155246f725dd : SUCCESS in 12m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9466b3884c3d43eebeede34eece29732 : SUCCESS in 12m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/921a1575ec0b4dca903781b2ae1d9b89 : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52a89e1160ee4af6bed9ccf9a1641cd8 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1af85a17353474e906e292ff1c9c480 : SUCCESS in 14m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/70d4cf91138049339a36798d1663d5eb : SUCCESS in 22m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68097edbef2e4ddf86d3494f46e732ef : SUCCESS in 33m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/54425383f05e452b85702cd6c7d2a719 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5f87d913bb5f473e886fad02aab3db7c : SUCCESS in 9m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/64b190a7650a437a9d1742b29b8cc8b6 : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9c3366aeee08452c96bdaad3a88ef8a9 : SUCCESS in 7m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6046a6ae4e854861a62cf09a56e70e2f : FAILURE in 13m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fbd6b695498541418e4be0e34b217ad1 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/55cf9ebc93ac46dda127c7162304367e : SUCCESS in 52m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/387a46f085094b53a7e9300b6267d688 : SUCCESS in 50m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3f687dbc9fc44d49e4929a34e2cd836 : SUCCESS in 33m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/885720794a4349baadb1f0b75362a31d : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc6168687f045f289092813e005b721 : FAILURE in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06eeda5adc724b388eaeeeb00fe46883 : SUCCESS in 18m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b386e117f5e849b8beeb46ba48c9968e : SUCCESS in 53m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e5e4f9eb99e4e76be591b969b3b0ea2 : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9685aabbe01cb11f4a4aad486f77d39e3e40eaf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 07:04:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755cfe0e7b1444d6acff301b99eb76f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e64cb30557342648d6997b263526637 : SUCCESS in 2h 21m 48s","accounts_in_message":[],"_revision_number":1},{"id":"49dfab6d430ac67ce6a98c6d28e0d49f6e131424","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-02-05 14:59:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6c82d7d919cb5cacaf262d464de747e53554630","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 19:21:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2835ae2187864c00584af850081b0fd5895e4dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-06 04:00:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b617f584ad2b1766c21b56bc14ecff6794de1913","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-06 04:02:09.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7a58fe4ffc9fdcb9fe43ae74d53aa3644878e126","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-06 04:04:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c96db9511faf53250545860f35442d38232dc445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 05:12:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9390036422244bbca269ecab5098480c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908cf6737b1c4dc38c0baa4edb3167f8 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/051fd7958ca844039cb9c9c80fc9acfc : SUCCESS in 44m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4324caced52a4481afe975f7c0a1117e : SUCCESS in 34m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48940613b67942228294b02bc7993e13 : SUCCESS in 54m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bd978fe53eb4b2cbc0650596360cc92 : SUCCESS in 51m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34fe646d9a094b74a10085174a029601 : SUCCESS in 12m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b02f0db06c54fd7bb4d31aebf60a049 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8770dbe70227402aa6328e274148ca29 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a5c2e48a70142dc8d9db93685b3257e : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b7006478cc40eab17e6b98e928b0c5 : SUCCESS in 19m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2cf97614793e4b77aca06bfa4803d9be : SUCCESS in 20m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d1f50e00459848b9a264d2ca1d814887 : SUCCESS in 26m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ea97bc70fa2a4f7e8ba7fd26e9e2197d : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8753f835943c446bb11cd471f7050210 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d2102d4d1d4401aa0a10389a53c325f : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0023420d879d40b9a9ac402956fe25b6 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b2a3ca7fa12464bafb15157c9d409fd : FAILURE in 14m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c4cfc5267e094f6ebc574c2d47d1be54 : SUCCESS in 8m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/41d30dd7db07493c90e1d3f510f0d8c5 : SUCCESS in 49m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4651523e4f264d04a46899c9092bd5c2 : SUCCESS in 30m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ef935a82544492e81df065684f9cf71 : SUCCESS in 31m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5928c39bef044e7d86b60548b2778eb8 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0572c0b0b2c4c7ab2f8f9d5b7f850e3 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd4a9b11b9ae4f3db8c90b31a506dfe5 : SUCCESS in 19m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b7a5bd8940346f6bf921a99e1dd5c6b : SUCCESS in 37m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/316375db350243f384e707916df7cf2d : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c78283475cbf7e217593e20ea8f041b600148667","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 05:19:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d7f89b7c6ed4d11a61ed666cdef29db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c2356e217f54fa8b04dfac5fefad46d : SUCCESS in 1h 15m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6e350273c8c30ffa33bcc7555fe2ef873bb54271","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-06 20:44:58.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fa36b59a6fe2fb7e46d8eda2627f7f17d49152c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-07 05:36:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5931c9ea171e80ea2eda5600b652c7f910463ed6","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-07 05:43:47.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"adef7ca38c21c1789a03c42a76831562de0b28db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 06:32:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c99b04ae86147eaa5aaad3b888a7828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4947046f0e1d421e8a20f4ba24133d02 : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/7f054f9525984da9af8d0b694763be8e : SUCCESS in 51m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0345805358d947ffa564952628db06cb : SUCCESS in 50m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14e3c1f0c8e841b184a38f05c8d84c3c : SUCCESS in 41m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba64cc6a98b84d32a885168f2dd095e1 : SUCCESS in 46m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9eeedccfe28840c8889e70cd7b10fab3 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4437b1c5ab5428aa0f10807320d6511 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/313cab087c344e25b14ae9e04e466621 : SUCCESS in 21m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f431c343f4ab44faa87461dcf913437a : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fecf4f5b1b34321a01f7c8a86f83428 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7116e8b523467e83d9c5c23dea0069 : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/153f9f9e9c32440f9e262efb0279c623 : SUCCESS in 13m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d0b19f94cfb742dab539be06dcccfc8b : SUCCESS in 9m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f89f782d7c644982a05da119b878790f : SUCCESS in 9m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8f1a51e997b0424a89fd25d2374d541c : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83ccfc7c35d64408ba319aa55de63298 : SUCCESS in 4m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad6e78d428254855b13b6f5ee7f8b3ab : FAILURE in 14m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f97d1b57b224a2499bd7dde6ac207d8 : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/365d0c07f25545ffb921c8a8b8543468 : SUCCESS in 37m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51335db9b00c44b290afe37ab7e3b2f5 : SUCCESS in 53m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1fcf61166f814414aa9eaa32dc8757de : SUCCESS in 49m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed4fe8b30f474092835d9ab68b25508b : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a56340e239c443584c8a25148cb22d8 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b73e6f5ecaa451fa9ace781019a717e : SUCCESS in 19m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/990ab85d01184c389cc553ed227483ce : SUCCESS in 39m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac0adf9c540d4b879eb26edf4c9bd3f4 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":3},{"id":"f45c04fb6b98ddfd1c489cdef513cb5fb07a44da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 07:07:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac86aabd5d641fab0d3d5f21ccbe23f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/981e1a78da1c4446bfeb7e786f34ccce : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":3},{"id":"622aa57423932fa53746f991837a0d301c81e8d8","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-02-18 04:13:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"881d1ab0eda6fa5b27aaace0b48bdb4ffc8efe65","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-19 21:34:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fa7a25e58925d3fac7990e0f769d7624594b23f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-03-17 19:24:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b019d38ff1a92ae1e22a666d5d0fb500e7bd95aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 20:00:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c761645090a34649aae0483c5c02f0a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/650d2fd5a06f4526aedee90144ef4c03 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":4},{"id":"cb909abcbea5d6b195afca2f83f203fe95c6b624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 20:36:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad63cea99104462d8817b2899c378f8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48954ab91cd24b9980ddfc5252807596 : SUCCESS in 4m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a1414ff3175b4302ae25b202b5fa392c : SUCCESS in 28m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d165f9b6d2304fa480dd873ed7241316 : SUCCESS in 28m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/072cf0c5808e4881a1f513b52430806d : SUCCESS in 32m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89a95112726f4ee999ae1ff4b0d0c0f9 : SUCCESS in 34m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57189e1a06d641a488537ae32ee9b5bc : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30d09eaaff5d42e0b2e66e826bf21557 : SUCCESS in 10m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12f13506797945999f6843c1862c2518 : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/740625d35b9b4b3bad6ed88d397042b4 : SUCCESS in 4m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/079827df457f477a872ebaccf3c04e05 : SUCCESS in 19m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2207ba6384b487f97d5498df99ed031 : SUCCESS in 9m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/632c96856024485cb7b2143aa7dc2f4c : SUCCESS in 19m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/065c32087f5841ffab36e911ac4a3277 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2363bcda4ee0475da6b0283002dbc372 : SUCCESS in 9m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad02e8f8e64a4dc793454674bf6c952b : SUCCESS in 9m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55b070289cfb4bbbacb3f7d0ab430f84 : SUCCESS in 5m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6bd462511534c4f8e635b8147c5ddfd : SUCCESS in 9m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/752fb21dc5414ec7a9424ab16ed26706 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/98d300841c3142bd88a128d208d825f1 : SUCCESS in 54m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7be58457281548af9231417da67b5ffb : SUCCESS in 52m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62d68981ad3d41b0945a0c1f55f77113 : SUCCESS in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2432c7a09f3a4325944416608927e82c : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ac5b0ca384435fb1f91b1e6e4c3467 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5c502f0aa02454abe87eab4089e3292 : SUCCESS in 24m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebe1e61d631042d9910e882cd4cf7478 : SUCCESS in 29m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/987eaa2bf9224ec4914deff9d3ab515d : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":4},{"id":"8b674cab03954036c24da9f79dff56d4982f520b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-24 02:35:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8aa6708e401c50c2f0fe7af360140228b9fc343f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-03-24 23:16:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b1fe86f951acf42c2884d506c7c6a432dda2fbdd","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-03-24 23:18:36.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_22816\u003e.","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":5},{"id":"70f9e3ac26949d2912806b5922c64b9a7e934987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:24:40.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f43197721ee410b8a2bceb54dbf80fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb302d03ddd8458882d5046c21f7f19f : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":5},{"id":"5c8cfb714f558316092808223beea9e1b3191df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-03-24 23:25:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e830a9acb49be99ef44a491a871da1d085f33caf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:36:26.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0bd9712227f4ac2ab08cea56e639d0e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b6ddc428363482294ce38ae3b988a36 : FAILURE in 9m 10s","accounts_in_message":[],"_revision_number":6},{"id":"b6206683f29dc6a39c5c367a5daf06591de4252e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 00:30:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bed842acef0a4cfd8ff3a89b63c7c7e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a350f8139a649f0ac2ede90a3e7abe7 : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/a460d7bfcbfb4962a755acb38c23288a : SUCCESS in 53m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dab97b402a7b408fabe9711dd453c0d2 : SUCCESS in 58m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b158be9ee5144c958588c6a22c139fbc : SUCCESS in 54m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab2f4feb93bf4e639e2e5c2682fece26 : SUCCESS in 59m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bed93dc93ed40f98c607c00e1ae8c35 : RETRY_LIMIT in 1m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f9e33bfc9b6499594b2ce7e94a95110 : RETRY_LIMIT in 1m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abec9c5fe57b458286e49d3e20aac75b : RETRY_LIMIT in 1m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01c3606157dd4ee6a2c004fbb2217263 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b41ce69f49c4f1b9fd2d32deb8580da : FAILURE in 6m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24af831f0692459394183694e818b658 : FAILURE in 6m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/78a149ce777c4dea963b9233c1563641 : FAILURE in 9m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/835296d459724fd89d858364a23eced7 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a672649fa3a4533bfb0d53a25c8beba : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5698364cdac64267acc22cb1d77aaaeb : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55268400ac9a48239f0033bd2cd1d5a5 : RETRY_LIMIT in 2m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/625b7a5710114b67a5857625481da61d : RETRY_LIMIT in 2m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/316ed75c4da14591b68dc8358a3cfdf3 : RETRY_LIMIT in 2m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b18680fc07524bc1ae9e508c28cb03a6 : RETRY_LIMIT in 1m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32f72ea99df4464985aea29b5cfcc020 : SUCCESS in 21m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cf603284d094a94b4b6b5f62e03cacb : SUCCESS in 32m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bf5475613cb49cea958595ad0134315 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e0f7825d723435b9abf742d4df25815 : SUCCESS in 2m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05e27576eac6444ebf7dec7eee0a0299 : RETRY_LIMIT in 2m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2caf15b65a44f19985e73eb859c16a2 : SUCCESS in 36m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/083b585da161458d9a75b7708e87a356 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":6},{"id":"13caba0edaffba2f398ef269f4b6912b2446ebfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-25 15:56:54.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"cdba107e3b90a6e320800cc3e95372fd702566fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 16:55:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2188ea58eede4acea0cd5d941a151609\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f178f506b744c1891fcdf0b189abd9 : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a70a3dc667cb4899a894776b55aa08b0 : SUCCESS in 48m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d64ac799fcf844cc8e631386038be8af : SUCCESS in 31m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c49f886dfd3d4290a3d7bc2a887b1df4 : SUCCESS in 34m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0b2db682ed54a49b23ccb256a576617 : SUCCESS in 35m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b72f0adef75402aba538e117e9eb654 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be7e5a41aeee4fa894a37b020452fda0 : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a07f8ff016b4591b94a10e8d281dd34 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0884925790764aff8cdf7d6628327f68 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5837bd47bbc45d2b9411cd34fceba9c : SUCCESS in 20m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cc4eb06a69c3442a8b9eb89cbb65ef5a : SUCCESS in 23m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ab366f1f05ce4a9183a242e034c82756 : SUCCESS in 28m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e194ab9e69d9432fb4aeeeeb741f3cf4 : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/297088c1fe43486c9b8766ac45aa2e7e : SUCCESS in 16m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/16c6d5f8b80b48149179267b3d977d05 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9610506417524bd39c6569308e060870 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7b02dd2bd67d4c4caec976de0acfed6f : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6062b4536d94372b3dabc7a42ef5eff : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c1a7bbdc0354e50a1117699d090f3ed : SUCCESS in 51m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fbff017dde1457196b75da834bbd1f7 : SUCCESS in 47m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/831d135b04ba4fdc95219660bfaa099d : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b257958d646041eeab1ff775a8d4a462 : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f0bdcd1cf6420d93b770bfef61677b : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa3a04a426844b77868d6d6538c83c9e : SUCCESS in 22m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/550b00de89f84ea9a1e1b49181e2062c : SUCCESS in 40m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f212f4128e18427d94c3e397c8a5d595 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":7},{"id":"c31f51d11fab6284a6b9080609e8c05a2e889876","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 17:04:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13c3be65bdf9455db28ef6a16af501d7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07f1b25adc20483aa4ffc55f09b6d475 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":7},{"id":"eb21760b81265c5a6c84ab58583e556876b43818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-04-04 01:09:45.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e4eae3f70ef10ab28cff5d971a80346f67c78a80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 01:18:48.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b3f4502e9854bdf8c31e96e096ea094\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23a511816ac946db9d9cec559771b2ab : FAILURE in 7m 11s","accounts_in_message":[],"_revision_number":8},{"id":"d8e577efe0389cd0d33485c704b1e6c1791c33be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 02:16:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec4d8dcc429b463ea8fcbddc5d99bb8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4634bd6e504202843616c9e9824988 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b7138cb44b18412aa2dcd272db2fe225 : SUCCESS in 56m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51b66560d4a842fe9757ee5bd90c66d0 : SUCCESS in 53m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d65ee7af46f046319a5b6d945b160d1c : SUCCESS in 49m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb029196552f418390bbacd76f7c44b9 : SUCCESS in 55m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d85de768e1214575974d9f432b006d90 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/148a152c27ea484fa045ef26f955e138 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97be10878bfc452dbae8bf97f1657e59 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba97180d9ec644b5ad8a99009ef2162d : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb7e724b06f247e29a3869dbe6d50da2 : FAILURE in 4m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38776862170c4d72baebe7de377409e0 : FAILURE in 5m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/191c38d37ec743cb8c516c2f6461dfb3 : FAILURE in 9m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/62bf905497ad4ddf84000d2a0d3af89c : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8b9749a7d8a4aaaa3d080de796ecd3c : SUCCESS in 17m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dcdaee74b3734890b0158124dd9b6fcf : SUCCESS in 13m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84a6ef8d4e1a4f81855df7b31ff9cc24 : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/27506e9e282c41daadf3a3c0ad61de75 : SUCCESS in 13m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83e234f3c9234b5a89ce2081b77125b8 : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1beacddda464daeafad1e8bd4397ce8 : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c299c0ac121b4f53ba74af4961767555 : SUCCESS in 31m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6bed2b85c1774d4fb988b96f4ada89e8 : SUCCESS in 47m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3afb4ac9e6834beb906b20549013e7ad : FAILURE in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a4b1dd435a49a1a34f112895d1eb59 : SUCCESS in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f3ab89a9335414dbc06f8871ee73beb : SUCCESS in 9m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa22c112205f48d9a0f91a63ea1bcc4e : SUCCESS in 34m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0856d2bfe8ab428aabb52b84757dbf5d : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":8},{"id":"a6070f00a375b9604d4fa01713cef0f6a360a690","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-28 22:52:28.000000000","message":"Patch Set 8:\n\n(9 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ef45059302fb69a295a7074cf8fd49d5d92267be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-01 01:11:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4ba1c448e6048d1db9ba44390473d78185265256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-01 01:12:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a60295ded4a76f0d27be8b03266163dcbdf61ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-01 01:14:37.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"12900d8d1001404d3adc1bcac48941431d04a978","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-01 01:16:41.000000000","message":"Patch Set 11: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":11},{"id":"bf40f7860b3e1eedafabea8c2614413dce02fc04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 02:16:35.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8632103178d44fee9d71be7fae48f914\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e07e5c7f3049ebac36fe777490a672 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c47596e93ff34da2930ddd18b4dc1c99 : SUCCESS in 56m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f35879b7e9db4d52a0f7840d6fbce90b : SUCCESS in 54m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be4d7b6bd6de46f69baf320cd2154f9b : SUCCESS in 38m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa4562df75964217ae60ebd1f4ae536c : SUCCESS in 49m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d8fe459d17f44b9bf6300f80ee15244 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16dd091be5594a43bba134361340d65b : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5166a00994564d9b920088a101538a38 : SUCCESS in 9m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be4313f1c4a04468bae6443e298039ad : SUCCESS in 3m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ba4222830a416e878b3cfb1e11aa44 : SUCCESS in 7m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d16253e2497477e8db20f1eee73e03d : SUCCESS in 16m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6ea5ba7f329f4bd88ffdd969fce0cd2f : SUCCESS in 9m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5d0906afb1ad46e8ac20a9645155e750 : SUCCESS in 21m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/61e07c6b15bb4690bd851dd20332d362 : SUCCESS in 7m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/82d58a7d1bd04e0e941c698082fdf874 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ece643b3fafa422dbd95b20f7ca4466a : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e1f892f791f4bc6a8ed981227035896 : SUCCESS in 5m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/675d58b833414ff5ba2466c1ed967b8f : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cefa63af20814469b4a35448cd14062b : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e99df9647c147c08b7e3988a04166b6 : SUCCESS in 52m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ac4ec2592d04ccaaf3c5fa86bf621a2 : SUCCESS in 40m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15234a7c6e7945238e2e6d8e65e168bf : SUCCESS in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67be158bbe694eb39fc70930e8b93665 : SUCCESS in 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11c68161152465fb9d92696d82c6a55 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/627f7150edb8465eaea135395479aaef : SUCCESS in 21m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/676928e1087a497f980052ba493855e5 : SUCCESS in 55m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2ec713b895841d1b4b033a74cd81705 : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":11},{"id":"65c4c40b75118d9a5824eb4c9919150574a3c261","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 02:19:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06796afa2c6c4a9f9bfe82b09f65c5e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/51811ceda32244268e248b590aa7bf8c : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":11},{"id":"0f0604d7ed89c3b444de99c8ce66d17e3f4fd279","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-05 21:10:55.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"adf74bab158b55247ed23dcefd0c2cbcbef13e05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-06 17:47:17.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"e8a993882aa2f18b1c8564170ebeb00da59aa1a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-06 17:52:45.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9be2153f424f9ec6b0c951c4d3ace9b1e60a9a22","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-06 18:17:38.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0b357c4b823469fc528175a23caa62f72f87e204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 18:43:10.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018fc92961194d27a2a82af273e9d4c5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/68dc6cce2cee4f378fea623c4f21452d : FAILURE in 48m 42s","accounts_in_message":[],"_revision_number":13},{"id":"a6a9b6eaa9df4f0247258f9f70f6c880435964d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 18:55:13.000000000","message":"Patch Set 13: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49b79aa7a1174d55a6b0b9a2e01f441d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe0216acab24fceb84d314b94b536c3 : SUCCESS in 4m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/dde581e3b8bb4785a173c58adc263cad : SUCCESS in 27m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88caea97f8d44b7fb058b0b5815c4577 : SUCCESS in 26m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21544fe5a5944af3a3f480093e5f98d5 : SUCCESS in 30m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55daadcccd524af59ef7d48a5c501b90 : SUCCESS in 32m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a90c6e974a1e40f6bebe32639f5d3036 : SUCCESS in 10m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50eb080592e84cb280ef32987d419819 : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/776dfd5dad6e4bf6859a8486efafe487 : SUCCESS in 9m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a78a5b72ff7f4ab59e009b9000fc4a22 : SUCCESS in 3m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca6b4eda54f94c609261625f184ff7af : FAILURE in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6f999e352744ae9ba4932f47fa5adc : FAILURE in 7m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0aa94cbce744441ca270cdce00c01977 : FAILURE in 10m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d32c66942b97448ba359e23e087fb79b : FAILURE in 22m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/02ae6091eff24861b64019135a63e4d0 : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/736b5f3e109847758e61b8c16a852e15 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ecee94d2ebc4023a71fd47e66c50e24 : SUCCESS in 9m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2988b83e43c84b0f8a189c759acfe6d3 : SUCCESS in 5m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/431afbd05e594b29a7462683fcc8b4c6 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/96c2310c3221441686c67ed42c4b5739 : SUCCESS in 4m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b655d04a29d14629bb0a84d131e6e929 : SUCCESS in 36m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bf011b857904e1da7444b27e6738fce : SUCCESS in 33m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ec5990707d045d4b940ab29e3938e95 : SUCCESS in 33m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aa43900e43a43a187d1b5730ba306ab : FAILURE in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e4658f5162400b80e92840e81ca56c : FAILURE in 2m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29664b3d6d4c4f70abc6a1df3143381d : SUCCESS in 10m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea6c8afec2a449bab4883ceaf1684943 : SUCCESS in 33m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6737da63603a46568eba581b6fce102b : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":13},{"id":"02b9c079b9929df360e15661c1e50bbe634b6f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-07 18:31:45.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"37771de86f8eaf88b143a09e2729fdd5f5c876d6","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-07 18:33:08.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"33cf80f498e5316c9ad01d819c4963b328c3ea7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:36:29.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31f0b8ba9bcf486a80cbc82faccf661a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329d4856d79447529e0fa4ca6c95c459 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/66ec2d44315f400ea5c5db564bd0f9c3 : SUCCESS in 57m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a06ee2a2ed4a46c1b7861e7a531234fd : SUCCESS in 55m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcf260bfd87c4f7eba1883200752f107 : SUCCESS in 33m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/166ea6bd87734038adad745eadcdedc7 : SUCCESS in 46m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d5adb4718284e76a5766041f5e60f34 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/831772d85eba4940a8a457abbd012da8 : SUCCESS in 23m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9c9b83951054b45b5314f2ba5a1ca3f : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50b367e6d0984cb2b7bbf20bf10d20e9 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80ce73baffa24ba8a40ba2dfe6d6defc : SUCCESS in 20m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f558a97c10004dd5978130cbb9c67a82 : SUCCESS in 13m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f791529f0da347c99807acf3f3587532 : SUCCESS in 16m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/195eb21e04dd4f148bf99c3888ab5ea5 : SUCCESS in 19m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6828af815c454c8da9ef35e383889cd7 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab033fcdb9f740f98e72891ab4635f38 : SUCCESS in 12m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7ab4409b1b1a477294e479db2d411817 : SUCCESS in 12m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/095aaf7af1244982bb8858613822a591 : SUCCESS in 6m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40ce40fcc13f4f81bfc58ac6da6b642a : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/48764c6bca3a4bb89609af0bf0d97560 : SUCCESS in 5m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8eec33181484977979a3b9901b90ccd : SUCCESS in 46m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/033b892fa4ff49b3a0f1c98fb0e356fe : SUCCESS in 31m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8fa014260ecf439e856c2bc439f07672 : SUCCESS in 33m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7746e31944244d1b8b3be2038cea6869 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735450d7edfb4139a1dfe561f2ca4af8 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45c60fc7c0394a589d553e398d7e506a : SUCCESS in 20m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22b4b064559d495f8d6837cf1c14c4e3 : SUCCESS in 41m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7ef47c8865847d98b4634926f1f7f21 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":14},{"id":"1c224921748c7c9f905fa6a5847d10f1135cc894","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-07 19:45:04.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"58c3f55d2c1bb98774ccf43f4baef3bd1c5b0433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 20:01:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5625bc8fad404ebfbb75fa82e2985241\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f3730beb94e4deb9ac72a8e78cc1787 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":14},{"id":"6456ec9fc06e18b5161a07beb46e25f7d4870d8f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-05-08 10:26:03.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"7556b9163a5ea0df8f3116295937f20b79553d1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-08 16:28:17.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"c425514780783ddbd13f5e7fcf3ddb624c399139","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-08 16:29:27.000000000","message":"Patch Set 15: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"1b7215e34d9a8d3e2bf6302030df0edc22a287d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:30:18.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce9da96a1c954a2995cc561f2614c5de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452857edb503449985095305f201663e : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/5acfb13047c041fb9d94a751e97816b1 : SUCCESS in 41m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7f20f7013ec484baccfec4e75e1e8ca : SUCCESS in 55m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b775848de1d044a8a10716f2947fc80f : SUCCESS in 35m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0777004272d846748f60293dd627f74f : SUCCESS in 57m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8339566c55f942b7b67fe0fea93b9052 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b71206c8f462438dbd7d453af42a0ec5 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abf94e3c33f8400ca038c5e0048786bb : SUCCESS in 15m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4e91ac016b544728d60ce70f9272254 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22a673082c934eb5951728931072361a : FAILURE in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22ff6ee562724406b8800c6d660c6ff6 : FAILURE in 21m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d76841974da4d4984c30822fb371263 : FAILURE in 17m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/143ecbec9b3d46ac9c957836b968f8cb : FAILURE in 28m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/93f386252a4a4fb99d0f86faa2323801 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f78de6114c894786b59ea04fee2d0668 : SUCCESS in 14m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/edee0d7ab946432c86febdb9491d3024 : SUCCESS in 15m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c91b447dc474bf2bdac04e5975fd43d : SUCCESS in 5m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e92be3dc577b4720982e8b27c04a46ce : SUCCESS in 10m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/38a6d211b2314beaa19808e26c466c9c : SUCCESS in 4m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf42bdbcf6f94538867f7dc274610558 : SUCCESS in 54m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c4777278ddb4fda8117dd48a1b0e160 : SUCCESS in 49m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40a72f2d40c9493ba1b1e1dbc679b9f5 : SUCCESS in 34m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e126ee3d3bf4c099b03d2aef7a29f7f : FAILURE in 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2799c81953e4dedb68c0e7da1b06d1f : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29bb6677b6b140e09042b8a3958d6543 : SUCCESS in 11m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ff79e18356b4a9ea04954406a2356ea : SUCCESS in 47m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ce1e1412b7c43678209bfdee990fa70 : SUCCESS in 55m 29s","accounts_in_message":[],"_revision_number":15},{"id":"a9867acad918ea288c56d211909ff70a9f6bf4ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:38:06.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cacbf72bbd4708a82c7b4b64acf413\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f904fb868c7d4076a27bd5ca18a9515d : FAILURE in 1h 08m 21s","accounts_in_message":[],"_revision_number":15},{"id":"1be117d5118bffd41b038d638ba5d9d91cb2a459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-08 18:25:31.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"1d452676059f842cb0e1808fab25269ad309a481","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-08 18:26:09.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"fe282e559e7c315e7666dbddfc50c092eec8ac82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 19:06:54.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b07b7ee262b445ae94d951072d939c67\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3368cd050c6b4bc4a402f29663e372ee : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":16},{"id":"a699b3572c3fdac281079bf690476fe5f28827eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 19:33:34.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7daf198fe124306873daf6797cd5601\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d555df18d7348029cef5873328a9dd0 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/f131203210b94c1796dddba2b018c486 : SUCCESS in 54m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/323b9180bc1345d8bb662839bc6e9ab0 : SUCCESS in 29m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ad68a45745e4cd99c6c7ca5964f6bea : SUCCESS in 49m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91defaae6f02420a87389fca2515d6d2 : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fc3079a287e4ed3af71a8fb65ac5946 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ac911a55b084988b2e531a061b16df2 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e9614c42bd84bdfa506d09446d7406e : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59761adbc5c5434c8547f1a8c20f213b : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/98102c5426444a589f934eadae7d8868 : SUCCESS in 19m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39d376f8fbcb442e84e9b8d22b45cc56 : SUCCESS in 8m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5e02d483bfa247ec873f505c61ff12f8 : SUCCESS in 33m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6b7895abe7e442bc9d3b43e57a1d6352 : SUCCESS in 23m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e968680d0ff44eacb9c8892cf5947a6a : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/998fc2e3078944f8a8d9d120992bb41b : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/180a423b59394294a5a9a5ad1ed7a68a : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce43564eba87481891c88b08d1789e40 : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d722947f5fb9485f8ed6de2e279f4de0 : SUCCESS in 10m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0bd673f225c4768848145fcd1d475a5 : SUCCESS in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a4cbe3302e041cbb718213f08fef820 : SUCCESS in 53m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7069c0f9fdb4cba807a1bddb666b169 : SUCCESS in 43m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/531c48f5c8a34bb899159a87db2b96fd : SUCCESS in 35m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdafc77dcde24a4e8f03db4670464c98 : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b200b6b277643b0ba8288bc733d8cbb : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/187297b922cf4570a481bd4c278f389c : SUCCESS in 22m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33fa07b77ab6449095eb7982ebac5c33 : SUCCESS in 56m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f129f7c6e9de41e8827c2b4cabf0bf79 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":16},{"id":"9554e27f43d24ed0f8aa7acbfcadb4fbb236b2e6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-09 16:45:44.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"25345dfe82c93f805b975b68fb780447d34608cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 16:46:13.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"5317b08704759ade3fc68d4157c284a7c169f68c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 17:52:33.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbd96f596edd493995c1cf1b8e05fa39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ffd531d3224005bfa7cf1014af6600 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/e875523669464eaea54205c9c317f21b : SUCCESS in 44m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e84ec9dec82046e488ff598b7bb54930 : SUCCESS in 45m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3bf6e430448949efb96e01f703062fa8 : SUCCESS in 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56894c46a7214447bb5a1a0e4db4aba5 : SUCCESS in 55m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4301cb11f30a4bfb8534e0a5f412f94e : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2e660d096174e91822e1b54aad4be1f : SUCCESS in 13m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e6ed54c8b5c544fdaa9edd3f01042339 : SUCCESS in 20m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4a436686de2942f7b2ff0c2739b16a86 : SUCCESS in 22m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e07a8a90424c4451ae6ebce9503b633f : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f917a08c0e024c0bbd5917f1f507050c : SUCCESS in 7m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6da8f5e4718e4bf1a3221784408900b3 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9c590013e9704ff0a5d244f537a925e4 : SUCCESS in 5m 36s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e67b9bc4b4214cd1a170145eb5d91f79 : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df33b5dd8b8949f39a9492188f63cb55 : SUCCESS in 36m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d60260a575c4838a2b8ef3aaf31fc29 : SUCCESS in 42m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d07b986b52034d1b9f4a2411c4528960 : SUCCESS in 22m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/703c838aca9b4bc8a3887cfcf129a10a : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15aedcec8117400f988c34e66528be2c : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75bf3eefce544e10a426073a8813a253 : SUCCESS in 20m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26417fe41eed443a99068eb4ceeb8b8b : SUCCESS in 43m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/936e9941023b4e9289212eb9046520db : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":16},{"id":"509bfe015376adc522cecfa8cafe0931b6a31961","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 17:52:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"d5f8ba1d854d0ea9f392007e5ade470c5f775924","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 17:54:01.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e99dba4dca7043c0bcb5032fad4236bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6dcd87ef13c41e28bb4c9a285f42f17 : SUCCESS in 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9adb6ae6c226470087e57eec2e42165f : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"9a445503e934eaedfdceb3afc8908a3c6688d51c","revisions":{"34674738af1942fa313ae2f7df9eb17e2bd6997a":{"kind":"REWORK","_number":1,"created":"2025-02-05 04:40:05.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/1"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-05 04:39:38.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-05 04:40:03.000000000","tz":-300},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34674738af1942fa313ae2f7df9eb17e2bd6997a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34674738af1942fa313ae2f7df9eb17e2bd6997a"}]},"branch":"refs/heads/master"},"64d2920b601604cfc43c3f80e1dbbd1f1d5a99a4":{"kind":"REWORK","_number":2,"created":"2025-02-06 04:00:54.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/2"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-05 04:39:38.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-06 04:00:30.000000000","tz":-300},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64d2920b601604cfc43c3f80e1dbbd1f1d5a99a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64d2920b601604cfc43c3f80e1dbbd1f1d5a99a4"}]},"branch":"refs/heads/master"},"777f66f805445c2ef6f870054ef6d0ebd5fbebbe":{"kind":"REWORK","_number":3,"created":"2025-02-07 05:36:27.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/3"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-05 04:39:38.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-07 05:36:03.000000000","tz":-300},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777f66f805445c2ef6f870054ef6d0ebd5fbebbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777f66f805445c2ef6f870054ef6d0ebd5fbebbe"}]},"branch":"refs/heads/master"},"159804c6f41ac81945d2361bd14935967ed1499c":{"kind":"REWORK","_number":4,"created":"2025-03-17 19:24:40.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/4"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-02-05 04:39:38.000000000","tz":-300},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-03-17 19:24:27.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/159804c6f41ac81945d2361bd14935967ed1499c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/159804c6f41ac81945d2361bd14935967ed1499c"}]},"branch":"refs/heads/master"},"459ece04d3c1b145b71c3c1c29d08b88c9a04524":{"kind":"REWORK","_number":5,"created":"2025-03-24 23:16:25.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/5"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-03-24 23:13:10.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/459ece04d3c1b145b71c3c1c29d08b88c9a04524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/459ece04d3c1b145b71c3c1c29d08b88c9a04524"}]},"branch":"refs/heads/master"},"507eec8d3b994e73e3846b90a61210559bb417dc":{"kind":"REWORK","_number":6,"created":"2025-03-24 23:25:36.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/6"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-03-24 23:25:27.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/507eec8d3b994e73e3846b90a61210559bb417dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/507eec8d3b994e73e3846b90a61210559bb417dc"}]},"branch":"refs/heads/master"},"4bc5b0e0c935be37e549aa38b63aba03a433faa7":{"kind":"REWORK","_number":7,"created":"2025-03-25 15:56:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/940762/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/7"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-03-25 15:54:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-25 15:54:18.000000000","tz":-420},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bc5b0e0c935be37e549aa38b63aba03a433faa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bc5b0e0c935be37e549aa38b63aba03a433faa7"}]},"branch":"refs/heads/master"},"5436a6298fc68c23971f3870f0131f382602e8a0":{"kind":"REWORK","_number":8,"created":"2025-04-04 01:09:45.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/8"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-04-04 01:09:33.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5436a6298fc68c23971f3870f0131f382602e8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5436a6298fc68c23971f3870f0131f382602e8a0"}]},"branch":"refs/heads/master"},"37478a01afdad0516ebdc386bcd064b1daaea259":{"kind":"REWORK","_number":9,"created":"2025-05-01 01:11:49.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/9"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-01 01:08:56.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37478a01afdad0516ebdc386bcd064b1daaea259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37478a01afdad0516ebdc386bcd064b1daaea259"}]},"branch":"refs/heads/master"},"1175930cf220fb65d6e922b6c43b7fe819dc3f8e":{"kind":"REWORK","_number":10,"created":"2025-05-01 01:12:54.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/10"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-01 01:12:43.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1175930cf220fb65d6e922b6c43b7fe819dc3f8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1175930cf220fb65d6e922b6c43b7fe819dc3f8e"}]},"branch":"refs/heads/master"},"aedbaac1e71fff7ae73738ffb1b6995c698bb0eb":{"kind":"REWORK","_number":11,"created":"2025-05-01 01:14:37.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/11"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-01 01:14:33.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aedbaac1e71fff7ae73738ffb1b6995c698bb0eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aedbaac1e71fff7ae73738ffb1b6995c698bb0eb"}]},"branch":"refs/heads/master"},"87a481d91ce51fe062e4441cf7cafb71834826a9":{"kind":"REWORK","_number":12,"created":"2025-05-06 17:47:17.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/12"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-06 17:47:04.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87a481d91ce51fe062e4441cf7cafb71834826a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87a481d91ce51fe062e4441cf7cafb71834826a9"}]},"branch":"refs/heads/master"},"9e8533d5db4c47d578bc078fb320f3cc534fd87d":{"kind":"REWORK","_number":13,"created":"2025-05-06 17:52:45.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/13"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-06 17:52:35.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e8533d5db4c47d578bc078fb320f3cc534fd87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e8533d5db4c47d578bc078fb320f3cc534fd87d"}]},"branch":"refs/heads/master"},"d0d177c933a73ba9cd45f89ea00b6e63a2e200ed":{"kind":"REWORK","_number":14,"created":"2025-05-07 18:31:45.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/14"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-07 18:28:09.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0d177c933a73ba9cd45f89ea00b6e63a2e200ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0d177c933a73ba9cd45f89ea00b6e63a2e200ed"}]},"branch":"refs/heads/master"},"9bd6ffe879e314189181e7d9b8e518c24efe8c81":{"kind":"REWORK","_number":15,"created":"2025-05-08 16:28:17.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/15"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 16:27:59.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bd6ffe879e314189181e7d9b8e518c24efe8c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bd6ffe879e314189181e7d9b8e518c24efe8c81"}]},"branch":"refs/heads/master"},"9a445503e934eaedfdceb3afc8908a3c6688d51c":{"kind":"REWORK","_number":16,"created":"2025-05-08 18:25:31.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/62/940762/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/940762/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/940762/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/940762/16"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-14 14:26:48.000000000","tz":0},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 18:25:23.000000000","tz":-240},"subject":"Implement heartbeat response for COPY request","message":"Implement heartbeat response for COPY request\n\nThis will  prevent client timeouts during large object copies, similar to existing support in\nbulk operations and SLO manifest writes. The operator can set the frequency using the\nyield_frequency inside conf file.\n\nTo use heartbeat, simply add \"heartbeat\u003don\" query parameter, which will:\n- Returns a 202 Accepted response\n- Sends periodic 1 byte whitespace heartbeat every yield_frequency\n- Returns final status and headers\n- Supports both plain text and JSON\n\nCloses-Bug: #1738306\n\nChange-Id: Iaf4355f2e0edbd6dde226231f3d7a61a5e53ecfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a445503e934eaedfdceb3afc8908a3c6688d51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a445503e934eaedfdceb3afc8908a3c6688d51c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
